)]}'
{"id":"openstack%2Freleases~780444","triplet_id":"openstack%2Freleases~master~Ied5cdfa45b93a6e9b0e4479dd6d1cb925a2c3dae","project":"openstack/releases","branch":"master","topic":"new/adjutant_1_1_0rc1","hashtags":[],"change_id":"Ied5cdfa45b93a6e9b0e4479dd6d1cb925a2c3dae","subject":"release Adjutant and Adjutant-UI for Wallaby","status":"MERGED","created":"2021-03-15 02:51:13.000000000","updated":"2021-03-18 10:17:30.000000000","submitted":"2021-03-18 10:17:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"780444-new/adjutant_1_1_0rc1","meta_rev_id":"0ef4a315a984818ef68a1effdfba2c6050364adc","_number":780444,"virtual_id_number":780444,"owner":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-18 10:17:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2021-03-18 09:59:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2021-03-17 09:55:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2021-03-18 09:59:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-03-16 23:48:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-15 02:52:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-15 08:32:15.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-15 10:20:26.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"CC"},{"updated":"2021-03-18 09:59:33.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"63d57e075172b6c5d1fcd50fdbc700d943624d65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2021-03-15 02:51:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"16ea7bf47a259179468f87d905ec67e512b324b3","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 02:52:04.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"88a2f099da229a8962210e213624203e16efb6be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2021-03-15 02:55:28.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"308b4e9c9caba4d066a4397a19bd853a19eb86a1","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 02:56:20.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"4bf71bfb4fac27ce5d5cd5895b6ac1266f1d2020","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 03:07:18.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e71e320f76c94b63aff22c9885245f83 : SUCCESS in 6m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b374179a78eb489dbfd0e8b9feeff4d0 : FAILURE in 5m 07s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/83b382eeb07a4217807136fbbc3801f7 : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":2},{"id":"2ace1cf7a00140aba94e53d0a2a3a3160b8d58f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2021-03-15 04:22:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d16ab90b02dc2858af25335881b17dfd723e86de","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 04:23:37.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"bdd36de129531badac7150c1d3a9bd5deef97361","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 04:33:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac6bfd7dcdf8453fb3dd3885c3a1e5fd : SUCCESS in 5m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9d5b74bfc8984f8facecd0fc630fdcbd : SUCCESS in 4m 54s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/3a8226d44d6644fbb22689e32a6434dc : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":3},{"id":"94fa7c889faabcccf793acdd73bec9141e9b4059","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-15 08:32:15.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"f9cb06992e08c01aa5601d4a409d3c830c5af12d","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-03-15 10:20:26.000000000","message":"Patch Set 3:\n\nCycle-with-rc model generally mandates that we do a major bump between series... Holding W+1 until we get a chance to discuss in release management team.","accounts_in_message":[],"_revision_number":3},{"id":"99ee6deca2f9cbd3f01a9e58c4201c4d9dfdce22","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2021-03-16 00:52:10.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3:\n\u003e \n\u003e Cycle-with-rc model generally mandates that we do a major bump between series... Holding W+1 until we get a chance to discuss in release management team.\n\nIf that\u0027s a requirement, then I can update the patch. But in truth I wouldn\u0027t have tagged 1.0 even just yet. :P\n\nI\u0027d ideally have had Adjutant at 0.7.0 right now. So my preference for now is keep it 1.* and go to 2.0 when I actually get the time/chance to finally add my async workers for task processing.","accounts_in_message":[],"_revision_number":3},{"id":"2a08cdf8a7b125b852e7954a11bb9fef77de1a32","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-16 10:29:24.000000000","message":"Patch Set 3: -Code-Review\n\nHello,\n\nI agree with Thierry we usually start with a major version [1][2] and I think we should continue with that here.\n\nThe cycle-with-rc model is the historic OpenStack release model.\nIn this model, a single release is produced at the end of the development\ncycle for inclusion in the coordinated OpenStack release. The major release\nnumber (the X in X.Y.Z numbers) is incremented for each release series.\n\nI don\u0027t know exactly why we choosen this rule but think it be worth to continue to follow it until we have a related official discussion and until this point is socialized enough to give a chance to everybody to raise their hands.\n\n\n\n[1] https://docs.openstack.org/project-team-guide/release-management.html#services-horizon-plugins-and-other-deliverables\n[2]\n```\ngrep -ri \ndeliverables/victoria -e cycle-with-rc | awk -F \":\" \u0027{print $1}\u0027 | xargs cat | grep rc1 | less \n```","accounts_in_message":[],"_revision_number":3},{"id":"288f50f0bd5705516f66905593d43356f04bf771","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-16 10:35:18.000000000","message":"Patch Set 3:\n\nHere you can find more details about the time where this rule has been added:\n\nhttps://github.com/openstack/project-team-guide/commit/647bcbdc1cc6456a09a9580b58e51a32fbe7f567","accounts_in_message":[],"_revision_number":3},{"id":"ac8963ba62baadcd6f34edbfec5782d57fb2e2d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2021-03-16 23:00:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e0e8b20620983829df311b10e17969cf01db9baf","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 23:01:53.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"0702fd344beb57fed04d0faafef041b2f2ee90ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 23:12:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cb43c91d44a4ef4b84609fe01fc4029 : SUCCESS in 6m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/764f3ba00cdb42abb7fb653c0e1f328e : SUCCESS in 5m 12s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c5da6d94163d499aacb87159efbf17c8 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":4},{"id":"2e8582a20ebe7c648d9c1515773edee3e2e9df3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2021-03-16 23:47:28.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"f3cd806641b4c3c0140ff6a265b6edb4c09cd4de","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 23:48:32.000000000","message":"Patch Set 5: PTL-Approved+1","accounts_in_message":[],"_revision_number":5},{"id":"ddd54be9804074e51dca246429f90cb0ecf98e82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 23:54:29.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e07fa211a7c45f787e887c2eb800e5f : SUCCESS in 5m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5793a490c93a42eb82fd4aa5a7f3996c : SUCCESS in 4m 30s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d2fa58cdf27147eb86ba86b44351e00f : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":5},{"id":"c5a5b1cdb35d6cdfd242e27507e726a59c5d5e3c","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2021-03-16 23:56:51.000000000","message":"Patch Set 5:\n\nI\u0027ve always been a fan of semver style versioning but I don\u0027t really mind too much.\n\nJust feels like with the current openstack versioning, we essentially have two whole version numbers we barely touch, and the version number doesn\u0027t particularly convey anything useful in terms of API changes or major improvements/differences.\n\nPlus it feels weird when there aren\u0027t many actual changes to the service, but it gets a major version bump.","accounts_in_message":[],"_revision_number":5},{"id":"0a2707097a9c8ee26750563fac15959ac13db0dd","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-17 09:54:23.000000000","message":"Patch Set 5:\n\n\u003e Patch Set 5:\n\u003e \n\u003e I\u0027ve always been a fan of semver style versioning but I don\u0027t really mind too much.\n\u003e \n\u003e Just feels like with the current openstack versioning, we essentially have two whole version numbers we barely touch, and the version number doesn\u0027t particularly convey anything useful in terms of API changes or major improvements/differences.\n\u003e \n\u003e Plus it feels weird when there aren\u0027t many actual changes to the service, but it gets a major version bump.\n\nThanks for the update.\n\nI love semver too but it doesn\u0027t solve everything either. Semver have pros and cons...\n\nFrom the developer point of view APIs are all the interface provided by a project, from a user point of view APIs are these ones too + the associated user experience (performances, stability, etc...) associated to these APIs and semver doesn\u0027t prevent regression on this experience and can\u0027t represent them.\n\nProjects can be semver compatible and stick to X\u003c\u003d0 all along their life and never enter into the backward changes lifecycle and never take the risk to stabilize something for their users.\n\nProjects can propose major releases all the time and inform that the UX can be impacted from one series to another. I think that we should mostly consider our current versionning rules like that.\n\nBy example, changes on neutron can have several impacts on nova performances even if there aren\u0027t many actual major changes into nova. I think that major versions better translate this point. Besides the libs versioning rules (like on oslo.*) are differents and are more respectful of semver and the boths paradigms can cooexist without to much pain\n\nSemver is just a tool that mostly reflect the way that you use it.","accounts_in_message":[],"_revision_number":5},{"id":"14dbfabfa148602f205bb489a348b72b77a3a51d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-17 09:55:09.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"6fe9a6bd3f3308a6c71dc6cdd5222fe38ffeef4a","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-03-18 09:59:33.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"c29f7a96f4fe1110cf98196661db0118a2cf01bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 09:59:58.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"803d46141c1dcf2586b7893f936462c2616366a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 10:17:29.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0df450bedfc647c5aa6c7e49a2d343a2 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e4f6cc0bc404d04b8ddd0d842db1bdd : SUCCESS in 7m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a763786599194c16aaa1fd44de3cd3d8 : SUCCESS in 6m 52s","accounts_in_message":[],"_revision_number":5},{"id":"0ef4a315a984818ef68a1effdfba2c6050364adc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 10:17:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"11916536d29fcec70692390dce7ea69f3bf72434","revisions":{"25efa393e442ccc5fd0637038e651df99b630b02":{"kind":"REWORK","_number":1,"created":"2021-03-15 02:51:13.000000000","uploader":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"ref":"refs/changes/44/780444/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/780444/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/780444/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/780444/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/780444/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/780444/1"}}},"commit":{"parents":[{"commit":"6e7cbf618af7cce28356827cb7a0339532c68a42","subject":"Add Wallaby cycle highlights for Blazar","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6e7cbf618af7cce28356827cb7a0339532c68a42"}]}],"author":{"name":"Adrian Turjak","email":"adriant@catalystcloud.nz","date":"2021-03-15 02:50:32.000000000","tz":780},"committer":{"name":"Adrian Turjak","email":"adriant@catalystcloud.nz","date":"2021-03-15 02:50:32.000000000","tz":780},"subject":"release Adjutant and Adjutant-UI for Wallaby","message":"release Adjutant and Adjutant-UI for Wallaby\n\nChange-Id: Ied5cdfa45b93a6e9b0e4479dd6d1cb925a2c3dae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/25efa393e442ccc5fd0637038e651df99b630b02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/25efa393e442ccc5fd0637038e651df99b630b02"}]},"branch":"refs/heads/master"},"a7e1bae623baa75fb2b82fd80f79b36670047bdd":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-03-15 02:55:28.000000000","uploader":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"ref":"refs/changes/44/780444/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/780444/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/780444/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/780444/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/780444/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/780444/2"}}},"commit":{"parents":[{"commit":"6e7cbf618af7cce28356827cb7a0339532c68a42","subject":"Add Wallaby cycle highlights for Blazar","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6e7cbf618af7cce28356827cb7a0339532c68a42"}]}],"author":{"name":"Adrian Turjak","email":"adriant@catalystcloud.nz","date":"2021-03-15 02:50:32.000000000","tz":780},"committer":{"name":"Adrian Turjak","email":"adriant@catalystcloud.nz","date":"2021-03-15 02:51:57.000000000","tz":780},"subject":"release Adjutant and Adjutant-UI for Wallaby","message":"release Adjutant and Adjutant-UI for Wallaby\n\nNot worth tagging with major version change, so will\nstick to minor version bumps unless changes are enough to\nwarrant a major version.\n\nChange-Id: Ied5cdfa45b93a6e9b0e4479dd6d1cb925a2c3dae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a7e1bae623baa75fb2b82fd80f79b36670047bdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a7e1bae623baa75fb2b82fd80f79b36670047bdd"}]},"branch":"refs/heads/master"},"549341c5c3d93dbcb65d0b60aa7a6709fefcf091":{"kind":"REWORK","_number":3,"created":"2021-03-15 04:22:49.000000000","uploader":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"ref":"refs/changes/44/780444/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/780444/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/780444/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/780444/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/780444/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/780444/3"}}},"commit":{"parents":[{"commit":"6e7cbf618af7cce28356827cb7a0339532c68a42","subject":"Add Wallaby cycle highlights for Blazar","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6e7cbf618af7cce28356827cb7a0339532c68a42"}]}],"author":{"name":"Adrian Turjak","email":"adriant@catalystcloud.nz","date":"2021-03-15 02:50:32.000000000","tz":780},"committer":{"name":"Adrian Turjak","email":"adriant@catalystcloud.nz","date":"2021-03-15 04:22:40.000000000","tz":780},"subject":"release Adjutant and Adjutant-UI for Wallaby","message":"release Adjutant and Adjutant-UI for Wallaby\n\nNot worth tagging with major version change, so will\nstick to minor version bumps unless changes are enough to\nwarrant a major version.\n\nChange-Id: Ied5cdfa45b93a6e9b0e4479dd6d1cb925a2c3dae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/549341c5c3d93dbcb65d0b60aa7a6709fefcf091"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/549341c5c3d93dbcb65d0b60aa7a6709fefcf091"}]},"branch":"refs/heads/master"},"d4dd795add7e975158946490a902fd8cef826083":{"kind":"REWORK","_number":4,"created":"2021-03-16 23:00:18.000000000","uploader":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"ref":"refs/changes/44/780444/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/780444/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/780444/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/780444/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/780444/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/780444/4"}}},"commit":{"parents":[{"commit":"6e7cbf618af7cce28356827cb7a0339532c68a42","subject":"Add Wallaby cycle highlights for Blazar","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6e7cbf618af7cce28356827cb7a0339532c68a42"}]}],"author":{"name":"Adrian Turjak","email":"adriant@catalystcloud.nz","date":"2021-03-15 02:50:32.000000000","tz":780},"committer":{"name":"Adrian Turjak","email":"adriant@catalystcloud.nz","date":"2021-03-16 23:00:05.000000000","tz":780},"subject":"release Adjutant and Adjutant-UI for Wallaby","message":"release Adjutant and Adjutant-UI for Wallaby\n\nNot worth tagging with major version change, so will\nstick to minor version bumps unless changes are enough to\nwarrant a major version.\n\nChange-Id: Ied5cdfa45b93a6e9b0e4479dd6d1cb925a2c3dae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d4dd795add7e975158946490a902fd8cef826083"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d4dd795add7e975158946490a902fd8cef826083"}]},"branch":"refs/heads/master"},"11916536d29fcec70692390dce7ea69f3bf72434":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2021-03-16 23:47:28.000000000","uploader":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"ref":"refs/changes/44/780444/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/780444/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/780444/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/780444/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/780444/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/780444/5"}}},"commit":{"parents":[{"commit":"6e7cbf618af7cce28356827cb7a0339532c68a42","subject":"Add Wallaby cycle highlights for Blazar","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6e7cbf618af7cce28356827cb7a0339532c68a42"}]}],"author":{"name":"Adrian Turjak","email":"adriant@catalystcloud.nz","date":"2021-03-15 02:50:32.000000000","tz":780},"committer":{"name":"Adrian Turjak","email":"adriant@catalystcloud.nz","date":"2021-03-16 23:47:19.000000000","tz":0},"subject":"release Adjutant and Adjutant-UI for Wallaby","message":"release Adjutant and Adjutant-UI for Wallaby\n\nChange-Id: Ied5cdfa45b93a6e9b0e4479dd6d1cb925a2c3dae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/11916536d29fcec70692390dce7ea69f3bf72434"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/11916536d29fcec70692390dce7ea69f3bf72434"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
