)]}'
{"id":"openstack%2Freleases~780559","triplet_id":"openstack%2Freleases~master~I129bed0051631c04ea53c088dfe5c9823e306a93","project":"openstack/releases","branch":"master","topic":"puppet-openstack-wallaby-m2","hashtags":[],"change_id":"I129bed0051631c04ea53c088dfe5c9823e306a93","subject":"Puppet OpenStack wallaby Milestone 2","status":"MERGED","created":"2021-03-15 12:12:12.000000000","updated":"2021-03-22 08:58:48.000000000","submitted":"2021-03-22 08:58:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"780559-puppet-openstack-wallaby-m2","meta_rev_id":"88f024b09a494fd90a309e28a66af7a299acd139","_number":780559,"virtual_id_number":780559,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-22 08:58:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2021-03-22 08:47:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2021-03-19 11:01:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2021-03-22 08:47:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-03-19 08:45:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-15 12:13:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-15 20:29:35.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-18 09:51:35.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"bf48f30a2ac4303abd4ed438e9e8928107474665","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-03-15 12:12:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1bfdd77da5a0af4eaa4f21fcd2fb32bafc9fd15d","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 12:13:10.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"3cdc1f18f5d31a14dcb352aeaa9faccf0585ce5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 12:26:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5cb4065a9064411af555761bf609319 : SUCCESS in 6m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a9a77e999fae4f23a6d934e930da76ad : SUCCESS in 6m 13s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/08641558e81242a99037da43df4deee6 : SUCCESS in 12m 39s","accounts_in_message":[],"_revision_number":1},{"id":"37991396e54596d5c9e3914d21cc98376a92b46c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-15 20:29:35.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)\n\nLGTM trailing project","accounts_in_message":[],"_revision_number":1},{"id":"227c5b1f130f61ef7dc18137926e6e7de950b01d","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-03-17 09:50:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bb9aa51d14a38e57276d66766d9eb48799002d9e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-17 10:14:51.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"c208f3cd9fda38de11e24865999ef8d16a3c621d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-17 10:15:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"84c7b1d4f5fce8b8b326380389bcbeb6cbfa6aaf","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-03-17 10:54:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cb6594a3ed4275545a045f543ff66bae4fc112e6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-17 13:29:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6a8390d8d5fdcf8191925fed1a63e77bf07e6db1","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-17 13:36:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2797aa7904792a18dabeb19a2fc86d7f0f70d9db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-03-18 01:33:55.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b3c52f1ff69f0c115f2412c10f1ca6ecb97cbf40","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 01:35:04.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"ed47c114945876a259783caf8ac60693db0675fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 01:43:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c7495831e4142cfbeb43de1c66c4696 : SUCCESS in 7m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/722e2d491cc94ffb8963eeff454c3717 : SUCCESS in 7m 42s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/065e8781fd574d908d992bcec3a1e3ca : SUCCESS in 7m 38s","accounts_in_message":[],"_revision_number":2},{"id":"e5e74174678cc1b67a6c07e8ca0938a9a0936729","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-03-18 09:51:35.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"98487dba55dc6901f953e01beead0bfcfa95cacb","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-18 14:03:42.000000000","message":"Patch Set 2:\n\ni think you have to do the opposite. I think you need to abandon this patch and you need to reopen the M3 patch.\n\nhttps://review.opendev.org/c/openstack/releases/+/781057\n\nAll the used SHA in the current patch are outdated and correspond to M2. By using this current patch you will skip lot of changes and especially those which correspond to your M3 preparation, example:\n\nhttps://opendev.org/openstack/puppet-aodh/commit/8b1d40cf8da2294f1c383cb2dc7fd22dd6d4b6f0","accounts_in_message":[],"_revision_number":2},{"id":"7d8ba5c1ba78f0dceb7c947d22eaba81c1a4ed79","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-18 14:04:31.000000000","message":"Patch Set 2: Workflow-1\n\nHolding this patch for awhile","accounts_in_message":[],"_revision_number":2},{"id":"97b0fdf51cd06dbc9f23b6192e7aa85b452990a4","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-03-19 07:03:35.000000000","message":"Patch Set 2:\n\nthe new M3 patch is https://review.opendev.org/c/openstack/releases/+/781601","accounts_in_message":[],"_revision_number":2},{"id":"1496eed5f10fef1bf5779176a4c71aaca67a4f8c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-19 08:04:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"add46c2c3c6062a50292e14ea6770340c3fd9107","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-19 08:06:57.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e the new M3 patch is https://review.opendev.org/c/openstack/releases/+/781601\n\nThanks.\n\nFinally, if you want to continue with this one too then I\u0027ll approve it. Just don\u0027t forget to update the milestone number to avoid confusing","accounts_in_message":[],"_revision_number":2},{"id":"48fa683e565a81999a181d7f43dc2d7164dab3de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-03-19 08:43:59.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"65cb5915fefa1a89054324ef00b39e9786f1675b","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 08:45:06.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"741d4965fc9fd6bad14d1d9719c18d9605f60c8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 08:52:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35af24c0aacc4aeeb05af7c1b72b8796 : SUCCESS in 6m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f3dd593c1932424fa90b930c4862d010 : SUCCESS in 7m 34s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/23dc938625a94d2d94fec305c9960aac : SUCCESS in 7m 29s","accounts_in_message":[],"_revision_number":3},{"id":"ba53e35b4f026f807b21eb872d1166149f3c3cde","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-19 11:01:29.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"9678b0780d9848ffba26e32eb239c69812617267","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-03-22 08:47:03.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ce532e1a6fac1d3634b2708a40b102c1d6a42b9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 08:47:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0c0f0634f8713cb5535b38083b37b7f6cc185f18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 08:58:33.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c89c31ef9007499b85538f312f37e3f7 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a49d2170e554f8dad64e5437219d826 : SUCCESS in 6m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a75568154674470f9b98a0cc213841b2 : SUCCESS in 8m 09s","accounts_in_message":[],"_revision_number":3},{"id":"88f024b09a494fd90a309e28a66af7a299acd139","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 08:58:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f9ef52f3f00d12833cbc0a5fedd527be0a5a52b7","revisions":{"6269a4f40ed1a7ceec486b069c844033ee6f6653":{"kind":"REWORK","_number":1,"created":"2021-03-15 12:12:12.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/59/780559/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/59/780559/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/59/780559/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/59/780559/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/59/780559/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/59/780559/1"}}},"commit":{"parents":[{"commit":"3e83517ff6710d93a4609c2f3523f2bce5e810d9","subject":"Merge \"Add ovn-octavia-provider cycle highlights\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3e83517ff6710d93a4609c2f3523f2bce5e810d9"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-03-15 12:06:13.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-03-15 12:06:13.000000000","tz":480},"subject":"Puppet OpenStack wallaby Milestone 2","message":"Puppet OpenStack wallaby Milestone 2\n\nChange-Id: I129bed0051631c04ea53c088dfe5c9823e306a93\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6269a4f40ed1a7ceec486b069c844033ee6f6653"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6269a4f40ed1a7ceec486b069c844033ee6f6653"}]},"branch":"refs/heads/master"},"8cef1564c50d27277e44cb3b48e1433508a3b9c1":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-03-18 01:33:55.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/59/780559/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/59/780559/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/59/780559/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/59/780559/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/59/780559/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/59/780559/2"}}},"commit":{"parents":[{"commit":"3e83517ff6710d93a4609c2f3523f2bce5e810d9","subject":"Merge \"Add ovn-octavia-provider cycle highlights\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3e83517ff6710d93a4609c2f3523f2bce5e810d9"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-03-15 12:06:13.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-03-18 01:33:33.000000000","tz":480},"subject":"Puppet OpenStack wallaby Milestone 3","message":"Puppet OpenStack wallaby Milestone 3\n\nChange-Id: I129bed0051631c04ea53c088dfe5c9823e306a93\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8cef1564c50d27277e44cb3b48e1433508a3b9c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8cef1564c50d27277e44cb3b48e1433508a3b9c1"}]},"branch":"refs/heads/master"},"f9ef52f3f00d12833cbc0a5fedd527be0a5a52b7":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-03-19 08:43:59.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/59/780559/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/59/780559/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/59/780559/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/59/780559/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/59/780559/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/59/780559/3"}}},"commit":{"parents":[{"commit":"3e83517ff6710d93a4609c2f3523f2bce5e810d9","subject":"Merge \"Add ovn-octavia-provider cycle highlights\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3e83517ff6710d93a4609c2f3523f2bce5e810d9"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-03-15 12:06:13.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-03-19 08:43:34.000000000","tz":480},"subject":"Puppet OpenStack wallaby Milestone 2","message":"Puppet OpenStack wallaby Milestone 2\n\nChange-Id: I129bed0051631c04ea53c088dfe5c9823e306a93\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f9ef52f3f00d12833cbc0a5fedd527be0a5a52b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f9ef52f3f00d12833cbc0a5fedd527be0a5a52b7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
