)]}'
{"id":"openstack%2Freleases~780574","triplet_id":"openstack%2Freleases~master~Ia046f5eb91659f1c556e0a3750372ce39ee555f1","project":"openstack/releases","branch":"master","topic":"wallaby-stable-branches","hashtags":[],"change_id":"Ia046f5eb91659f1c556e0a3750372ce39ee555f1","subject":"[freezer] Create for all client and non-client lib wallaby branch","status":"MERGED","created":"2021-03-15 13:15:01.000000000","updated":"2021-03-18 10:21:29.000000000","submitted":"2021-03-18 10:21:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"780574-wallaby-stable-branches","meta_rev_id":"56e99033633c8d775e1d9b6da41874bc13f8a920","_number":780574,"virtual_id_number":780574,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-18 10:21:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2021-03-18 10:01:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2021-03-16 08:22:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-16 00:03:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2021-03-18 10:01:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-03-16 00:04:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13940,"name":"Saad Zaher","email":"eng.szaher@gmail.com","username":"szaher"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":13940,"name":"Saad Zaher","email":"eng.szaher@gmail.com","username":"szaher"},{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-15 13:16:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-15 13:20:48.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":13940,"name":"Saad Zaher","email":"eng.szaher@gmail.com","username":"szaher"},"state":"REVIEWER"},{"updated":"2021-03-15 13:20:49.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"state":"REVIEWER"},{"updated":"2021-03-15 13:26:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-18 10:01:02.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"c3915155c9fa0daa9f4ee4c02982e612f373019e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-15 13:15:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f7f3c1ba9d01204b7924c69c4d0df5714429ef1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 13:26:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b54d04f0323442f951c152329468538 : SUCCESS in 6m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6a66393ca2db46f191d13cd69b813c4e : SUCCESS in 5m 21s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c0353b92178145e7a3ea90b63c11cf97 : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":1},{"id":"091523e58299b5d83208e7df52ff9906abd97fbf","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2021-03-16 00:03:56.000000000","message":"Patch Set 1: Code-Review+1\n\nThanks.","accounts_in_message":[],"_revision_number":1},{"id":"467c492bf5032b1c05844fcdbbe0d019cab71600","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 00:04:43.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"fcf231ee48262a0ce75b489d902a0b819646234f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-16 08:22:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d1fd94c8979119c5c31095f310579022e38da13b","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-03-18 10:01:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d6c400ae99ef1ae9ea44810fba7fe93ce580bdd1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 10:04:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"539cdcff3cd4667efb442581efcf8b98a8bbe9a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 10:21:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eece6b962424471b9e37587f0cbc2dca : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c35eca9fd305414abd29deb31bb022f1 : SUCCESS in 7m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bf5d7613c77447b6933dc67db412062b : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":1},{"id":"56e99033633c8d775e1d9b6da41874bc13f8a920","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 10:21:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5c2763a3a9351b2c704a93f9059ff9e3fc96a5d2","revisions":{"5c2763a3a9351b2c704a93f9059ff9e3fc96a5d2":{"kind":"REWORK","_number":1,"created":"2021-03-15 13:15:01.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/74/780574/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/74/780574/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/780574/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/780574/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/780574/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/74/780574/1"}}},"commit":{"parents":[{"commit":"8109a32344d42803cd53304e27aec6426c83d232","subject":"Merge \"Release Octavia 6.2.0 for stable/ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8109a32344d42803cd53304e27aec6426c83d232"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-15 13:13:54.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-15 13:13:54.000000000","tz":60},"subject":"[freezer] Create for all client and non-client lib wallaby branch","message":"[freezer] Create for all client and non-client lib wallaby branch\n\nThis creates the stable/wallaby branch for all client and non-client\nlibraries\nnow that we are past the lib freeze on March 11.\n\nIf a team wants to wait for a specific patch to make it to the library,\nsomeone from the team can -1 the patch to have it held,or update that\npatch with a different commit SHA.\n\nPatches with no response will be merged by Friday, March 19.\n\nChange-Id: Ia046f5eb91659f1c556e0a3750372ce39ee555f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5c2763a3a9351b2c704a93f9059ff9e3fc96a5d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5c2763a3a9351b2c704a93f9059ff9e3fc96a5d2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
