)]}'
{"id":"openstack%2Freleases~781143","triplet_id":"openstack%2Freleases~master~Ide8d9cf6d823ee783bca4d96baecbcd13bad0217","project":"openstack/releases","branch":"master","topic":"glance-wallaby-rc1","hashtags":[],"change_id":"Ide8d9cf6d823ee783bca4d96baecbcd13bad0217","subject":"Release Glance Wallaby rc1 (22.0.0.0rc1)","status":"ABANDONED","created":"2021-03-17 18:21:20.000000000","updated":"2021-03-23 08:39:47.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1c5ef2235be993e58801594f53d699400bb0b775","_number":781143,"virtual_id_number":781143,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-03-18 14:18:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":-1,"date":"2021-03-18 13:40:25.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-17 18:22:10.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-03-17 18:24:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-17 18:33:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-17 19:20:08.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2021-03-18 14:21:09.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"}],"messages":[{"id":"4f7b640f1624fde07b6f0b2848a8b2819fc0eb96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-17 18:21:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b789f4a2a9e3fb9f56ca7e0e545d01f2a28389c","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-17 18:22:10.000000000","message":"Patch Set 1: Workflow-1\n\nJust getting this lined up for next week in case we don\u0027t have any other important changes. Will keep -W until then.","accounts_in_message":[],"_revision_number":1},{"id":"774d4f7cfb66f180bd1a42d0f7d677e84ba19e03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 18:33:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d02878830d04cf09bfd7c53b81afe30 : SUCCESS in 7m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d417620c65334a5dac0a88f836db4ddb : SUCCESS in 5m 59s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/a894b1d09c43468cbe9135c3dea7f8f1 : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":1},{"id":"755c8ae8ab7cac1b852f8872c271d7bc5fdc0b5d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-18 06:05:35.000000000","message":"Patch Set 1:\n\nI think we should cut the stable branch at this point as well.","accounts_in_message":[],"_revision_number":1},{"id":"13ec89a7bb999bf9332c5704cfe70cc61bb14873","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-18 13:39:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ac46a77e8172a75a2f35b9c60e84bee95f6dbae3","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-18 13:40:25.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"9a4ea511b3b15089dac4113e27c047e5a2835e59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 14:18:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a50e3d093af4bed92d3cdc09597c7a7 : SUCCESS in 7m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fc52e1e9a240482d94b5f9d883956036 : SUCCESS in 5m 21s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ea921915156b4ed686bd48cf86cf6dc0 : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":2},{"id":"b1e8f55c7eecee7504dad123e9d34535f5846c59","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-18 14:21:09.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 1:\n\u003e \n\u003e I think we should cut the stable branch at this point as well.\n\nYes","accounts_in_message":[],"_revision_number":2},{"id":"9cd2397916d20f2aef61f3f80da0f1e391014dea","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-19 12:26:01.000000000","message":"Patch Set 2:\n\nLet me know if you prefer to use my patch or this one https://review.opendev.org/c/openstack/releases/+/781653\n\nMy patch refer to the HEAD of glance","accounts_in_message":[],"_revision_number":2},{"id":"df19770cd1786f5e1bf42dcc8129d1e80c7c6813","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-22 13:56:45.000000000","message":"Patch Set 2:\n\n\u003e Let me know if you prefer to use my patch or this one https://review.opendev.org/c/openstack/releases/+/781653\n\u003e \n\u003e My patch refer to the HEAD of glance\n\nI just pushed this up at Abhi\u0027s request last week to get something staged and was going to rebase it this morning. However, the auto-generated one looks fine to me, so I think we should go with that.","accounts_in_message":[],"_revision_number":2},{"id":"b0a54019358a696e8e2a065ecef73d32bd4792c0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-23 08:39:28.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e \u003e Let me know if you prefer to use my patch or this one https://review.opendev.org/c/openstack/releases/+/781653\n\u003e \u003e \n\u003e \u003e My patch refer to the HEAD of glance\n\u003e \n\u003e I just pushed this up at Abhi\u0027s request last week to get something staged and was going to rebase it this morning. However, the auto-generated one looks fine to me, so I think we should go with that.\n\nSo I think we can abandon this patch.\n\nThanks for details Dan.","accounts_in_message":[],"_revision_number":2},{"id":"1c5ef2235be993e58801594f53d699400bb0b775","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-23 08:39:47.000000000","message":"Abandoned\n\nduplicated with https://review.opendev.org/c/openstack/releases/+/781653","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5cab867b2bd3030e1a76aedbedc76154d99f5fe3","revisions":{"a7bf0715f250ea79232c6646877e88b753fa544d":{"kind":"REWORK","_number":1,"created":"2021-03-17 18:21:20.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/43/781143/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/43/781143/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/781143/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/781143/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/781143/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/43/781143/1"}}},"commit":{"parents":[{"commit":"39056601340abd8df4bdc9efbbc53d2038b371d3","subject":"Merge \"barbican-tempest-plugin: release 1.2.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/39056601340abd8df4bdc9efbbc53d2038b371d3"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-17 18:19:55.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-17 18:19:55.000000000","tz":-420},"subject":"Release Glance Wallaby rc1 (22.0.0.0rc1)","message":"Release Glance Wallaby rc1 (22.0.0.0rc1)\n\nChange-Id: Ide8d9cf6d823ee783bca4d96baecbcd13bad0217\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a7bf0715f250ea79232c6646877e88b753fa544d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a7bf0715f250ea79232c6646877e88b753fa544d"}]},"branch":"refs/heads/master"},"5cab867b2bd3030e1a76aedbedc76154d99f5fe3":{"kind":"REWORK","_number":2,"created":"2021-03-18 13:39:20.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/43/781143/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/43/781143/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/781143/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/781143/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/781143/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/43/781143/2"}}},"commit":{"parents":[{"commit":"39056601340abd8df4bdc9efbbc53d2038b371d3","subject":"Merge \"barbican-tempest-plugin: release 1.2.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/39056601340abd8df4bdc9efbbc53d2038b371d3"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-17 18:19:55.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-18 13:39:08.000000000","tz":-420},"subject":"Release Glance Wallaby rc1 (22.0.0.0rc1)","message":"Release Glance Wallaby rc1 (22.0.0.0rc1)\n\nThis adds the rc1 for Wallaby and also creates the stable/wallaby\nbranch.\n\nChange-Id: Ide8d9cf6d823ee783bca4d96baecbcd13bad0217\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5cab867b2bd3030e1a76aedbedc76154d99f5fe3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5cab867b2bd3030e1a76aedbedc76154d99f5fe3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
