)]}'
{"id":"openstack%2Freleases~781601","triplet_id":"openstack%2Freleases~master~I48bc230ab69c6b802518aed9f7fc4b4ccaf0432d","project":"openstack/releases","branch":"master","topic":"puppet-openstack-wallaby-m3","hashtags":[],"change_id":"I48bc230ab69c6b802518aed9f7fc4b4ccaf0432d","subject":"Puppet OpenStack wallaby Milestone 3","status":"MERGED","created":"2021-03-19 06:47:32.000000000","updated":"2021-03-25 08:59:23.000000000","submitted":"2021-03-25 08:59:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"781601-puppet-openstack-wallaby-m3","meta_rev_id":"84d3e79518b14054d2938deb7e04e49b887d5308","_number":781601,"virtual_id_number":781601,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-25 08:59:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2021-03-25 08:48:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2021-03-23 08:36:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2021-03-25 08:48:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-03-23 02:19:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-19 06:48:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-19 08:03:53.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-25 08:48:03.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"96f792e817687e6ed1822f974b02a193a5d1c234","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-03-19 06:47:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1de621b4e0ed0edbd30b8d7b51c29f4056299fc1","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 06:48:41.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"8a61d2d92fff44b13a2b47367a48404cb6e9a8f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-03-19 06:51:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dadd2ec3f592c0a99d498a7f87eaf84021cb84a3","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 06:52:56.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"9a2f28f15e32fecdf30517e914199edf02d35ed1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 07:04:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee1c4ac686df45b48a4718661e8a4ada : SUCCESS in 6m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f0da795b95994816bd03860d0b87b6e0 : SUCCESS in 7m 27s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/a5f2a22c98164d81a3990f2637c9b20f : SUCCESS in 8m 15s","accounts_in_message":[],"_revision_number":2},{"id":"083808f41acaab5f0b75f6b84965d59fe1f6e68c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-19 08:03:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"06daebe7e50b9fe7e00858ee6d85bdf562e38065","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-19 08:05:16.000000000","message":"Patch Set 2: Workflow-1\n\nWaiting for https://review.opendev.org/c/openstack/releases/+/780559","accounts_in_message":[],"_revision_number":2},{"id":"f5fce5149913bfe3a4fa785bf412c930873436e5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-22 09:35:56.000000000","message":"Patch Set 2: -Workflow\n\nWe good to go now","accounts_in_message":[],"_revision_number":2},{"id":"f53db03132767db3934f3ecf70b53fdc5c8fbb12","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 09:36:12.000000000","message":"Patch Set 2: -PTL-Approved","accounts_in_message":[],"_revision_number":2},{"id":"e64707e5c180c5cfca389d6c1c079a417661aaa8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-22 09:36:44.000000000","message":"Patch Set 2:\n\nMilestone 2 patch has been merged.","accounts_in_message":[],"_revision_number":2},{"id":"42a5b0963cac6934ce488ed8aa7973cddd0518de","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-22 09:39:01.000000000","message":"Patch Set 2:\n\nConflicts needs to be fixed first","accounts_in_message":[],"_revision_number":2},{"id":"924fa7463be2b28794c5c75999ecb1ea9eca19fe","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-22 09:39:07.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"fcdcf591d8a62382dde7477ad2c5e4f1453283ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-22 10:02:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1fa39e77aea0c92c1e90f6e20d811cbbf5079371","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 10:26:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/489ca857de824b07af40bca606755819 : SUCCESS in 8m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b28af170029048aeb03d67b9192d2229 : SUCCESS in 9m 09s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/98df25859cee4665b012eaf75d52de8c : SUCCESS in 8m 09s","accounts_in_message":[],"_revision_number":3},{"id":"138327f68655af859ec28d0c4e59490886746569","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-03-23 02:18:29.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"c30f18c328acc9dc3a434325527b819d813a137d","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 02:19:38.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"1a0c27ba112ce591034432c59c124066890bb942","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 02:32:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bb16eb67cf94c68a901262e1109a90f : SUCCESS in 8m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cd35366584974e0583839f4ac79ede6f : SUCCESS in 8m 14s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/3b28bfe88e8143c1b6c84645aeefde70 : SUCCESS in 12m 16s","accounts_in_message":[],"_revision_number":4},{"id":"4592fe3554e962fe7672b5f3ff36a4c6f00df0c8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-23 08:36:15.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"137ce329a0dbc2352d73156e3ce2f6710c441fbc","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-03-25 08:48:03.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"e61fd84b6b1506ae536bae34d0e9bee924e9059e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 08:48:53.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f01b282090ad2c42b1ac5553a67c45f26268c120","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 08:59:20.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/549da19f04a54d6da0517550cfad9ee3 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2a1cd06b0dc4df88c150f8795ddc1c8 : SUCCESS in 8m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/466993793d9e46909f79650e8d2b2939 : SUCCESS in 8m 49s","accounts_in_message":[],"_revision_number":4},{"id":"84d3e79518b14054d2938deb7e04e49b887d5308","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 08:59:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"cf11f66e55e6d0d3b2080e010f9aeec80c265599","revisions":{"d6860097ffa796f8dcd8ff14d92a3bd553644d1a":{"kind":"REWORK","_number":1,"created":"2021-03-19 06:47:32.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/01/781601/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/01/781601/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/781601/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/781601/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/781601/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/01/781601/1"}}},"commit":{"parents":[{"commit":"3e83517ff6710d93a4609c2f3523f2bce5e810d9","subject":"Merge \"Add ovn-octavia-provider cycle highlights\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3e83517ff6710d93a4609c2f3523f2bce5e810d9"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-03-19 06:32:12.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-03-19 06:40:50.000000000","tz":480},"subject":"Puppet OpenStack wallaby Milestone 3","message":"Puppet OpenStack wallaby Milestone 3\n\nChange-Id: I48bc230ab69c6b802518aed9f7fc4b4ccaf0432d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d6860097ffa796f8dcd8ff14d92a3bd553644d1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d6860097ffa796f8dcd8ff14d92a3bd553644d1a"}]},"branch":"refs/heads/master"},"8111b9a41c3887a454f19e207bbf006857e6fec9":{"kind":"REWORK","_number":2,"created":"2021-03-19 06:51:05.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/01/781601/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/01/781601/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/781601/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/781601/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/781601/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/01/781601/2"}}},"commit":{"parents":[{"commit":"3e83517ff6710d93a4609c2f3523f2bce5e810d9","subject":"Merge \"Add ovn-octavia-provider cycle highlights\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3e83517ff6710d93a4609c2f3523f2bce5e810d9"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-03-19 06:32:12.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-03-19 06:50:48.000000000","tz":480},"subject":"Puppet OpenStack wallaby Milestone 3","message":"Puppet OpenStack wallaby Milestone 3\n\nChange-Id: I48bc230ab69c6b802518aed9f7fc4b4ccaf0432d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8111b9a41c3887a454f19e207bbf006857e6fec9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8111b9a41c3887a454f19e207bbf006857e6fec9"}]},"branch":"refs/heads/master"},"d1c40ec32b9736eb1f38865e3a9b93d7d103d911":{"kind":"REWORK","_number":3,"created":"2021-03-22 10:02:06.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/01/781601/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/01/781601/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/781601/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/781601/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/781601/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/01/781601/3"}}},"commit":{"parents":[{"commit":"bf7ec4e437c3779b5d0dcceb4e4ef64ff5581f36","subject":"Merge \"Proposing Wallaby RC1 for aodh\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bf7ec4e437c3779b5d0dcceb4e4ef64ff5581f36"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-03-19 06:32:12.000000000","tz":480},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-22 10:01:44.000000000","tz":60},"subject":"Puppet OpenStack wallaby Milestone 3","message":"Puppet OpenStack wallaby Milestone 3\n\nChange-Id: I48bc230ab69c6b802518aed9f7fc4b4ccaf0432d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d1c40ec32b9736eb1f38865e3a9b93d7d103d911"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d1c40ec32b9736eb1f38865e3a9b93d7d103d911"}]},"branch":"refs/heads/master"},"cf11f66e55e6d0d3b2080e010f9aeec80c265599":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-03-23 02:18:29.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/01/781601/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/01/781601/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/781601/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/781601/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/781601/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/01/781601/4"}}},"commit":{"parents":[{"commit":"e2bc748ca2fb25fbd8179e416f605da82d4c7aa5","subject":"Merge \"Proposing Wallaby RC1 for sahara-plugin-spark\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e2bc748ca2fb25fbd8179e416f605da82d4c7aa5"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-03-19 06:32:12.000000000","tz":480},"committer":{"name":"zhongshengping","email":"chdzsp@163.com","date":"2021-03-23 02:18:29.000000000","tz":0},"subject":"Puppet OpenStack wallaby Milestone 3","message":"Puppet OpenStack wallaby Milestone 3\n\nChange-Id: I48bc230ab69c6b802518aed9f7fc4b4ccaf0432d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cf11f66e55e6d0d3b2080e010f9aeec80c265599"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cf11f66e55e6d0d3b2080e010f9aeec80c265599"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
