)]}'
{"id":"openstack%2Freleases~781640","triplet_id":"openstack%2Freleases~master~Ibc2594ab7a8ddff7f92e98b868e6b30050932c89","project":"openstack/releases","branch":"master","topic":"wallaby-rc1-deadline","hashtags":[],"change_id":"Ibc2594ab7a8ddff7f92e98b868e6b30050932c89","subject":"Proposing Wallaby RC1 for cinder","status":"MERGED","created":"2021-03-19 12:07:23.000000000","updated":"2021-03-26 10:07:04.000000000","submitted":"2021-03-26 10:07:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"781640-wallaby-rc1-deadline","meta_rev_id":"acb319e06bd37d1ddac06bb5cc1ee5fdd39c1567","_number":781640,"virtual_id_number":781640,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-26 10:06:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2021-03-26 09:55:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2021-03-26 06:18:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2021-03-25 22:15:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2021-03-26 09:55:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-03-25 22:12:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-19 12:08:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-19 12:18:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-19 12:21:04.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-03-19 12:21:08.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-03-25 22:15:54.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2021-03-26 09:55:55.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"87cf5d1808cd660fab451eb019b1159be2cb652b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-19 12:07:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9327c80f2440d71b9ff70e09039db00912befe8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 12:18:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bb6f2f2ed754b08abbc5ea6f73167c5 : SUCCESS in 6m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ac627a905cd446e6b9764ab265a40073 : SUCCESS in 5m 25s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c5444ead1b16443f971973935e718b8f : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":1},{"id":"271edccc90b3ffd2812ae98b9c0134a2b262e6fd","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-24 09:17:15.000000000","message":"Patch Set 1:\n\nAny update?","accounts_in_message":[],"_revision_number":1},{"id":"35315d6669be86ff26487956b1204cbb80b1e0b2","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-24 18:16:28.000000000","message":"Patch Set 1: Code-Review-1\n\nWe\u0027re still working on some final bugfixes.","accounts_in_message":[],"_revision_number":1},{"id":"8b4df353105ae6b58b2a640cca4c12f256a3630f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-25 22:11:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"06bc2e9a5105f6a21594de5d8bc1dfcc94edfd16","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 22:12:31.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"b7f8fe27ca39ac9768306545434e2dd9576aa1e3","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-03-25 22:15:54.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"0089641055f74899e33fa600bb4bef8e701f74a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 22:18:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c163b1526363491fbe54a3d3ddca4936 : SUCCESS in 6m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2b6da364090442d9acc049ee2453e994 : SUCCESS in 4m 49s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/3978a2306f9b4f60a057d9116e235734 : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":2},{"id":"15e54930b8d9b50046724da949729b20a3893d1c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-26 06:18:43.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3f19283a941b54a2a44f9c086d2db084e55bebf6","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-26 09:55:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLGTM:\n* version bump is OK\n* patch updated by PTL\n* validator passed\n* list-changes-results.log doesn\u0027t show any issue","accounts_in_message":[],"_revision_number":2},{"id":"363a38486a266f0820952ccf02bc4973bac3138c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 09:56:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2abd0d557a6de51453331e3d82cfa530eead7fba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 10:06:50.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/840bb7208c0d4f0f9c22dc76945a7ae7 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/679a03ab2165438d8cebc506e501dd74 : SUCCESS in 6m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4205b2464e5748e49fbac42e452b084f : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":2},{"id":"acb319e06bd37d1ddac06bb5cc1ee5fdd39c1567","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 10:07:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f2945c883934024545f0da7f91bea45957542ab7","revisions":{"b766b80ef994f60c88609d6013804a3efae4ea15":{"kind":"REWORK","_number":1,"created":"2021-03-19 12:07:23.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/40/781640/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/40/781640/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/40/781640/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/40/781640/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/40/781640/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/40/781640/1"}}},"commit":{"parents":[{"commit":"c5845e766ef8a97d4682c018b3b553786e20dd1a","subject":"Merge \"Amend Wallaby Cycle Highlight for Nova\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c5845e766ef8a97d4682c018b3b553786e20dd1a"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-19 12:07:12.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-19 12:07:12.000000000","tz":60},"subject":"Proposing Wallaby RC1 for cinder","message":"Proposing Wallaby RC1 for cinder\n\nIf you wants to wait for a specific patch to make it to the release,\nsomeone from your team can -1 the patch to have it held, or update that patch\nwith a different commit SHA.\n\nPatches with no response will be merged by Friday March 26.\n\n$ git log --oneline --no-merges 18.0.0.0b1..0dd28a16c\n6f1531ab1 [SVF]: Volume name is not validated for host\n7576313b9 Changed whitelist to allowlist in tox\nfd77dfa22 [SVF]:Storwize hyperswap volume is not deleting.\n09124fa68 Resolve SADeprecationWarning for joinedload_all\n5ad7fe926 Always constraint dependencies in tox\n0d06d497c Add support for consistency groups in the Nimble Storage driver\n81e5a9290 NFS: Fix for groups and cloning\n6ff5591fc Add virtualenv requirement to tox.ini\n0cb7322ac NetApp ONTAP: Implement FlexGroup pool\n8da91b74d API validation: Use cinder_host for services checks\nace174821 Support format info in fs type drivers\nae00a173b Add Consistency Groups support in PowerStore driver\nf328341ed Add OpenStack volume replication v2.1 support in PowerStore driver\n8687925ce API validation: Add cinder_host type to support ipv6 in manage\n886c5b4c8 [SVF] Set volume IOPS based on volume size\n79b2a4f46 Add QoS Suport for Pure Storage\n24c9065df Imported Translations from Zanata\n7d9dc8dec NetApp ONTAP: Add support for dynamic Adaptive QoS policy group creation\nfb358e45f NetApp ONTAP: Add support for QoS minimums specs\ncb02f9dfe PowerMax Driver - Offline r1 promotion fixes\n2f5885e4b [SVF]: Fixed host and group failback issues\nbb219d6b9 Pure Storage: check volumename length does not exceed maximum\n17a25eafc [DS8K]: Support revert to snapshot\n234f3358c [Pure] Add missing FC host personality support\n9f0be9994 Tests: Don\u0027t assert notifier not called\ncead51cea [Storwize] Provide IOPS based storage offering\n881f72569 PowerMax Driver - Extend replicated volume\n5e5ae0764 Hitachi: Trace REST API input/output logs\n3343ca773 Hitachi: Use get_volume_stats in the base driver\n41101463a [Pure] Fix failing consistency group tempest tests\n868646d28 Fix CI_WIKI_NAME entries\ndc938a405 Use TOX_CONSTRAINTS_FILE\nd99322ede Drop policy check failures to DEBUG\n40d93a611 Fusionstorage Cinder Driver Support OceanStor 100D Storage.(dsware)\n01c9aba1f [SVF] Update volume replication properties\n9b418e76a smbfs: set VHD UUID using volume UUID\n11185189b TOYOU:  Abandon the target parameter and  Report SAN driver options\nf8e569864 Simplify composite check strings for project personas\n0c87d14f5 Make sure we pass context objects directly to policy enforcement\n118da903b Properly handle InvalidScope exceptions\n9fc6fab79 Update secure RBAC check strings with descriptions\ne175d64ca Add ports filtering support to Dell EMC XtremIO driver\n2ae735bff Update to hacking 4.0.0\n2563dd3a1 Bump pylint to 2.6.0\n02521d19f LVM: fix _create_vg\nbc0fdbc7e vmware: Use cookiejar from oslo.vmware client directly\n698ab86a7 mypy: annotate volume manager\nf91aec586 Add encryption_key_id to volume and backup details\n203eb2cf8 Correct format string in error message\n01f010011 Tests: Add another mock psutil in quobyte tests\n3ce9d773a Label temporary files created by image_utils\nd9c90cf48 tox mypy: Call mypywrap.sh directly\nacfb3361c Hitachi: Wait until the volume can be deleted\nf04d905a6 [RBD] Fix snapshot backup name\n4ba6664de RBD: Change rbd_exclusive_cinder_pool\u0027s default\n0a11cac11 Exception type is not iterable so should change it to string type\nd729522ac [IBM DS8000]: Fix compatability issue in get_host\n00ad39a60 PowerMax Driver - Fix assign SRP during promotion retype\nceb502254 Tests: Move glance_stubs to glance dir\nb62f250bc PowerMax Driver - Port status check\ne1f517927 Remove install unnecessary packages\n61d612e64 PowerMax Docs - Incorrect property set on volume group\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: Ibc2594ab7a8ddff7f92e98b868e6b30050932c89\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b766b80ef994f60c88609d6013804a3efae4ea15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b766b80ef994f60c88609d6013804a3efae4ea15"}]},"branch":"refs/heads/master"},"f2945c883934024545f0da7f91bea45957542ab7":{"kind":"REWORK","_number":2,"created":"2021-03-25 22:11:26.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/40/781640/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/40/781640/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/40/781640/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/40/781640/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/40/781640/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/40/781640/2"}}},"commit":{"parents":[{"commit":"71acccba6e3a1c198d1771d6d718d0e3141061c7","subject":"Merge \"Proposing Wallaby RC1 for storlets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/71acccba6e3a1c198d1771d6d718d0e3141061c7"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-19 12:07:12.000000000","tz":60},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2021-03-25 22:11:20.000000000","tz":-240},"subject":"Proposing Wallaby RC1 for cinder","message":"Proposing Wallaby RC1 for cinder\n\nRelease cinder 18.0.0.0rc1 and cut the stable/wallaby branch.\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: Ibc2594ab7a8ddff7f92e98b868e6b30050932c89\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f2945c883934024545f0da7f91bea45957542ab7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f2945c883934024545f0da7f91bea45957542ab7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
