)]}'
{"id":"openstack%2Freleases~781656","triplet_id":"openstack%2Freleases~master~Ifdf0474c200ad3a56426b63dcaa18b321279a17f","project":"openstack/releases","branch":"master","topic":"wallaby-rc1-deadline","hashtags":[],"change_id":"Ifdf0474c200ad3a56426b63dcaa18b321279a17f","subject":"Proposing Wallaby RC1 for heat","status":"MERGED","created":"2021-03-19 12:32:11.000000000","updated":"2021-03-26 15:14:22.000000000","submitted":"2021-03-26 15:14:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"781656-wallaby-rc1-deadline","meta_rev_id":"3fe1fe89bb9eaf18860af1eedbcf30bc9ecac600","_number":781656,"virtual_id_number":781656,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-26 15:14:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"recommended":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"all":[{"value":1,"date":"2021-03-26 11:47:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":2,"date":"2021-03-26 15:06:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2021-03-26 13:14:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":1,"date":"2021-03-26 15:06:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-03-26 11:55:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-19 12:34:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-19 12:43:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-19 13:23:23.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"state":"REVIEWER"},{"updated":"2021-03-26 15:06:48.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"d4237026462e0871d888d72a969faeeca723ffdd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-19 12:32:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c785a4c207029128ab85ab4b8403868e5713b00d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 12:43:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d6188398359460c81e7c9bc2bf49c16 : SUCCESS in 7m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b3756a83acb24870877475c450ba8043 : SUCCESS in 5m 53s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/0c719f8f6245482da52167753e13461e : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":1},{"id":"6b4bd86233c9dc16fb2ff87ffca0803e4e0678cf","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-24 09:21:15.000000000","message":"Patch Set 1:\n\nAny update?","accounts_in_message":[],"_revision_number":1},{"id":"431e112e8f495af22dd8c33da8cda0651f2875e3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-25 12:50:36.000000000","message":"Patch Set 1:\n\nHello team,\n\nDeadline is tomorrow, please have a look before. If no response is provided until tomorrow we\u0027ll force it.\n\nThank you for your understanding.","accounts_in_message":[],"_revision_number":1},{"id":"736830b6e4b706a789b71fd23fc9e14d4664ef4a","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2021-03-26 11:47:12.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b29fd026f41165f24517375324f86229f1919c92","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 11:55:32.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"e4afa6d579c9f8055826d758556566c35e7c3cef","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-26 13:14:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"825fb391e68f4d096fea05f4703fe37c530811f9","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-26 15:06:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM:\n* version bump is OK\n* approved by PTL\n* test jobs logs don\u0027t show any relevant issue","accounts_in_message":[],"_revision_number":1},{"id":"1b7e7bffa44cc9d18e45ac8ac5374eae63ccc5eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 15:07:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e1ba0bbac485df33de4f974afca219a8f70e4d9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 15:14:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a73e2365bc6b470ab4cc114b5fff59cf : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7840dc10a08d4ec98c6bda1f2b86f1f7 : SUCCESS in 6m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5de581fcaad74811bd5e6d3bc66a3d7d : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":1},{"id":"3fe1fe89bb9eaf18860af1eedbcf30bc9ecac600","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 15:14:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"21fa5247781a4f24ee6b57e22bfd755a396ab3a1","revisions":{"21fa5247781a4f24ee6b57e22bfd755a396ab3a1":{"kind":"REWORK","_number":1,"created":"2021-03-19 12:32:11.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/56/781656/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/56/781656/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/781656/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/781656/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/781656/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/56/781656/1"}}},"commit":{"parents":[{"commit":"c5845e766ef8a97d4682c018b3b553786e20dd1a","subject":"Merge \"Amend Wallaby Cycle Highlight for Nova\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c5845e766ef8a97d4682c018b3b553786e20dd1a"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-19 12:32:00.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-19 12:32:00.000000000","tz":60},"subject":"Proposing Wallaby RC1 for heat","message":"Proposing Wallaby RC1 for heat\n\nIf you wants to wait for a specific patch to make it to the release,\nsomeone from your team can -1 the patch to have it held, or update that patch\nwith a different commit SHA.\n\nPatches with no response will be merged by Friday March 26.\n\n$ git log --oneline --no-merges 15.0.0..af0dd44da\naf7f8e380 Add separate policy for updates with no changes\n7230082f8 Fix avoid deprecation warnings from policy\n1dcbac064 Add segments attribute to OS::Neutron::ProviderNet\n0b593c218 Allow deleting Server/DeployedServer without swift/zaqar\n93594c30e Implement secure RBAC\n8daa7e938 Allow using database configs on db retry\n45750c603 Use neutron client for server.addresses\nc37a72e47 Fix iter bug with SQlAlchemy 1.3.23\n79fb6509a Don\u0027t create deployment when changing transport\n46d31c729 Fix typo in documentation\n164daa691 Remove references to basestring\n57e975409 Switch to collections.abc.*\n9292264aa Allow creating a Neutron port without fixed ips\n6a9c35d89 Move testing to Fedora 33\n042345703 Bump requirements to support secure RBAC effort\nf4ab9d0bc Fix multiple gate issues\nf616f43ce Update doc8 version\n823a30322 Imported Translations from Zanata\n7e12f6984 Imported Translations from Zanata\n4370af157 Align lower-constraints for new pip\n555c49b89 Add openstack-python3-wallaby-jobs-arm64 job\ndefb617ce Fix args for initializing trove client\n896166161 Restore to vote for grenade job\n61e53c4ec Replace Fedora test image with F32\n041e500ea Imported Translations from Zanata\naf53e4b4c Fix invalid argument formatting in exception messages\nd7d2ce077 Adds backups and backups_gigabytes parameters to cinder quota resource\n54956ada8 Imported Translations from Zanata\na5e3ebe76 Use TOX_CONSTRAINTS_FILE\n960428ca7 Remove deprecated tail_log function\nc17eef90d Always reset CONF when starting the wsgi app\nd2be2f9cf Add test cases to check user{domain} pattern.\nd8efcd178 Regenerate trust when update with different user\n762879a14 [S2007220]: Added more image properties to web_image\nba7d105c8 Adopt grenade to multinode setting\nc6ae01c30 Use assertCountEqual instead of assertItemsEqual\ne39b03e8e Add Vitrage template resource\n2c831ae25 Use TOX_CONSTRAINTS_FILE\n32413431e Imported Translations from Zanata\n6fe5230a3 Add more information for tests\naab68ced6 Update get_hosts to use available API\nb068f45ad As of nova microversion 2.57, personality is deprecated\n20b599cdf Add SOURCE_IP_PORT to LB_ALGORITHM allowed values\nfc0aadf66 Imported Translations from Zanata\n85b693c31 Remove gabbi tempest plugin\ne5b00ddfa Delete the default value of domain in role creation\n253aff806 Remove grenade-heat and heat-functional-non-apache\n2afd44dfe Imported Translations from Zanata\n71a9c3d69 Modify \u0027if\u0027 Macro to allow optional properties\n674a62ae9 Add template version 2021-04-16 for Wallaby\n8cce06a75 Increase max_stacks_per_tenant for tests\nf0d7425db Imported Translations from Zanata\nf419923a0 Configure swap as before\ne9671342a Dep\u0027s should be restricted by upper-constraints\nb52da6a7c Delete default security groups created by stack.\na78ddafde Use err.errno to access errno for socket errors\nd9ed15907 Add Python3 wallaby unit tests\n2a82f17b1 Update master for stable/victoria\n539b2a4c4 Net attr: Sort segments without name attribute first\n3709d0887 Mark support for python3.8\nf0c7508ab Switch from unittest2 compat methods to Python 3.x methods\ne79e76dad S2007837: Fixes domain_id parameter in the user find call\nd43bb063b Fix pygments style\n4deef7e72 Allow null values to be returned from Macros\n4ddc58b79 Refactor Properties._get_property_value()\n6b3d15d67 Optimise resource type listing\nf8ef790e9 Use resource_id instead of phy rsrc name in InstanceGroup\n0777837c1 Ignore old \u0027vN-branch\u0027 tags when scanning for release notes\n9a4012ba9 Use set instead of list of server_deferred_statuses\n2cad99202 API ref: Update sample output of resource type APIs\ndbe7ed82a Fix duplicated words issue like \"can be be used\"\n049c962f1 Workaround client race in legacy nested stack delete\n6b7d64988 Remove handling for client status races\nc5fdc0884 Allow turning off max_stacks limit\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: Ifdf0474c200ad3a56426b63dcaa18b321279a17f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/21fa5247781a4f24ee6b57e22bfd755a396ab3a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/21fa5247781a4f24ee6b57e22bfd755a396ab3a1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
