)]}'
{"id":"openstack%2Freleases~781670","triplet_id":"openstack%2Freleases~master~I0b1bbdf182676ccde286c9624b6423b6c4dfa82c","project":"openstack/releases","branch":"master","topic":"wallaby-rc1-deadline","hashtags":[],"change_id":"I0b1bbdf182676ccde286c9624b6423b6c4dfa82c","subject":"Proposing Wallaby RC1 for magnum","status":"MERGED","created":"2021-03-19 12:52:06.000000000","updated":"2021-03-26 10:19:37.000000000","submitted":"2021-03-26 10:19:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"781670-wallaby-rc1-deadline","meta_rev_id":"bbde88b58e92a8f1f5d617a8b8a0a383073bca39","_number":781670,"virtual_id_number":781670,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-26 10:19:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"recommended":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"all":[{"value":1,"date":"2021-03-25 21:09:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"value":2,"date":"2021-03-26 10:01:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2021-03-26 06:19:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"value":1,"date":"2021-03-26 10:01:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-03-25 21:10:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-19 12:59:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-19 13:18:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-19 13:20:48.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"state":"REVIEWER"},{"updated":"2021-03-26 10:01:36.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"6648ba23378ecb80fbc5e39c438932d3a3134d5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-19 12:52:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f743d1f48ee3a18ad7e5057a024efd2616d58c7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 13:18:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/284c88b28ab045d6b528907e744110c4 : SUCCESS in 7m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d2b753c0606a4c23b4a5df6a6f2c2cfb : SUCCESS in 5m 19s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/044b8c5efa764db6b73f6a88f1ac761a : SUCCESS in 6m 15s","accounts_in_message":[],"_revision_number":1},{"id":"010e4ecbd47a933a955cd0c924f11c49c33f18bd","author":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"date":"2021-03-19 15:30:39.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nACK will te","accounts_in_message":[],"_revision_number":1},{"id":"6e4d961f1f6af4ae25b296e295a4b87b979f185f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-23 08:54:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0a830e4eff638b44a94dcc865e8b9f59eccdd2b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 09:07:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4715527c91d24197b5fc24639da30f7a : SUCCESS in 6m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7391e48ea80b4f1b91e4b36e3dcfb5d1 : SUCCESS in 5m 15s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/97773d734ec144079f7b7dbf4aac9bc2 : SUCCESS in 7m 34s","accounts_in_message":[],"_revision_number":2},{"id":"6277c0a8cafe708295609a1f9e8daaa31b564a2a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-24 09:31:14.000000000","message":"Patch Set 2:\n\nAny update?","accounts_in_message":[],"_revision_number":2},{"id":"17c8b1d25037b7c020027bdf0efb9e025e30b4ef","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-25 13:06:51.000000000","message":"Patch Set 2:\n\nHello team,\n\nFriendly reminder, deadline is tomorrow, please have a look before. If no response is provided until tomorrow we\u0027ll force it.\n\nThank you for your understanding.","accounts_in_message":[],"_revision_number":2},{"id":"424673774071ab9e6dca8adec7a25af588197544","author":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"date":"2021-03-25 13:35:40.000000000","message":"Patch Set 2:\n\nThank you for the reminders, we try to push one more patch. If we don\u0027t manage today (in Europe) we can merge this and then we can cherry-pick. I will leave another comment today.","accounts_in_message":[],"_revision_number":2},{"id":"842ebd461915865aa4ed5c67cedc702928a8b40e","author":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"date":"2021-03-25 21:09:03.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"804a2d363d2dafeaf99c44a0271b9a982385b3ae","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 21:10:06.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"2a86de050ea33e158cc43c46f3bb4d1d6856edcd","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-26 06:19:42.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"cc9e08cf30347049b5c384995c2fc1ad928668c1","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-26 10:01:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLGTM:\n* version bump is OK\n* approved by Spyros at the end of Thursday\n* validator passed\n* list-changes-results.log looks OK","accounts_in_message":[],"_revision_number":2},{"id":"42bc233b62b68d83a36698a8145a02e2c082e040","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 10:02:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6109f109be75562a938df5cbbf159cbba54336f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 10:19:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7c3d848923c4abdae3e6c162df39657 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25d58debc93a468d89ab7ac46231fac9 : SUCCESS in 8m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/24c3a07319dc45fba38f23d75e8f3562 : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":2},{"id":"bbde88b58e92a8f1f5d617a8b8a0a383073bca39","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 10:19:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"95fdc58565a5b71318ce7aafec08b1217e77dcb1","revisions":{"30b4c9080dece9e3ada0eb1e4ec81db43f116661":{"kind":"REWORK","_number":1,"created":"2021-03-19 12:52:06.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/70/781670/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/70/781670/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/781670/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/781670/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/781670/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/70/781670/1"}}},"commit":{"parents":[{"commit":"c5845e766ef8a97d4682c018b3b553786e20dd1a","subject":"Merge \"Amend Wallaby Cycle Highlight for Nova\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c5845e766ef8a97d4682c018b3b553786e20dd1a"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-19 12:51:53.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-19 12:51:53.000000000","tz":60},"subject":"Proposing Wallaby RC1 for magnum","message":"Proposing Wallaby RC1 for magnum\n\nIf you wants to wait for a specific patch to make it to the release,\nsomeone from your team can -1 the patch to have it held, or update that patch\nwith a different commit SHA.\n\nPatches with no response will be merged by Friday March 26.\n\n$ git log --oneline --no-merges 11.0.0..42f8c97b\n00f8aa5d Fix debug logging during cluster upgrade\nb336bb34 Re-factored rpc serializer\n3b7a33eb [goal] Deprecate the JSON formatted policy file\nff575939 [hca] Only build/push stable images if unpublished\n1f33b6c1 Build autoscaler 1.20\n0e6d1789 Only allow zero node count from microversion 1.10\n439548e3 Fix ostree_* upgrade\nafce5d9c [doc] Replace Atomic/CoreOS with Fedora CoreOS\n332e2b6f Add CT tags field to the database and API\na3d8b4fe 4. Update cluster monitoring documentation\nea64468a 3. Configure monitoring apps path based endpoints\naec5d469 2. Add persistency for grafana dashboards\n37497ccf 1. Configurable prometheus monitoring persistent storage\nd21d7d50 Update API version history doc\nbcf771bc Do not create constraints for boolean fields\nf46923cc Allow nodegroups with node_count equal to 0\n672b1195 Re-use transport for rpc server\nfd79dd4f Switch to uwsgi and enable named uri\n1b72456e k8s: Do not use insecure api port\n987c9d80 Re-use transport for rpc calls\n7bfd7519 [k8s-fcos] Fix insecure registry\n8018bf91 Fix cluster deletion when load balancers don\u0027t exist\n9b8f1555 Update docs for cluster resource\nd11f4e83 Make kubelet and kube-proxy use the secure port\n1843ad1b Drop lower constraints testing\nacc70848 Fix validation for master_lb_enabled\n8bdf0e76 Update containerd version and tarball URL\n55d3dada Imported Translations from Zanata\nfade2451 [k8s] Fix default admission controller\n94737878 Fix gate - update lower-constraints\n333331b7 Update helm charts origin repository\nbc051d52 CI: Install debianutils and vim\n327c12fc Add image prefix for grafana images\n34f65582 Use kube_master_ip for monitoring when no floating ip is used\n56583ac8 Fix Cinder CSI\n2c93427d k8s-fcos: Source bashrc for clusterconfig\n9adfc443 Fix misquoted comment\ne17e6b8a Revert \"Fix Cinder CSI\"\n2b5a4512 ci: Update dockerhub password\n290d60a0 Fix Cinder CSI\nf5cf6b95 Fix database migrations\na837b5c0 Update default k8s admission controller list\n1cdc0628 [fix] Sync nodegroup status before delete_complete\nd64fe4b7 Update master for stable/victoria\nc84653cd Remove cloud-config from k8s worker node\n385bc970 Update default values for docker nofile and vm.max_map_count\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I0b1bbdf182676ccde286c9624b6423b6c4dfa82c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/30b4c9080dece9e3ada0eb1e4ec81db43f116661"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/30b4c9080dece9e3ada0eb1e4ec81db43f116661"}]},"branch":"refs/heads/master"},"95fdc58565a5b71318ce7aafec08b1217e77dcb1":{"kind":"REWORK","_number":2,"created":"2021-03-23 08:54:17.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/70/781670/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/70/781670/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/781670/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/781670/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/781670/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/70/781670/2"}}},"commit":{"parents":[{"commit":"e2bc748ca2fb25fbd8179e416f605da82d4c7aa5","subject":"Merge \"Proposing Wallaby RC1 for sahara-plugin-spark\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e2bc748ca2fb25fbd8179e416f605da82d4c7aa5"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-19 12:51:53.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-23 08:53:56.000000000","tz":60},"subject":"Proposing Wallaby RC1 for magnum","message":"Proposing Wallaby RC1 for magnum\n\nIf you wants to wait for a specific patch to make it to the release,\nsomeone from your team can -1 the patch to have it held, or update that patch\nwith a different commit SHA.\n\nPatches with no response will be merged by Friday March 26.\n\n$ git log --oneline --no-merges 11.0.0..42f8c97b\n00f8aa5d Fix debug logging during cluster upgrade\nb336bb34 Re-factored rpc serializer\n3b7a33eb [goal] Deprecate the JSON formatted policy file\nff575939 [hca] Only build/push stable images if unpublished\n1f33b6c1 Build autoscaler 1.20\n0e6d1789 Only allow zero node count from microversion 1.10\n439548e3 Fix ostree_* upgrade\nafce5d9c [doc] Replace Atomic/CoreOS with Fedora CoreOS\n332e2b6f Add CT tags field to the database and API\na3d8b4fe 4. Update cluster monitoring documentation\nea64468a 3. Configure monitoring apps path based endpoints\naec5d469 2. Add persistency for grafana dashboards\n37497ccf 1. Configurable prometheus monitoring persistent storage\nd21d7d50 Update API version history doc\nbcf771bc Do not create constraints for boolean fields\nf46923cc Allow nodegroups with node_count equal to 0\n672b1195 Re-use transport for rpc server\nfd79dd4f Switch to uwsgi and enable named uri\n1b72456e k8s: Do not use insecure api port\n987c9d80 Re-use transport for rpc calls\n7bfd7519 [k8s-fcos] Fix insecure registry\n8018bf91 Fix cluster deletion when load balancers don\u0027t exist\n9b8f1555 Update docs for cluster resource\nd11f4e83 Make kubelet and kube-proxy use the secure port\n1843ad1b Drop lower constraints testing\nacc70848 Fix validation for master_lb_enabled\n8bdf0e76 Update containerd version and tarball URL\n55d3dada Imported Translations from Zanata\nfade2451 [k8s] Fix default admission controller\n94737878 Fix gate - update lower-constraints\n333331b7 Update helm charts origin repository\nbc051d52 CI: Install debianutils and vim\n327c12fc Add image prefix for grafana images\n34f65582 Use kube_master_ip for monitoring when no floating ip is used\n56583ac8 Fix Cinder CSI\n2c93427d k8s-fcos: Source bashrc for clusterconfig\n9adfc443 Fix misquoted comment\ne17e6b8a Revert \"Fix Cinder CSI\"\n2b5a4512 ci: Update dockerhub password\n290d60a0 Fix Cinder CSI\nf5cf6b95 Fix database migrations\na837b5c0 Update default k8s admission controller list\n1cdc0628 [fix] Sync nodegroup status before delete_complete\nd64fe4b7 Update master for stable/victoria\nc84653cd Remove cloud-config from k8s worker node\n385bc970 Update default values for docker nofile and vm.max_map_count\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I0b1bbdf182676ccde286c9624b6423b6c4dfa82c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/95fdc58565a5b71318ce7aafec08b1217e77dcb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/95fdc58565a5b71318ce7aafec08b1217e77dcb1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
