)]}'
{"id":"openstack%2Freleases~781673","triplet_id":"openstack%2Freleases~master~Iac76024f04bdb06d7a98203ace334ed50bc52743","project":"openstack/releases","branch":"master","topic":"wallaby-rc1-deadline","hashtags":[],"change_id":"Iac76024f04bdb06d7a98203ace334ed50bc52743","subject":"Proposing Wallaby RC1 for manila","status":"MERGED","created":"2021-03-19 12:56:58.000000000","updated":"2021-03-26 18:47:51.000000000","submitted":"2021-03-26 18:47:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"781673-wallaby-rc1-deadline","meta_rev_id":"dbdb8027717a46a1e6f4262a2c156988b6e4dfd4","_number":781673,"virtual_id_number":781673,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-26 18:47:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"recommended":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2021-03-26 18:08:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2021-03-26 18:32:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2021-03-26 18:17:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2021-03-26 18:32:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-03-26 18:07:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"}],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-19 12:59:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-19 13:15:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-19 13:21:15.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2021-03-19 13:21:17.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2021-03-26 18:32:39.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"cb129d7f9155835c23920a61a37848337dc2df04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-19 12:56:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1359ccc9229aa8da88af9b2fac88ca4b1d23e166","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 13:15:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/009699b00cd044f28e54153cff16f59e : SUCCESS in 7m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/79c13bf73c0049fbb9b7330234aa806e : SUCCESS in 5m 47s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/712efde0b0244281b15abd2f3e303568 : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":1},{"id":"2875235c1961ba2d09dcf5e8b1b1852cd5823951","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-24 09:19:22.000000000","message":"Patch Set 1:\n\nAny update?","accounts_in_message":[],"_revision_number":1},{"id":"444d3ed232a2bccc1f605abe190de92323abafc7","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-03-24 23:30:22.000000000","message":"Patch Set 1: Code-Review-1\n\n\u003e Patch Set 1:\n\u003e \n\u003e Any update?\n\nThanks for checking here Hervé. Nope, this isn\u0027t ready yet. I\u0027ll propose an update with more changes soon...","accounts_in_message":[],"_revision_number":1},{"id":"2bed05b24d790f5909ab056eb5a3109a40a17660","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-03-26 18:06:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9e63e4d35a7c2924d0846b3d1379bd180fd83284","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 18:07:55.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"834484c778399569de2349831adbccacb783cb52","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-03-26 18:08:21.000000000","message":"Patch Set 2: Code-Review+1\n\nShip it :)","accounts_in_message":[],"_revision_number":2},{"id":"9043328df33e12d376c82f850ed646d6a3cdb7c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 18:15:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d47af49f4ebe4c7a82ea2a706956f5b8 : SUCCESS in 7m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5b12dd427e8741b4bb2b2c088b73421d : SUCCESS in 4m 33s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9bd609a1e87c4c74baa737452162fcf1 : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":2},{"id":"f6eaa90b5e89ea94f6d908267a888fd5bb20026e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-26 18:17:24.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"458d9573e691191832c1eebd3b97fae7b2def40f","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-26 18:32:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLGTM:\n* version bump is OK\n* patch updated by PTL\n* test jobs logs don\u0027t show any issue","accounts_in_message":[],"_revision_number":2},{"id":"ae6c92d5476127a7105827ff93fcfee219970cd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 18:32:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3e1e278bfddb3ce79a2573a6f55d6576b745323a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 18:47:50.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f72825b6b12943a5a776841820e2a1a3 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6d35d35a37b4ad18051293aa29a2ebc : SUCCESS in 6m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2a05b839a7cb4145a703ca0c809c1d25 : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":2},{"id":"dbdb8027717a46a1e6f4262a2c156988b6e4dfd4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 18:47:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0859d5569f4c9700ecf44712736ec84cc16a687e","revisions":{"5f708fd1ae32bdd72c8884378389f7c0750216bf":{"kind":"REWORK","_number":1,"created":"2021-03-19 12:56:58.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/73/781673/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/73/781673/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/781673/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/781673/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/781673/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/73/781673/1"}}},"commit":{"parents":[{"commit":"c5845e766ef8a97d4682c018b3b553786e20dd1a","subject":"Merge \"Amend Wallaby Cycle Highlight for Nova\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c5845e766ef8a97d4682c018b3b553786e20dd1a"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-19 12:56:37.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-19 12:56:37.000000000","tz":60},"subject":"Proposing Wallaby RC1 for manila","message":"Proposing Wallaby RC1 for manila\n\nIf you wants to wait for a specific patch to make it to the release,\nsomeone from your team can -1 the patch to have it held, or update that patch\nwith a different commit SHA.\n\nPatches with no response will be merged by Friday March 26.\n\n$ git log --oneline --no-merges 11.0.0..398c849a\n6981031e Fix share server lookup\nb3c20c01 Fix inconsistent ordering caused by low datetime precision.\nff91db3e [NetApp] Implement security service update\n733d6218 Add security service update support to the container driver\n2bc27c56 Add security service update for in-use share networks\n4c81cc4c [ci] Part 2: Temporarily set docs job to non-voting\n470ecda8 [ci] Temporarily set docs job to non-voting\n0b04d8d6 [NetApp] Add support for FPolicy native mode\nf1ed7c3c Implement secure RBAC for shares\nfcd559f2 Implement secure RBAC for share replicas\n9243d994 Change RBAC for share group snapshots\nce55a993 Implement secure RBAC for share snapshots\n8943e57e [NetApp] Fix security service configuration for LDAP servers\n904089f3 Implement secure RBAC for share type extra spec\n3388e9ae Implement secure RBAC for share snapshot instances\n00452939 Add config option to set per_share_size_limit.\n81cbc239 Implement secure RBAC for share group type specs\n3ecad318 Implement secure RBAC for share network subnets\n19d182c4 Implement secure RBAC for share networks\n2fb871cb Add share server limits\na7eb8257 Implement secure RBAC for share group types\n18d96c14 Implement secure RBAC for share groups\n26463657 Implement secure RBAC for share instances\nf6bf1e43 Implement secure RBAC for group snapshots\n498e5ac7 Implement secure RBAC for share types\n62119602 Implement secure RBAC for share servers\n0bb780e8 Implement secure RBAC for share snapshot instance export locations\na017d238 Implement secure RBAC for share snapshot locations\nf7fbe983 Implement secure RBAC for share replica locations\nb1dacdab Implement secure RBAC for share instance export location\n3bedb15b Fixed some syntax errors\n4705a7d5 Adding Zadara Manila driver\n74415f6d Adds support min/max share size limited by share_type\n472277c0 Implement secure RBAC for share locations\nf5950553 Implement secure RBAC for access rule metadata\n72de265c Implement secure RBAC for share access rules\n8d0b6acf Implement secure RBAC for services\nd7dd466a Implement secure RBAC for security services\na13ff5d5 Fix traceback in scheduler-stats API\nfc0f669d RBAC tightening for share access rule\n85539629 Clean up some policy code\n00133f93 Drop \"system_scope\" from context constructor\n3b0358a1 Doc: Add profiler support documentation\nfbc2b4d2 Integrate OSprofiler and Manila\n1269a866 [NetApp] Fix deepcopy of sqlalchemy objects inside the driver\ne3e0486d Remove deprecated public share policies\nd487c2db Fix unit tests to use requests\nc97f89f6 Catch up to changes in RequestContext\n0d8f415e tox miniversion update to 3.1.1\nde293981 [doc] remove project_id from api endpoints\n27f01aaa [devstack] Setup a \"shared-file-system\" service\n9efcb475 [devstack] create endpoint without project_id\n263d5438 Advertise v2 API routes without project_id\n1c17ca43 Enable healthcheck middleware\nda3ab2cf [Native CephFS] Add messages for async ACL ops\n3831e550 remove usage of six library from unit tests\n6fb2e851 remove usage of six library\nbe794fb1 Remove backend deprecated variables\naa62a809 add additional driver data to migration get progress view\nba3275e9 Adds missing caps for mgr in cephfs configuration\n68ad590c [NetApp] Fix kerberos security service issues\ndcdc1a98 Update doc for manila-service-image download\nd7407afd Update manila and Ceph matrix support doc\n6bb02137 Fix missing group and group_snapshots in absolute limits\n7f72147f Implement secure RBAC for storage pool statistics\n46f65eaf Implement secure RBAC for quotas\n5011b671 Implement secure RBAC for quota classes\n9ef2156c Implement secure RBAC for messages\na43911d7 Implement secure RBAC for storage availability zones\nfbc4ecb6 Always use new_size when extend\nd7d83a3f Uses local variable with ipv4 gateway config\n3fa5cf4e [Doc] Add admin doc for share group\n281b7dd1 Add developer reference to share server migration\nbab89906 Delete unavailable py2 package\n0cc7cbc3 [goal] Deprecate the JSON formatted policy file\n39734a70 Stop manila when unstacking\nc0ee8848 Update requirements\n8d1dcb1e Updates to support in Ceph local.conf sample\n8bef5e25 Switch from base64 encodestring to encodebytes\n914d8737 [glusterfs] don\u0027t reinit volume list on deletion\n2a8aaf43 Introduce base personas for secure RBAC\na10927c6 Bump oslo.log version to 4.3.0\nc8087fa0 Docs: NFS-Ganesha and dbus\n5e01d9e2 [NetApp] Support NFS shares on windows\n70f7859f [NetApp] Fix hard-coded CA cert path for SSL\n8a691d86 Retry unmount operation on the LVM driver\n5e8b2a47 Replace deprecated UPPER_CONSTRAINTS_FILE variable\ne3fea147 Retry unmount operation on the ZFSOnLinux driver\n136a8993 Update share replicas export location API ref\n074b0835 fix reno file location and indention\n55ef4999 Remove the unused coding style modules\n21c731e7 Move shares filtering to database layer\n1b5771ef Fix logic that determines a share exists before manage\ncc33a4b2 disable M325\n6d8c99a7 Set Victoria max version in the api doc\n2e27c718 Fix \u0027update_share_replica\u0027 doesn\u0027t provide share server model to drivers\n3ca9c767 Add share server migration admin documentation\n0c4d1e33 [NetApp] Fix access rules for CIFS promoted replica\n52c926e3 Add Python3 wallaby unit tests\n44fe2e5e Update master for stable/victoria\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: Iac76024f04bdb06d7a98203ace334ed50bc52743\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5f708fd1ae32bdd72c8884378389f7c0750216bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5f708fd1ae32bdd72c8884378389f7c0750216bf"}]},"branch":"refs/heads/master"},"0859d5569f4c9700ecf44712736ec84cc16a687e":{"kind":"REWORK","_number":2,"created":"2021-03-26 18:06:54.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/73/781673/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/73/781673/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/781673/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/781673/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/781673/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/73/781673/2"}}},"commit":{"parents":[{"commit":"c5845e766ef8a97d4682c018b3b553786e20dd1a","subject":"Merge \"Amend Wallaby Cycle Highlight for Nova\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c5845e766ef8a97d4682c018b3b553786e20dd1a"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-19 12:56:37.000000000","tz":60},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-03-26 18:06:32.000000000","tz":-420},"subject":"Proposing Wallaby RC1 for manila","message":"Proposing Wallaby RC1 for manila\n\n$ git log --oneline --no-merges 11.0.0..4c7d4b\n7a99c6a1 Add release note for secure rbac work\n02cc5c67 Set \"context_is_admin\" to system scope admin roles\n3e3e7911 Bump RPC version for the scheduler rpc call\n7f0737b7 [NetApp] Follow up fixes for update security service\n643083ee Support host assisted share migration for Zadara manila driver\n3d65f46d Add exception for insufficient privileges when using security services\n268686c4 Fix wrong totalcount returned by share listing query\n5c812643 Add create share from snapshot in CephFS\na8307109 Update cephfs drivers to use ceph-mgr client\n6f68d7e6 Move params from DocumentedRule to DeprecatedRule\n75acc514 Remove unused model properties and increase unit test coverage\na8e25b85 Put ensure_share into thread pool to speed up the startup of share service\n73d0a74f fix unmange share with manage_error status will lead to quota error\n163bda5f Disable some policy warnings\n25862526 Fix generic share resize with 0.0.0.0/24 access\nf24fff95 Filter storage protocol in the scheduler\nb2e17978 [NetApp] Implement cached status pool\n6981031e Fix share server lookup\nb3c20c01 Fix inconsistent ordering caused by low datetime precision.\n70bb650e scheduler should ignore earlier time service capabilities\nff91db3e [NetApp] Implement security service update\n733d6218 Add security service update support to the container driver\n2bc27c56 Add security service update for in-use share networks\n4c81cc4c [ci] Part 2: Temporarily set docs job to non-voting\n470ecda8 [ci] Temporarily set docs job to non-voting\n0b04d8d6 [NetApp] Add support for FPolicy native mode\nf1ed7c3c Implement secure RBAC for shares\nfcd559f2 Implement secure RBAC for share replicas\n9243d994 Change RBAC for share group snapshots\nce55a993 Implement secure RBAC for share snapshots\n8943e57e [NetApp] Fix security service configuration for LDAP servers\n904089f3 Implement secure RBAC for share type extra spec\n3388e9ae Implement secure RBAC for share snapshot instances\n00452939 Add config option to set per_share_size_limit.\n81cbc239 Implement secure RBAC for share group type specs\n3ecad318 Implement secure RBAC for share network subnets\n19d182c4 Implement secure RBAC for share networks\n2fb871cb Add share server limits\na7eb8257 Implement secure RBAC for share group types\n18d96c14 Implement secure RBAC for share groups\n26463657 Implement secure RBAC for share instances\nf6bf1e43 Implement secure RBAC for group snapshots\n498e5ac7 Implement secure RBAC for share types\n62119602 Implement secure RBAC for share servers\n0bb780e8 Implement secure RBAC for share snapshot instance export locations\na017d238 Implement secure RBAC for share snapshot locations\nf7fbe983 Implement secure RBAC for share replica locations\nb1dacdab Implement secure RBAC for share instance export location\n3bedb15b Fixed some syntax errors\n4705a7d5 Adding Zadara Manila driver\n74415f6d Adds support min/max share size limited by share_type\n472277c0 Implement secure RBAC for share locations\nf5950553 Implement secure RBAC for access rule metadata\n72de265c Implement secure RBAC for share access rules\n8d0b6acf Implement secure RBAC for services\nd7dd466a Implement secure RBAC for security services\na13ff5d5 Fix traceback in scheduler-stats API\nfc0f669d RBAC tightening for share access rule\n85539629 Clean up some policy code\n00133f93 Drop \"system_scope\" from context constructor\n3b0358a1 Doc: Add profiler support documentation\nfbc2b4d2 Integrate OSprofiler and Manila\n1269a866 [NetApp] Fix deepcopy of sqlalchemy objects inside the driver\ne3e0486d Remove deprecated public share policies\nd487c2db Fix unit tests to use requests\nc97f89f6 Catch up to changes in RequestContext\n2eab5de8 [api-ref] Fix incorrect parameters\n0d8f415e tox miniversion update to 3.1.1\nde293981 [doc] remove project_id from api endpoints\n27f01aaa [devstack] Setup a \"shared-file-system\" service\n9efcb475 [devstack] create endpoint without project_id\n263d5438 Advertise v2 API routes without project_id\n1c17ca43 Enable healthcheck middleware\nda3ab2cf [Native CephFS] Add messages for async ACL ops\n3831e550 remove usage of six library from unit tests\n6fb2e851 remove usage of six library\nbe794fb1 Remove backend deprecated variables\naa62a809 add additional driver data to migration get progress view\nba3275e9 Adds missing caps for mgr in cephfs configuration\n68ad590c [NetApp] Fix kerberos security service issues\ndcdc1a98 Update doc for manila-service-image download\nd7407afd Update manila and Ceph matrix support doc\n6bb02137 Fix missing group and group_snapshots in absolute limits\n7f72147f Implement secure RBAC for storage pool statistics\n46f65eaf Implement secure RBAC for quotas\n5011b671 Implement secure RBAC for quota classes\n9ef2156c Implement secure RBAC for messages\na43911d7 Implement secure RBAC for storage availability zones\nfbc4ecb6 Always use new_size when extend\nd7d83a3f Uses local variable with ipv4 gateway config\n3fa5cf4e [Doc] Add admin doc for share group\n281b7dd1 Add developer reference to share server migration\nbab89906 Delete unavailable py2 package\n0cc7cbc3 [goal] Deprecate the JSON formatted policy file\n39734a70 Stop manila when unstacking\nc0ee8848 Update requirements\n8d1dcb1e Updates to support in Ceph local.conf sample\n8bef5e25 Switch from base64 encodestring to encodebytes\n914d8737 [glusterfs] don\u0027t reinit volume list on deletion\n2a8aaf43 Introduce base personas for secure RBAC\na10927c6 Bump oslo.log version to 4.3.0\nc8087fa0 Docs: NFS-Ganesha and dbus\n5e01d9e2 [NetApp] Support NFS shares on windows\n70f7859f [NetApp] Fix hard-coded CA cert path for SSL\n8a691d86 Retry unmount operation on the LVM driver\n5e8b2a47 Replace deprecated UPPER_CONSTRAINTS_FILE variable\ne3fea147 Retry unmount operation on the ZFSOnLinux driver\n136a8993 Update share replicas export location API ref\n074b0835 fix reno file location and indention\n55ef4999 Remove the unused coding style modules\n21c731e7 Move shares filtering to database layer\n1b5771ef Fix logic that determines a share exists before manage\ncc33a4b2 disable M325\n6d8c99a7 Set Victoria max version in the api doc\n2e27c718 Fix \u0027update_share_replica\u0027 doesn\u0027t provide share server model to drivers\n3ca9c767 Add share server migration admin documentation\n0c4d1e33 [NetApp] Fix access rules for CIFS promoted replica\n52c926e3 Add Python3 wallaby unit tests\n44fe2e5e Update master for stable/victoria\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: Iac76024f04bdb06d7a98203ace334ed50bc52743\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0859d5569f4c9700ecf44712736ec84cc16a687e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0859d5569f4c9700ecf44712736ec84cc16a687e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
