)]}'
{"id":"openstack%2Freleases~781693","triplet_id":"openstack%2Freleases~master~Ibe5bd5a901a08580764f40a3e8fd285a45f1f304","project":"openstack/releases","branch":"master","topic":"wallaby-rc1-deadline","hashtags":[],"change_id":"Ibe5bd5a901a08580764f40a3e8fd285a45f1f304","subject":"Proposing Wallaby RC1 for monasca-thresh","status":"MERGED","created":"2021-03-19 13:11:32.000000000","updated":"2021-03-26 10:35:43.000000000","submitted":"2021-03-26 10:35:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"781693-wallaby-rc1-deadline","meta_rev_id":"e30bbef8d601fc81f5af9aff81fc6d9162f82c0d","_number":781693,"virtual_id_number":781693,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-26 10:35:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2021-03-26 10:20:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2021-03-26 06:16:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-25 23:21:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2021-03-26 10:20:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-03-25 23:22:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-19 13:20:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-19 13:34:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-25 17:22:11.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"state":"REVIEWER"},{"updated":"2021-03-26 10:20:37.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"858a92b26a2ab6ddc8ad9c935df091540529f099","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-19 13:11:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d5d9fe0a51d83c09e74089ca222533abdde233d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 13:34:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7305af7824b40b2a0f82e9fc1702307 : SUCCESS in 7m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a5b0aef75c0d4db181193fd3d0b38f38 : SUCCESS in 5m 30s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/39aac8f4a7774e28bbdf66992e1bdc42 : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":1},{"id":"520a51c0f35e7ceb0505005e9f6f03adabdd212d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-24 09:22:34.000000000","message":"Patch Set 1:\n\nAny update?","accounts_in_message":[],"_revision_number":1},{"id":"a7d5a8227a3c4118b8a85800acaf3358deda8d45","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-25 13:04:04.000000000","message":"Patch Set 1:\n\nHello team,\n\nDeadline is tomorrow, please have a look before. If no response is provided until tomorrow we\u0027ll force it.\n\nThank you for your understanding.","accounts_in_message":[],"_revision_number":1},{"id":"da92045e688854aff560db653e4e54a2a42402b2","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2021-03-25 23:21:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"354894bc512c4a72d00bac404ddc487144c51b9e","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 23:22:01.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"9938f6e214aa3360c148e333aedf6e403482916c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-26 06:16:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1a3a26b71444bdd062c2ecd7c0f0565582d61991","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-26 10:20:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM:\n* version bump is OK\n* approved by PTL\n* test job results don\u0027t show relevant issues","accounts_in_message":[],"_revision_number":1},{"id":"48a6f52b55ca352cb3be989aabbef37254f3153e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 10:26:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3d5534526a342b434d20dd57077d1f3135a0c3b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 10:35:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/964602fdd94b455889cf3ef2de44e185 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/984c953860f64c5b83b6333402e0dbb0 : SUCCESS in 7m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a1ed6228ea3a453aa1c096e5c4fec09a : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":1},{"id":"e30bbef8d601fc81f5af9aff81fc6d9162f82c0d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 10:35:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1e6e9946feeb12dc8bd875639565bbedb87b234d","revisions":{"1e6e9946feeb12dc8bd875639565bbedb87b234d":{"kind":"REWORK","_number":1,"created":"2021-03-19 13:11:32.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/93/781693/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/93/781693/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/93/781693/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/93/781693/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/93/781693/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/93/781693/1"}}},"commit":{"parents":[{"commit":"c5845e766ef8a97d4682c018b3b553786e20dd1a","subject":"Merge \"Amend Wallaby Cycle Highlight for Nova\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c5845e766ef8a97d4682c018b3b553786e20dd1a"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-19 13:11:23.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-19 13:11:23.000000000","tz":60},"subject":"Proposing Wallaby RC1 for monasca-thresh","message":"Proposing Wallaby RC1 for monasca-thresh\n\nIf you wants to wait for a specific patch to make it to the release,\nsomeone from your team can -1 the patch to have it held, or update that patch\nwith a different commit SHA.\n\nPatches with no response will be merged by Friday March 26.\n\n$ git log --oneline --no-merges 3.0.0..d8420f5\nd8420f5 switch to storm version 1.2.3\nbcb469e Fix zuul publish docker image job\n05dc74a Change storm config to eliminate storm warnings\nb50c1d1 Fix zuul publish docker image job\n1464b7b Enable usage of storm.yaml\n0ee0e79 Add mariadb-connector-j and bump version to 2.4.0\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: Ibe5bd5a901a08580764f40a3e8fd285a45f1f304\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1e6e9946feeb12dc8bd875639565bbedb87b234d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1e6e9946feeb12dc8bd875639565bbedb87b234d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
