)]}'
{"id":"openstack%2Freleases~781696","triplet_id":"openstack%2Freleases~master~I6b9f419080afd534f25fcdf484dd4f22b90619a5","project":"openstack/releases","branch":"master","topic":"wallaby-rc1-deadline","hashtags":[],"change_id":"I6b9f419080afd534f25fcdf484dd4f22b90619a5","subject":"Abandon monasca-transform","status":"MERGED","created":"2021-03-19 13:12:11.000000000","updated":"2021-03-29 17:17:08.000000000","submitted":"2021-03-29 17:17:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"781696-wallaby-rc1-deadline","meta_rev_id":"e2e07eed57876bfdb44d966721303fad920053e2","_number":781696,"virtual_id_number":781696,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-29 17:17:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2021-03-29 17:05:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2021-03-29 15:42:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-29 15:34:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2021-03-29 17:05:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-19 13:20:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-19 13:30:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-25 17:20:49.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"state":"REVIEWER"},{"updated":"2021-03-29 11:50:54.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2021-03-29 11:53:40.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-03-29 15:34:16.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"76abdcb41ace78ab82973b7a50053008e9633229","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-19 13:12:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b81a7f4a6dcca94efd5a85f5d03568442d37e78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 13:30:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a44a998913d14c3dbee4795a6f1fc1f9 : SUCCESS in 8m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/66f847f866994038a103c1f7f8e4671f : FAILURE in 5m 58s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/881f681188aa41d8a06e36e882ff9f0c : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":1},{"id":"742c18a60e6396fe2b7bf475eb079bb69158aed3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-19 13:56:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6d764299b7ffd7b116662073eeceec2b5526d893","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 14:09:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a6887f650f746aa9d97ab4442854ac9 : SUCCESS in 7m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/36a2211882e547bbb0e78893f9129aee : FAILURE in 5m 26s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/172dff9440bf499bb36742da85a500e7 : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":1},{"id":"e88d2fd397ec1b512a72783eb52d8f6762eaa59f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-23 08:56:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"71b4e70c5973a4d5a9e295cfc4547d7da02c90da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 09:14:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbb4bc976dbd481d975972604feae122 : SUCCESS in 6m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aeca4ebda6d248eebeb0bbd8fd35717c : FAILURE in 6m 12s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/a1e2b7960a124a1fadebef4979ea2823 : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":1},{"id":"cef5726d8b6c50b3a1040c787fbb0a6c1c5d73d7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-23 13:25:54.000000000","message":"Patch Set 1:\n\nWeird validation issue:\n\n```\ndeliverables/wallaby/monasca-transform.yaml: validate_pypi_permissions: Could not determine a valid PyPI dist name for monasca-transform\n```\nThe pypi repo exist:\nhttps://pypi.org/project/monasca-transform/\n\nI didn\u0027t see significative change in the deliverable config since the last series... I need to dig further.","accounts_in_message":[],"_revision_number":1},{"id":"ae23235ac24eaa1a056ff2995cfef408492c4578","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-23 13:32:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b3f78a7aabe0f30f6f7775f7c328a13d8ac34fbc","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-23 13:33:10.000000000","message":"Patch Set 2:\n\nPS2 helped to fix the issue locally","accounts_in_message":[],"_revision_number":2},{"id":"5bcae486155c2fa9239bf9043de4fe7f0203c6d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 13:41:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/882836fd37b3449aa9ebd132388ae7fa : SUCCESS in 5m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/985d34ed080e4097aca8a905e149acc7 : SUCCESS in 7m 06s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/67ff4bdc7c634262a31d1c67e8421e9d : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":2},{"id":"f344ef06a40c84ee6d65e8895f1f652f1424bc09","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-24 09:31:26.000000000","message":"Patch Set 2:\n\nAny update?","accounts_in_message":[],"_revision_number":2},{"id":"ddd7b3b85cf58ee758eb9e4bb0d01e039018bf1f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-25 13:07:02.000000000","message":"Patch Set 2:\n\nHello team,\n\nFriendly reminder, deadline is tomorrow, please have a look before. If no response is provided until tomorrow we\u0027ll force it.\n\nThank you for your understanding.","accounts_in_message":[],"_revision_number":2},{"id":"bfff0bd816ffaa51a200b4c2f92e7f32cfc8694c","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2021-03-25 23:26:47.000000000","message":"Patch Set 2: Code-Review-1\n\nHi, monasca-transform has been deprecated [1]\n\n[1] https://opendev.org/openstack/monasca-transform","accounts_in_message":[],"_revision_number":2},{"id":"1dfc065bbe37aae799419c80f0aecd490274aef2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-29 10:29:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7b01f3f7c7676f497b801ec2abc0b0000b34340c","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2021-03-29 10:33:40.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"287bda3faa794af2c09a10055d5ea47d447702e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 10:45:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6b8cc7d849e4a779df530e6d968735a : SUCCESS in 6m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cfbaadb54bdc460885e85537162fd545 : SUCCESS in 5m 29s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/3d4bf2ab8b884b75a75f3955e5f8e22f : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":3},{"id":"4d4e6d0f794e06a0262a6feb8d3687e9415afb9a","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-03-29 11:50:54.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"f7d49c38a76381c91438a4b1e11f7312d3451776","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-29 11:53:40.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e4baa081937af0c5c25bf6104515ebfc5d2a38bd","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-29 12:08:06.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"3ba57344706591fe653b07f1cc8ab8dc276239f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-29 14:15:37.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"4edab13e0e5fbf49aa3ec368a26afaa9a24cbe78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 14:30:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4ba2262a86c4111998d6fcb78ed7c94 : SUCCESS in 6m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5029c80d070844a29593bcacd01f5397 : SUCCESS in 5m 15s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/67321652ace548babddac85e8a3554b3 : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":4},{"id":"39111bfbaf7f41fb81b3f6a90866173ef4534246","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-03-29 15:34:16.000000000","message":"Patch Set 4: Code-Review+1\n\nlgtm, repo is retired and documentation changes in governance is also up. We can proceed on release things without waiting for governance patch merge which can take 1-2 or more days and might be late for Wallaby release processes.","accounts_in_message":[],"_revision_number":4},{"id":"7c7b0c32c708968b5376e1997ff42a1d3ec767c7","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-29 15:42:51.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c6baccda7fc4f3497ebae58aa2112077b517fbb4","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-03-29 16:48:10.000000000","message":"Patch Set 4:\n\ngovernance patch is merged https://review.opendev.org/c/openstack/governance/+/783624","accounts_in_message":[],"_revision_number":4},{"id":"1cd319721941bce469914ee53c1b81f48411b39d","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-03-29 17:05:45.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"fe4625043d0609a7dd45feda194882437b04e53f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 17:06:25.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ce68a0d2abe3e42b9837a13ff936598e8c6a3a7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 17:17:03.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a73f110a1c5f40ca9f759ba20b5b9fd8 : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f62c3c8573254f949aa7b7bca62f0525 : SUCCESS in 8m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e85ebd33de5e4665ba64ceb50c6cdf5a : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":4},{"id":"e2e07eed57876bfdb44d966721303fad920053e2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 17:17:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"eab9017547398b6f10c5e698d19a539a4d7a8e2e","revisions":{"7d8c564a17d969b9980d6607ba465a30a4f56d77":{"kind":"REWORK","_number":1,"created":"2021-03-19 13:12:11.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/96/781696/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/96/781696/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/781696/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/781696/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/781696/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/96/781696/1"}}},"commit":{"parents":[{"commit":"c5845e766ef8a97d4682c018b3b553786e20dd1a","subject":"Merge \"Amend Wallaby Cycle Highlight for Nova\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c5845e766ef8a97d4682c018b3b553786e20dd1a"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-19 13:11:55.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-19 13:11:55.000000000","tz":60},"subject":"Proposing Wallaby RC1 for monasca-transform","message":"Proposing Wallaby RC1 for monasca-transform\n\nIf you wants to wait for a specific patch to make it to the release,\nsomeone from your team can -1 the patch to have it held, or update that patch\nwith a different commit SHA.\n\nPatches with no response will be merged by Friday March 26.\n\n$ git log --oneline --no-merges 2.0.0..811acd7\n811acd7 Remove project content on master branch\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I6b9f419080afd534f25fcdf484dd4f22b90619a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7d8c564a17d969b9980d6607ba465a30a4f56d77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7d8c564a17d969b9980d6607ba465a30a4f56d77"}]},"branch":"refs/heads/master"},"7348f5097d279649e8b84bf2964b5c126b88d13d":{"kind":"REWORK","_number":2,"created":"2021-03-23 13:32:31.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/96/781696/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/96/781696/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/781696/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/781696/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/781696/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/96/781696/2"}}},"commit":{"parents":[{"commit":"c5845e766ef8a97d4682c018b3b553786e20dd1a","subject":"Merge \"Amend Wallaby Cycle Highlight for Nova\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c5845e766ef8a97d4682c018b3b553786e20dd1a"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-19 13:11:55.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-23 13:32:09.000000000","tz":60},"subject":"Proposing Wallaby RC1 for monasca-transform","message":"Proposing Wallaby RC1 for monasca-transform\n\nIf you wants to wait for a specific patch to make it to the release,\nsomeone from your team can -1 the patch to have it held, or update that patch\nwith a different commit SHA.\n\nPatches with no response will be merged by Friday March 26.\n\n$ git log --oneline --no-merges 2.0.0..811acd7\n811acd7 Remove project content on master branch\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I6b9f419080afd534f25fcdf484dd4f22b90619a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7348f5097d279649e8b84bf2964b5c126b88d13d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7348f5097d279649e8b84bf2964b5c126b88d13d"}]},"branch":"refs/heads/master"},"3e277b3fa7bbba6a39d04b892218e3f784f9bc5a":{"kind":"REWORK","_number":3,"created":"2021-03-29 10:29:49.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/96/781696/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/96/781696/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/781696/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/781696/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/781696/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/96/781696/3"}}},"commit":{"parents":[{"commit":"c5845e766ef8a97d4682c018b3b553786e20dd1a","subject":"Merge \"Amend Wallaby Cycle Highlight for Nova\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c5845e766ef8a97d4682c018b3b553786e20dd1a"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-19 13:11:55.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-29 10:25:50.000000000","tz":120},"subject":"Abandon monasca-transform","message":"Abandon monasca-transform\n\nThis project have been tagged as deprecated 6 months ago and a patch to\nabandon it have been proposed on the governance repo.\n\nThese changes simply remove the deliverable for Wallaby.\n\nThe project will be no longer released.\n\nDepends-On: https://review.opendev.org/c/openstack/governance/+/783624\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I6b9f419080afd534f25fcdf484dd4f22b90619a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3e277b3fa7bbba6a39d04b892218e3f784f9bc5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3e277b3fa7bbba6a39d04b892218e3f784f9bc5a"}]},"branch":"refs/heads/master"},"eab9017547398b6f10c5e698d19a539a4d7a8e2e":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-03-29 14:15:37.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/96/781696/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/96/781696/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/781696/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/781696/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/781696/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/96/781696/4"}}},"commit":{"parents":[{"commit":"c5845e766ef8a97d4682c018b3b553786e20dd1a","subject":"Merge \"Amend Wallaby Cycle Highlight for Nova\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c5845e766ef8a97d4682c018b3b553786e20dd1a"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-19 13:11:55.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-29 14:14:34.000000000","tz":120},"subject":"Abandon monasca-transform","message":"Abandon monasca-transform\n\nThis project have been tagged as deprecated 6 months ago and a patch to\nabandon it have been proposed on the governance repo [1].\n\nHowever, looks like deprecation was not properly marked in governance.\nIt means was is not done completely that is why release things did not get.\n\ngmann proposed to follow up with Monasca team if that is retired or\ndeprecated and update governance reference too.\n\nWe can remove it from release and the TC will take care of\ngovernance things properly after discussion with Monasca team.\n\nAccording to our discussion with the TC these changes simply remove the\ndeliverable for Wallaby.\n\nThe project will be no longer released.\n\n[1] https://review.opendev.org/c/openstack/governance/+/783624\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I6b9f419080afd534f25fcdf484dd4f22b90619a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/eab9017547398b6f10c5e698d19a539a4d7a8e2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/eab9017547398b6f10c5e698d19a539a4d7a8e2e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
