)]}'
{"id":"openstack%2Freleases~781726","triplet_id":"openstack%2Freleases~master~Idf8d6c4221702c35e835a8dcf6a9cdd868ddbd55","project":"openstack/releases","branch":"master","topic":"wallaby-rc1-deadline","hashtags":[],"change_id":"Idf8d6c4221702c35e835a8dcf6a9cdd868ddbd55","subject":"Proposing Wallaby RC1 for nova","status":"MERGED","created":"2021-03-19 14:15:20.000000000","updated":"2021-03-26 06:57:04.000000000","submitted":"2021-03-26 06:57:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"781726-wallaby-rc1-deadline","meta_rev_id":"7bc2e9c62e045e504c72440c2d62270794452940","_number":781726,"virtual_id_number":781726,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-26 06:57:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2021-03-26 06:44:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2021-03-26 06:13:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-26 06:12:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2021-03-26 06:44:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-03-26 06:12:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-19 14:17:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-19 14:20:19.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2021-03-19 14:20:24.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2021-03-19 14:25:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-25 21:30:26.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2021-03-26 06:44:41.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"45f14444b438532be59cb664671226f43fdf54ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-19 14:15:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bcee6f91d6f50c260773a245f2abeafed54b94a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 14:25:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4cf52a95b484944aeedaa9c0843373d : SUCCESS in 7m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/559aa5e67fe641438bb91e48c4fbb29e : SUCCESS in 5m 41s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/673593c06c5a4e59bccf33c5a6567dbc : SUCCESS in 6m 39s","accounts_in_message":[],"_revision_number":1},{"id":"4d46989fba40c17e3d6516f4b654ff74c361f5a7","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-03-22 09:41:28.000000000","message":"Patch Set 1: Code-Review-1\n\nWe need:\n* reno prelude: \u003cTBD\u003e\n* we expect RPC 6.0 bump: https://review.opendev.org/c/openstack/nova/+/761452\n* Wallaby RPC version alias: \u003cTBD\u003e","accounts_in_message":[],"_revision_number":1},{"id":"dfd5b5d7fc0ee2a4c554eaa665bcf46f3e7b26d2","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-03-22 09:49:31.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e We need:\n\u003e * reno prelude: \u003cTBD\u003e\n\u003e * we expect RPC 6.0 bump: https://review.opendev.org/c/openstack/nova/+/761452\n\u003e * Wallaby RPC version alias: if ^^ merge then that defines the alias too, If that does not merges then we have a separate patch to mark 5.13 as Wallaby: https://review.opendev.org/c/openstack/nova/+/782115","accounts_in_message":[],"_revision_number":1},{"id":"2b9a7228b2b4936d0339cc0033cf04b776781d61","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-03-22 12:34:17.000000000","message":"Patch Set 1:\n\nWe need:\n* reno prelude: https://review.opendev.org/c/openstack/nova/+/782172\n* we expect RPC 6.0 bump: https://review.opendev.org/c/openstack/nova/+/761452\n* Wallaby RPC version alias: if ^^ merge then that defines the alias too, If that does not merges then we have a separate patch to mark 5.13 as Wallaby: https://review.opendev.org/c/openstack/nova/+/782115","accounts_in_message":[],"_revision_number":1},{"id":"c377374a631fce9db29f8ad35f53cf10210da9ec","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-25 12:45:24.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e We need:\n\u003e * reno prelude: https://review.opendev.org/c/openstack/nova/+/782172\n\nLooks mostly done ^\n\n\u003e * we expect RPC 6.0 bump: https://review.opendev.org/c/openstack/nova/+/761452\n\nLooks mostly done ^\n\n\u003e * Wallaby RPC version alias: if ^^ merge then that defines the alias too, If that does not merges then we have a separate patch to mark 5.13 as Wallaby: https://review.opendev.org/c/openstack/nova/+/782115","accounts_in_message":[],"_revision_number":1},{"id":"89e3f4cb22d4c964afcb3ef9f810d7c729287265","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-03-25 15:19:15.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e We need:\n\u003e * reno prelude: https://review.opendev.org/c/openstack/nova/+/782172\n\u003e * we expect RPC 6.0 bump: https://review.opendev.org/c/openstack/nova/+/761452\n\u003e * Wallaby RPC version alias: if ^^ merge then that defines the alias too, If that does not merges then we have a separate patch to mark 5.13 as Wallaby: https://review.opendev.org/c/openstack/nova/+/782115\n\nprelude and RPC 6.0 bump has been landed. But we found a bug in the RPC6.0 bump so we need https://review.opendev.org/c/openstack/nova/+/783026 to land before we can cut RC1 safely.","accounts_in_message":[],"_revision_number":1},{"id":"49b0774f789ea6ca277add9adaaf012e63cd5404","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-03-25 21:30:17.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e We need:\n\u003e \u003e * reno prelude: https://review.opendev.org/c/openstack/nova/+/782172\n\u003e \u003e * we expect RPC 6.0 bump: https://review.opendev.org/c/openstack/nova/+/761452\n\u003e \u003e * Wallaby RPC version alias: if ^^ merge then that defines the alias too, If that does not merges then we have a separate patch to mark 5.13 as Wallaby: https://review.opendev.org/c/openstack/nova/+/782115\n\u003e \n\u003e prelude and RPC 6.0 bump has been landed. But we found a bug in the RPC6.0 bump so we need https://review.opendev.org/c/openstack/nova/+/783026 to land before we can cut RC1 safely.\n\nthe fix for the RPC 6.0 bump failed on the gate and I will need to recheck it as soon as Zuul votes. Then I will go to bed. I will be back around 6:00 UTC tomorrow. I delegate the PTL power to Dan to update and +1 the release patch if the fix merges while I\u0027m asleep.","accounts_in_message":[],"_revision_number":1},{"id":"78efba9a400326650885b931765914b13ae02092","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-26 04:08:34.000000000","message":"Patch Set 1:\n\nWell, end of night watch for me and the nova patch still isn\u0027t merged. It\u0027s currently #5 in the gate and looks like it might make it this time, but I won\u0027t be around to see it. Hopefully we can get it early tomorrow if it needs another recheck.","accounts_in_message":[],"_revision_number":1},{"id":"130274eb7edd3e49e3ecd1066f0cb6b780ac9e8b","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-03-26 06:02:32.000000000","message":"Patch Set 1:\n\nfix merged, I\u0027m updating the release patch now with new hash...","accounts_in_message":[],"_revision_number":1},{"id":"edf9c943cf7db0a11718e2a6404ebfd6bf1a8ca9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-03-26 06:09:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0bf23499b9e2472d2ef75d937759f2102ba065ed","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-03-26 06:10:57.000000000","message":"Patch Set 2:\n\nnova is ready for RC1","accounts_in_message":[],"_revision_number":2},{"id":"82a5efce071905002e8210cf1906303da6dc185b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-26 06:12:03.000000000","message":"Patch Set 2:\n\nThanks Balazs","accounts_in_message":[],"_revision_number":2},{"id":"95bf4fab725b7406e729c9b564bf8d83fa3e42d9","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-03-26 06:12:03.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a37b1026ec815b97ce34bd031a042b2445374f92","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 06:12:58.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"ccb5ddefd24c2f89269beee9e7ff0801dbdc37f6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-26 06:13:01.000000000","message":"Patch Set 2: Code-Review+2\n\nRebased on the HEAD of master LGTM","accounts_in_message":[],"_revision_number":2},{"id":"b84db2379cbdcdccabc028da5fb919f803935489","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 06:22:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2f3f93c3ca54d2b809e3707f639323c : SUCCESS in 8m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6cccb77c31864af898acd527bc8bb8a0 : SUCCESS in 6m 59s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c7b92896872547c78155fdad6aa9f47c : SUCCESS in 7m 05s","accounts_in_message":[],"_revision_number":2},{"id":"1717674a114557fa24fed2372b375ed49ac9ef28","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-26 06:44:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLGTM:\n* version bump is OK\n* uploaded by PTL\n* validator passed\n* I don\u0027t see any issue in list-changes-results.log","accounts_in_message":[],"_revision_number":2},{"id":"20b817cecb169f8bba7cf1cc971d8860ef239372","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 06:45:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c2d32d76536de10ce5c48f788e1e7386b9e57bbf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 06:57:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbd1bd13914944439ce00fa5cbd58a5d : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb25e12a31a54b8c904d87dfc7ca48df : SUCCESS in 6m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c352d6066df44dd8a554201b82384262 : SUCCESS in 6m 22s","accounts_in_message":[],"_revision_number":2},{"id":"7bc2e9c62e045e504c72440c2d62270794452940","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 06:57:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"53e1fe0fbc088f068d42e221cb2e390c52984499","revisions":{"afe00c86d86dff3a2a542cd38f970bfdb39ebcda":{"kind":"REWORK","_number":1,"created":"2021-03-19 14:15:20.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/26/781726/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/26/781726/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/26/781726/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/26/781726/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/26/781726/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/26/781726/1"}}},"commit":{"parents":[{"commit":"753fc318dec0b4afaca6a4408e3d5ec7aaa734dc","subject":"Merge \"[OpenstackSDK] Create for all client and non-client lib wallaby branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/753fc318dec0b4afaca6a4408e3d5ec7aaa734dc"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-19 14:15:10.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-19 14:15:10.000000000","tz":60},"subject":"Proposing Wallaby RC1 for nova","message":"Proposing Wallaby RC1 for nova\n\nIf you wants to wait for a specific patch to make it to the release,\nsomeone from your team can -1 the patch to have it held, or update that patch\nwith a different commit SHA.\n\nPatches with no response will be merged by Friday March 26.\n\n$ git log --oneline --no-merges 22.0.0..0c64f4c3ea\n5d5ff82bab Add config parameter \u0027live_migration_scheme\u0027 to live migration with tls guide\nab04eb2196 pci: Add vDPA vnic to PCI request mapping and filtering\n14071dfb11 libvirt: Deprecate `live_migration_tunnelled`\nf4c249c692 docs: Document UEFI secure boot feature\ne9c80da2db libvirt: Add guest generation for vDPA\n9a673a8faa libvirt: Add vDPA nodedev parsing\ndcfe40c30b tests: Remove duplicated \u0027start_compute\u0027 helper\nb202fa3a0e tests: Add functional tests for UEFI, secure boot\n52b505c9cb libvirt: Move PCI host device parsing to \u0027host\u0027\n7313832d58 objects: Add \u0027VDPA\u0027 to \u0027PciDeviceType\u0027\n5cd1acc3bf Support per port numa policies with SR-IOV\nb2471dd578 pci: implement the \u0027socket\u0027 NUMA affinity policy\n658f2327bc add constants for vnic type vdpa\n46704f7702 Make nova-ceph-multistore use policy.yaml\nd79f366cf5 fakelibvirt: make kB_mem default not laughable\n46e3502c30 libvirt: Report secure boot support to scheduler\ne7d704899f libvirt: Wire up \u0027os_secure_boot\u0027 property\n9fff6893ce libvirt: Use firmware metadata files to configure instance\n7d08a360d7 setup.cfg: Resolve warning\n452d2fb3a0 libvirt: Move some host arch checks to guest checks\nfaad45b632 libvirt: Add parsing of firmware metadata files\n890b6d54a6 pci: track host NUMA topology in stats\n6c3175d3ee pci manager: replace node_id parameter with compute_node\n1862971faa scheduler: Translate secure boot requests to trait\nd2fef725b1 compute: Report COMPUTE_SECURITY_UEFI_SECURE_BOOT\nbc020efab3 hardware: Start parsing \u0027os_secure_boot\u0027\ne0e969857b trivial: Clarify purpose of \u0027Host.supports_*\u0027 properties\n141f3e7d4f libvirt: Add missing type hints\n94d16e7c77 libvirt: Stop passing around virt_type, caps\n31889ce296 libvirt: Add tests for insecure UEFI configuration\n8f317f442c nova-next: Start testing the q35 machine type\n598e247437 Differentiate between InstanceNotFound and ConstraintNotMet\naeb8de892b Add device event constants to fakelibvirt\n8da39de007 libvirt: Add firmware auto-configuration support\n3f63c68195 libvirt: Add support for virtio-based input devices\na65e4201cc tests: Stop mocking \u0027nova.virt.libvirt.utils.get_arch\u0027\n3d4e53f832 Add functional test for bug 1837995\nbcd6b42047 libvirt: Allow disabling CPU flags via `cpu_model_extra_flags`\n4b32f9c9e3 Remove VFSLocalFS\nd06a10f096 Remove non-libguestfs file injection for libvirt\ndf89596edb apidb: Compact Train database migrations\ndae3c89874 apidb: Compact Stein database migrations\n6665bd8cb9 apidb: Compact Rocky database migrations\n86b423f9f3 apidb: Compact Queens database migrations\n31f01aa44f apidb: Compact Pike database migrations\n743693c740 apidb: Compact Ocata database migrations\n9020896cb7 docs: Add admin docs for configuring and updating machine types\n0b44c7ccef libvirt: Add a config update workflow test for [libvirt]hw_machine_type\n947f813e56 Fix ImageBackendFixture not to support direct_snapshot\n8cddd243bf nova-status: Add hw_machine_type check for libvirt instances\n9a5b07d9fc nova-manage: Add libvirt list_unset_machine_type command\nc70cde057d nova-manage: Add libvirt update_machine_type command\n20692c245c nova-manage: Add libvirt get_machine_type command\nf0eeba5c8c hyper-v rbd volume support\n95fbbce8b2 requirements.txt: Bump os-brick to 4.2.0\n98713d69ac libvirt: Deprecate disable_native_luksv1 and rbd_volume_local_attach\ne64744b92f rpc: Rework \u0027get_notifier\u0027, \u0027wrap_exception\u0027\nc23bda400a libvirt: Remove dead code\n59986cfd4f Replace md5 for fips\nc8478e40bd Remove short call timeout from func test\ndb7517d5a8 scheduler: Add \u0027socket\u0027 PCI NUMA affinity policy trait\nbe80dfdc71 objects: Add \u0027socket\u0027 PCI NUMA affinity\nfef371a425 Follow up for QoS interface attach\n5d3fb6cdbd Docs: Correct ``Password injection using the dashboard`` Explanation\n96072a1ef1 libvirt: parse alias out from device config\na634103b15 libvirt: allow querying devices from the persistent domain\nf5cd6e2dea libvirt: add AsyncDeviceEventsHandler\n4a70fc9cfb libvirt: Define and emit DeviceRemovedEvent and DeviceRemovalFailedEvent\n95b9481aa4 libvirt: start tracking NUMACell.socket for hosts\n4e23d93410 libvirt: Parse the \u0027os\u0027 element from domainCapabilities\nf7e3e17991 tests: Poison os.uname\nb367cf114c tests: Remove unnecessary mock\n1053ae0b61 tests: Add \u0027nova.virt.libvirt.utils.get_arch\u0027 stub to fixture\n5c35a5905e FUP: Catch and reraise routed nets exception\na1673d3118 Add a routed networks scheduler pre-filter\nc4b28a5496 Add net \u0026 utils methods for routed nets \u0026 segments\n123f6262f6 Handle instance \u003d None in _local_delete_cleanup\nf88a8b0c1b Fixes the elapsed time logged during a live migration\n838370a490 libvirt: add IP address to libvirt guest metadata\nf7975d640c Add regression test for bug 1914777\nd23cf26ec7 Change API unexpected exception message\n85cb4af075 tox: Add passenv DISABLE_CHERRY_PICK_CHECK to pep8\n5b20c628b2 libvirt: Record the machine_type of instances in system_metadata\n21656d4e9c Run nova-live-migration job with neutron-trunk enabled\n172024db71 Centralize sqlite FK constraint enforcement\n2e8e04a8f6 libvirt: Don\u0027t drop CPU flags with policy\u003d\u0027disable\u0027 from guest XML\n94b5db6d2e Make a couple test jobs run async devstack\n18ad9d97ab Remove unused Instance method\n5af28c4de5 Support interface attach with QoS ports\n9bdfdf7222 Uncap PrettyTable\n302f2bc4f7 apidb: Compact Newton database migrations\nd2449232bd apidb: Compact Mitaka database migrations\nd97fd31131 apidb: Compact Liberty database migrations\na172b75587 apidb: Add manage.py script\n35b58a5035 cmd: Remove \u0027nova-manage db ironic_flavor_migration\u0027\n737cae76e4 conf: Clean up docs for scheduler options\n9d53f366ff conf: Drop \u0027[scheduler] periodic_task_interval\u0027\n5aeb3a3874 Drop support for custom schedulers\n81b689e5bb fup: Merge duplicate volume attachment checks\n1252588d4e api: Reject volume attach requests when an active bdm exists\nee3a8f0225 Add regression test for bug #1908075\ncea5bf3d93 Remove __unicode__() from nova unit test Exception\nf6bacd3fde Refactor and rename test_tcp_rst_no_compute_rpcapi\n9e96f64126 Rename ensure_network_metadata to amend requested_networks\nedaaac6aaf Add requested_networks field to RequestSpec object\n2d4eff0ef7 cmd: Remove \u0027nova-manage db null_instance_uuid_scan\u0027\n7961cc3428 libvirt: Always enable USB controller on PPC64\n21af4dd4bb libvirt: Stub out \u0027nova.virt.libvirt.utils.get_arch\u0027\n96cf885ecd zuul: Mark nova-lvm as voting\n7d66f97e0a zuul: Increase SWIFT_LOOPBACK_DISK_SIZE within nova-lvm job\n936039d386 [FUP] this change removes an unused paramater\na19d25b67f docs: Move the LibvirtDistroSupportMatrix wiki page into our docs\ncbe3cd9dc3 Restart glance after updating policy in job\n25a632a4e1 Disallow CONF.compute.max_disk_devices_to_attach \u003d 0\nf9901ca927 glance: Remove [glance]/allowed_direct_url_schemes\nb62a1abd61 libvirt: Use specific user when probing encrypted rbd disks during extend\n7062e1db8b tox: Enable parallel docs build\n5beec0feab libvirt: Rationalize attachment of USB tablet\ndb8213af83 api: Reject requests to detach a volume when the compute is down\na3697ed013 Add regression test for bug #1909120\ncb56ae6aad Fix ForbiddenWithAccelerators to HTTPForbidden for shelve API\nee603927b0 libvirt: Remove \u0027hypervisor_version\u0027 from \u0027libvirt_info\u0027\n6b4db089b2 zuul: Allow encryption tests in nova-ceph-multistore again\n95724bbaef libvirt: Bump MIN_{LIBVIRT,QEMU}_VERSION and NEXT_MIN_{LIBVIRT,QEMU}_VERSION\n3a390c2c82 libvirt: Drop support for Xen\nd02ce3c4f0 libvirt: Drop support for UML\n0aff191514 libvirt: Remove MIN_LIBVIRT_VIR_ERR_DEVICE_MISSING\nd5420bbb50 docs: Add reference docs for internal block device structures\n684780ac0d libvirt: Load and cache volume drivers on-demand\n0f558749d1 libvirt: Stop NVMe and ScaleIO unit tests from calling os-brick\n6eae97843a zuul: Skip test_attach_cloned_encrypted_volume in nova-ceph-multistore\naf15488be9 add openstack-python3-wallaby-jobs-arm64 job\n74f5eb2ce9 Add generate schemas tool\ndda179d3f9 Do not inherit os_glance properties on snapshot\nc3804efd42 Refactor ResourceRequest constructor\ne6ff3651e6 Enable mypy on scheduler/utils.py\nbe9dd3d9db Refactor update_pci_request_spec_with_allocated_interface_name\n20fe077343 Factor out port resource deallocation\nb7f133898e Make remove allocation symmetric with add allocation\n54865a17cd Allow extending an existing instance allocation\n672b288324 Refactor _claim_pci_device_for_interface_attach to prepare for qos\n160e229f85 Move query param generation to RequestGroup o.vo\n534b4f3895 api: Log os-resetState as an instance action\n40e0578683 api-ref: Clarify \u0027all_tenants\u0027 command\n7fbd787b1b Cyborg shelve/unshelve support\na72ace1c6c Reuse code from oslo lib for JSON policy migration\n594df9aeb0 use psycopg2 binary instead of source package\n9046f0fff4 utils: Remove periods from instance hostnames\n5f93b680e7 Fix typo in warning message\ndc09bcd311 libvirt: Mock get_arch during some stable rescue unit tests\n7aa7fb94fd tests: Clean up \u0027test_hypervisors\u0027\n1f67ce2496 api: Drop statistics-style fields from os-hypervisors\nef7598ac28 api: Normalize exception handling for os-hypervisors\n5c9c81e0f9 db: Compact Train database migrations\nf0175a346a db: Compact Stein database migrations\n02d6a96588 db: Compact Rocky database migrations\nd83b02c5f6 db: Compact Queens database migrations\n7674fc6cd8 db: Compact Pike database migrations\neb5a1e4d72 db: Compact Ocata database migrations\nc57cc047ee db: Compact Newton database migrations\n76e8653ba1 db: Compact Mitaka database migrations\n1da542ff7a db: Compact Liberty database migrations\nb4fc1c141b db: Compact Kilo database migrations\n15bb50daff db: Compact Juno database migration\n833b7b60dd db: Compact Icehouse database migrations\n4701d28da3 db: Drop \"remove\" of index\nb7ba576cdd db: Merge in unique constraints\nd693d755db db: Clean up base migration\nf86bf703e5 tests: Remove \u0027test_extended_hypervisors\u0027\nc960fee835 Use the non polling notification waiter in func test\n6c6bca0326 Remove dead code from SchedulerReportClient\nb4f560d4e1 Improving the description for unshelve request body\n606a5dfecc api-ref: Add notes about volume attach and detach being async\n800254c0bc Run the db migration tests in the same test worker\n0be00cd58a Trival Change: Remove redundant code in instance delete\n9a0bc6e9cf Revert \"nova-multi-cell: Skip test_cold_migrate_unshelved_instance\"\nb247e69923 stabilize set_host libvirt driver unit test\nc501e23cb7 stabilize unit test asserting log\n1c056c384d Fallback to same-cell resize with qos ports\nf96ade2726 Reproduce bug 1907522 in functional test\n8b33ac0644 only wait for plugtime events in pre-live-migration\n00ad3bb1f1 tests: Merge \u0027test_hypervisor_status\u0027 into \u0027test_hypervisors\u0027\n032bbc3960 Refactoring of functional.regression.test_bug_1702454\n0617cebcc9 Fix a deprecation warning of \u0027collections\u0027\nbd0ead2d63 Add a description in the PTL guide\n1cf2431f4b Remove six.text_type (2/2)\n383e2a8bdc Remove six.text_type (1/2)\n07462dd005 Remove six.binary_type/integer_types/string_types\nf02899418d libvirt: make cross cell resize spawn from snapshot image\neaa98e3340 nova-grenade-multinode: Skip test_live_block_migration_paused\n048250a4b7 Disable qos resize tempest test for nova-multi-cell job\n01f0424930 Enhance simple usage API documentation\n1811731424 nova-evacuate: Remove leftover playbook from standalone job\n7a9d4df24b lower-constraints: Bump packaging to 20.4\n610396f8ad tools: Allow check-cherry-picks.sh to be disabled by an env var\n95da142a15 Remove outdated comment from tox.ini\n0b942dcaa5 Omit resource inventories from placement update if zero\n3cb1215f19 Docs: correct cpu_thread_policy explanation\nc11f4e5336 nova-multi-cell: Skip test_cold_migrate_unshelved_instance\nf0ffed8c7f libvirt: check for AMD SEV only on x86-64\na107a5099e libvirt: Skip encryption metadata lookups if secret already exists on host\n4bea68f12b Revert \"zuul: Skip test_attach_scsi_disk_with_config_drive in nova-ceph-multistore\"\n671dc5bfe8 pci: Improve testing of \u0027nova.pci.request\u0027 module\nd81ef45041 pci: Add logging for filtering\nc2357ab9f3 pci: Improve docs for \u0027_filter_*\u0027 functions\n76360e566b nova-live-migration: Disable *all* virt services during negative tests\n7f7a7a163e zuul: Add devstack-plugin-ceph-compute-local-ephemeral to experimental\n26c46a409f compute: Don\u0027t detach volumes when RescheduledException raised without retry\n836e13cd57 zuul: Skip test_attach_scsi_disk_with_config_drive in nova-ceph-multistore\nadc705ee0c pci: Add a \u0027_filter_pools\u0027 helper function\n9c0ea4a901 doc: Fix rendering in the PTL guide\nd38b7e13a6 Support sys.argv in wsgi app\n454b8d30f4 Fix config option default value for sample config file\n9d1683f70f virt: Remove \u0027get_all_bw_counters\u0027 API\n1273c5ee0b Make PCI claim NUMA aware during live migration\n68bc87876f virt: Remove \u0027reset_network\u0027 API\n67454b307b Fix invalid argument formatting in exception messages\ned756d07b1 Fix the instance_uuid for test\n528740128a Revert \"Temporarily disable parts of heal port allocation test\"\n30067be9bd virt: Remove \u0027change_instance_metadata\u0027 API\n160ed6ff65 Add missing exception\ne13e8c8cd4 api-ref: Move \u0027os-agents\u0027 API to obsolete section\n65fb14c36e doc: require openstack client change for every new API microversion\n8c9d6fc8f0 Ignore PCI devices with 32bit domain\n976ac722d3 Reproduce bug 1897528\n2c28b74340 [Trivial]Add missing white space in the log message\na1ec6e69fa doc: Update the PTL guide\n433bee58bc Restore retrying the RPC connection to conductor\n52c8137b4b [docs] Fix a placement client\u0027s command\n1d23b358d6 functional: Wait for revert resize to complete\n3fcaf579a2 Add functional-py39 testing\nbf47afd70a Update contributor guide for Wallaby\n798f06abb2 Remove [database]connection defaulting\n3b44275868 Improve error handling during service level check\n66c7f00e1d Fix unplugging VIF when migrate/resize VM\nb62f9a04a0 Add os-volume_attachments reference docs\nd718adb108 functional: Expand SR-IOV live migration tests with NUMA\nec8732a4b8 functional: Add live migration tests for PCI, SR-IOV servers\nc163205489 Remove compute service level check for qos ops\n4916ab7a4c rbd: Only log import failures when the RbdDriver is used\nbe752b8175 zuul: Reintroduce nova-dsvm-multinode-base\nf6d74eabfc Remove six.byte2int/int2byte\n28ad7c517d Remove six.iteritems/itervalues/iterkeys\n1d0a0e8c20 Remove six.moves\n4ac4a04d18 zuul: Replace nova-live-migration with zuulv3 jobs\nc0fe95fcc5 zuul: Merge nova-evacuate into nova-multinode-live-migration\n8d9f298f4c Fix the compute RPC 5.12 issue\nf2ebe107d1 Bump the lowest eventlet version to 0.26.1\n8f79afd448 Add a regression test for 5.12 compute API issue\n3b8257cd23 Add upgrade check about old computes\n7675964af8 Set instance host and drop migration under lock\n3f348602ae Reproduce bug 1896463 in func env\nf419f1a4b1 Use _evacuate_server helper in func test\naa7c6f8769 Prevent starting services with older than N-1 computes\n033a325578 Remove unused bindir config option\nc82ce37635 Replace md5 with oslo version\ne7c7a2851e Modify glance\u0027s copy_image permission for nova-ceph-multistore\n7bd2bef3b4 tox: Stop linting as part of docs target\nc716e7608e releasenotes: Add fixes note for bug #1882521\n0d2ca53bb8 libvirt: Only ask tpool.Proxy to autowrap vir* classes\nb32a685595 docs: Make JSON valid\n5b9d587e74 trivial: Remove irrelevant comment\nf5a68826c7 [doc]: Fix glance image_metadata link\n2317b92f4e Add regression test for bug #1899835\ne728fe668a Use subqueryload() instead of joinedload() for (system_)metadata\nfbf6178762 Ignore InstanceNotFound exception when host rejoin cluster\n78ad4a4efd Add placeholder migrations for Victoria backports\neeeca4ceff Handle disabled CPU features to fix live migration failures\n2bd8900d9b Fix the vGPU dynamic options race\nb8695de6da Update pci stat pools based on PCI device changes\ne620e45790 CinderFixture: Return a driver_type of `fake` within connection_info\n1f06418549 Remove self.server dependency form _evacuate_server\n770c139edb Add regression test for bug #1899649\n4a4f12678a Fix virsh domifstat to get vhostuser vif statistics\n1c256cf774 Prevent archiving of pci_devices records because of \u0027instance_uuid\u0027\n82528c83ac hardware: Check inventory of shared CPUs for \u0027mixed\u0027 policy\n9e1fe29d76 functional: Add test for #1898272\n583e85622b functional: Add tests for mixed CPU policy\n12cc805d36 objects: Fix issue in exception type\n450213f1f0 Clean up the DynamicVendorData constructor\n1390eecf8d Use cell targeted context to query BDMs for metadata\ncf0e519a0e Allow excluding image type capabilities\nf44700935f Include removal of ephemeral backing files in the image cache manager\n6cf449bdd0 compute: Lock by instance.uuid lock during swap_volume\n383bcbca70 optimize the shelve code flow\n1e10461c71 Follow up for cherry-pick check for merge patch\nc34a17db6f libvirt: Only add a USB controller if it\u0027s necessary\n5b0343d3e1 libvirt: Remove support for \u0027[libvirt] use_usb_tablet\u0027\nf2ca089bce libvirt: \u0027video.vram\u0027 property must be an integer\ndd1e6d4b0c libvirt: Increase incremental and max sleep time during device detach\naa16dd09eb libvirt: Log exception when unable to import rbd or rados\nead034d969 bindep: Install python3 and python3-devel on CentOS 8 and Fedora\n015f7dadcc libvirt: Remove MIN_{LIBVIRT,QEMU}_PMEM_SUPPORT\n0e7cd9d1a9 libvirt: Remove MIN_LIBVIRT_VIDEO_MODEL_VERSIONS\n226250beb6 nova-evacuate: Disable libvirtd service and sockets during negative tests\nd309e3cdf5 Temporarily disable parts of heal port allocation test\n8d3c2ce92b Add a hacking rule for assert_has_calls\n1175081294 Fix a hacking test\n6b22b697f7 libvirt: Remove MIN_LIBVIRT_BETTER_SIGKILL_HANDLING\n6c8e287156 libvirt: Remove MIN_{LIBVIRT,QEMU}_NATIVE_TLS_VERSION\n99850a0481 Add reproducer for bug #1894095\ned4a748bda Add Python3 wallaby unit tests\n851cc12f67 Update master for stable/victoria\nf357d80407 zuul: Introduce nova-evacuate\n73cc85d910 Log stack trace when fails launching a defined domain\n416cd1ab18 libvirt: make mdev types name attribute be optional\n34cc68ac6e libvirt: Remove MIN_LIBVIRT_FILE_BACKED_DISCARD_VERSION\n48123d6321 hyperv: Configures chassis asset tags for VMs.\n50ff47d553 remove python warnning  from tox\nb933312579 Use absolute path during qemu img rebase\n4861eebc91 fixtures: Handle binding of first port\n0fa6f4e7e1 objects: Remove Agent, AgentList\n7ac52e643c api: Remove \u0027os-agents\u0027 API\n8aea747c97 virt: Remove \u0027get_per_instance_usage\u0027 API\n83ae149c72 virt: Remove various aggregate APIs\ncaa5f9ef94 virt: Remove \u0027set_bootable\u0027 API\n39fe22161c virt: Remove \u0027post_interrupted_snapshot_cleanup\u0027 API\n912b717207 virt: Remove \u0027get_console_pool_info\u0027 API\n696c8a95f6 Raise 409 when removing security group from instance\n17184e7d56 vmware: Use cookiejar from oslo.vmware client directly\ndf8a658bd9 vmware tests: Support different moref backend representations\nae8c3067c5 libvirt: Use local variable instead of CONF\nb6e9023751 Fix misleading documentation for live_migration_inbound_addr\nde10c126e9 Improve documentation of (unpin|pin)_cpus_with_siblings()\n5477c0045f Add an online migration for PciDevice.uuid\n9d4725bcb6 Modify PciDevice.uuid generation code\n76c60c25bc Update supported transports for iscsi connector\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: Idf8d6c4221702c35e835a8dcf6a9cdd868ddbd55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/afe00c86d86dff3a2a542cd38f970bfdb39ebcda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/afe00c86d86dff3a2a542cd38f970bfdb39ebcda"}]},"branch":"refs/heads/master"},"53e1fe0fbc088f068d42e221cb2e390c52984499":{"kind":"REWORK","_number":2,"created":"2021-03-26 06:09:27.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/26/781726/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/26/781726/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/26/781726/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/26/781726/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/26/781726/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/26/781726/2"}}},"commit":{"parents":[{"commit":"753fc318dec0b4afaca6a4408e3d5ec7aaa734dc","subject":"Merge \"[OpenstackSDK] Create for all client and non-client lib wallaby branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/753fc318dec0b4afaca6a4408e3d5ec7aaa734dc"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-19 14:15:10.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-03-26 06:07:38.000000000","tz":60},"subject":"Proposing Wallaby RC1 for nova","message":"Proposing Wallaby RC1 for nova\n\nIf you wants to wait for a specific patch to make it to the release,\nsomeone from your team can -1 the patch to have it held, or update that patch\nwith a different commit SHA.\n\nPatches with no response will be merged by Friday March 26.\n\n$ git log --oneline --no-merges 22.0.0..4a285b1fb\n4a285b1fb9 Fix check_instance_shared_storage() call\n1780005eed Bump the Compute RPC API to version 6.0\n020e46094a [Arm64][libvirt] Fix read bytes from json files\nbc2c19bb2d Reset global wsgi app state in unit test\n49767beaa9 docs: Add SEV guide\nc5ebaef610 docs: Remove duplicated PCI passthrough extra spec info\n94c03e8d66 docs: Remove duplicate TPM extra spec info\n777c02485f docs: Add a real-time guide\n8528eaa602 docs: Add a resource limits guide\ndd15a09af4 Wallaby 23.0.0 prelude section\n064349a888 releasenotes: Fix typo\n7e7209b63a doc: mark the max microversion for wallaby\n5d5ff82bab Add config parameter \u0027live_migration_scheme\u0027 to live migration with tls guide\n78be7de1e5 Add missed accel_uuids for _poll_shelved_instances\n7326e46aae Add release note for vDPA\n22fcfcda23 tests: Add functional test for vDPA device\n45798adf5a api: Block unsupported actions with vDPA\nab04eb2196 pci: Add vDPA vnic to PCI request mapping and filtering\n14071dfb11 libvirt: Deprecate `live_migration_tunnelled`\nf4c249c692 docs: Document UEFI secure boot feature\ne9c80da2db libvirt: Add guest generation for vDPA\n9a673a8faa libvirt: Add vDPA nodedev parsing\nbecb94ae64 Dynamically archive FK related records in archive_deleted_rows\ndcfe40c30b tests: Remove duplicated \u0027start_compute\u0027 helper\nb202fa3a0e tests: Add functional tests for UEFI, secure boot\nff3fd84636 docs: Add note about rescuing bfv instances with the 2.87 microversion\n52b505c9cb libvirt: Move PCI host device parsing to \u0027host\u0027\n7313832d58 objects: Add \u0027VDPA\u0027 to \u0027PciDeviceType\u0027\n5cd1acc3bf Support per port numa policies with SR-IOV\nb2471dd578 pci: implement the \u0027socket\u0027 NUMA affinity policy\n658f2327bc add constants for vnic type vdpa\n46704f7702 Make nova-ceph-multistore use policy.yaml\nd79f366cf5 fakelibvirt: make kB_mem default not laughable\n46e3502c30 libvirt: Report secure boot support to scheduler\ne7d704899f libvirt: Wire up \u0027os_secure_boot\u0027 property\n9fff6893ce libvirt: Use firmware metadata files to configure instance\n7d08a360d7 setup.cfg: Resolve warning\n452d2fb3a0 libvirt: Move some host arch checks to guest checks\nfaad45b632 libvirt: Add parsing of firmware metadata files\n8318eb9c2f docs: Add version{added,changed} notes to the hw_machine_type guide\n890b6d54a6 pci: track host NUMA topology in stats\n6c3175d3ee pci manager: replace node_id parameter with compute_node\n1862971faa scheduler: Translate secure boot requests to trait\nd2fef725b1 compute: Report COMPUTE_SECURITY_UEFI_SECURE_BOOT\nbc020efab3 hardware: Start parsing \u0027os_secure_boot\u0027\ne0e969857b trivial: Clarify purpose of \u0027Host.supports_*\u0027 properties\n141f3e7d4f libvirt: Add missing type hints\n94d16e7c77 libvirt: Stop passing around virt_type, caps\n31889ce296 libvirt: Add tests for insecure UEFI configuration\n8f317f442c nova-next: Start testing the q35 machine type\n598e247437 Differentiate between InstanceNotFound and ConstraintNotMet\naeb8de892b Add device event constants to fakelibvirt\n8da39de007 libvirt: Add firmware auto-configuration support\n3f63c68195 libvirt: Add support for virtio-based input devices\na65e4201cc tests: Stop mocking \u0027nova.virt.libvirt.utils.get_arch\u0027\n3d4e53f832 Add functional test for bug 1837995\nbcd6b42047 libvirt: Allow disabling CPU flags via `cpu_model_extra_flags`\n4b32f9c9e3 Remove VFSLocalFS\nd06a10f096 Remove non-libguestfs file injection for libvirt\ndf89596edb apidb: Compact Train database migrations\ndae3c89874 apidb: Compact Stein database migrations\n6665bd8cb9 apidb: Compact Rocky database migrations\n86b423f9f3 apidb: Compact Queens database migrations\n31f01aa44f apidb: Compact Pike database migrations\n743693c740 apidb: Compact Ocata database migrations\n9020896cb7 docs: Add admin docs for configuring and updating machine types\n0b44c7ccef libvirt: Add a config update workflow test for [libvirt]hw_machine_type\n947f813e56 Fix ImageBackendFixture not to support direct_snapshot\n8cddd243bf nova-status: Add hw_machine_type check for libvirt instances\n9a5b07d9fc nova-manage: Add libvirt list_unset_machine_type command\nc70cde057d nova-manage: Add libvirt update_machine_type command\n20692c245c nova-manage: Add libvirt get_machine_type command\nf0eeba5c8c hyper-v rbd volume support\n95fbbce8b2 requirements.txt: Bump os-brick to 4.2.0\n98713d69ac libvirt: Deprecate disable_native_luksv1 and rbd_volume_local_attach\ne64744b92f rpc: Rework \u0027get_notifier\u0027, \u0027wrap_exception\u0027\nc23bda400a libvirt: Remove dead code\n59986cfd4f Replace md5 for fips\nc8478e40bd Remove short call timeout from func test\ndb7517d5a8 scheduler: Add \u0027socket\u0027 PCI NUMA affinity policy trait\nbe80dfdc71 objects: Add \u0027socket\u0027 PCI NUMA affinity\nfef371a425 Follow up for QoS interface attach\n5d3fb6cdbd Docs: Correct ``Password injection using the dashboard`` Explanation\n96072a1ef1 libvirt: parse alias out from device config\na634103b15 libvirt: allow querying devices from the persistent domain\nf5cd6e2dea libvirt: add AsyncDeviceEventsHandler\n4a70fc9cfb libvirt: Define and emit DeviceRemovedEvent and DeviceRemovalFailedEvent\n95b9481aa4 libvirt: start tracking NUMACell.socket for hosts\n4e23d93410 libvirt: Parse the \u0027os\u0027 element from domainCapabilities\nf7e3e17991 tests: Poison os.uname\nb367cf114c tests: Remove unnecessary mock\n1053ae0b61 tests: Add \u0027nova.virt.libvirt.utils.get_arch\u0027 stub to fixture\n5c35a5905e FUP: Catch and reraise routed nets exception\na1673d3118 Add a routed networks scheduler pre-filter\nc4b28a5496 Add net \u0026 utils methods for routed nets \u0026 segments\n123f6262f6 Handle instance \u003d None in _local_delete_cleanup\nf88a8b0c1b Fixes the elapsed time logged during a live migration\n838370a490 libvirt: add IP address to libvirt guest metadata\nf7975d640c Add regression test for bug 1914777\nd23cf26ec7 Change API unexpected exception message\n85cb4af075 tox: Add passenv DISABLE_CHERRY_PICK_CHECK to pep8\n5b20c628b2 libvirt: Record the machine_type of instances in system_metadata\n21656d4e9c Run nova-live-migration job with neutron-trunk enabled\n172024db71 Centralize sqlite FK constraint enforcement\n2e8e04a8f6 libvirt: Don\u0027t drop CPU flags with policy\u003d\u0027disable\u0027 from guest XML\n94b5db6d2e Make a couple test jobs run async devstack\n18ad9d97ab Remove unused Instance method\n5af28c4de5 Support interface attach with QoS ports\n9bdfdf7222 Uncap PrettyTable\n302f2bc4f7 apidb: Compact Newton database migrations\nd2449232bd apidb: Compact Mitaka database migrations\nd97fd31131 apidb: Compact Liberty database migrations\na172b75587 apidb: Add manage.py script\n35b58a5035 cmd: Remove \u0027nova-manage db ironic_flavor_migration\u0027\n737cae76e4 conf: Clean up docs for scheduler options\n9d53f366ff conf: Drop \u0027[scheduler] periodic_task_interval\u0027\n5aeb3a3874 Drop support for custom schedulers\n81b689e5bb fup: Merge duplicate volume attachment checks\n1252588d4e api: Reject volume attach requests when an active bdm exists\nee3a8f0225 Add regression test for bug #1908075\ncea5bf3d93 Remove __unicode__() from nova unit test Exception\nf6bacd3fde Refactor and rename test_tcp_rst_no_compute_rpcapi\n9e96f64126 Rename ensure_network_metadata to amend requested_networks\nedaaac6aaf Add requested_networks field to RequestSpec object\n2d4eff0ef7 cmd: Remove \u0027nova-manage db null_instance_uuid_scan\u0027\n7961cc3428 libvirt: Always enable USB controller on PPC64\n21af4dd4bb libvirt: Stub out \u0027nova.virt.libvirt.utils.get_arch\u0027\n96cf885ecd zuul: Mark nova-lvm as voting\n7d66f97e0a zuul: Increase SWIFT_LOOPBACK_DISK_SIZE within nova-lvm job\n936039d386 [FUP] this change removes an unused paramater\na19d25b67f docs: Move the LibvirtDistroSupportMatrix wiki page into our docs\ncbe3cd9dc3 Restart glance after updating policy in job\n25a632a4e1 Disallow CONF.compute.max_disk_devices_to_attach \u003d 0\nf9901ca927 glance: Remove [glance]/allowed_direct_url_schemes\nb62a1abd61 libvirt: Use specific user when probing encrypted rbd disks during extend\n7062e1db8b tox: Enable parallel docs build\n5beec0feab libvirt: Rationalize attachment of USB tablet\ndb8213af83 api: Reject requests to detach a volume when the compute is down\na3697ed013 Add regression test for bug #1909120\ncb56ae6aad Fix ForbiddenWithAccelerators to HTTPForbidden for shelve API\nee603927b0 libvirt: Remove \u0027hypervisor_version\u0027 from \u0027libvirt_info\u0027\n6b4db089b2 zuul: Allow encryption tests in nova-ceph-multistore again\n95724bbaef libvirt: Bump MIN_{LIBVIRT,QEMU}_VERSION and NEXT_MIN_{LIBVIRT,QEMU}_VERSION\n3a390c2c82 libvirt: Drop support for Xen\nd02ce3c4f0 libvirt: Drop support for UML\n0aff191514 libvirt: Remove MIN_LIBVIRT_VIR_ERR_DEVICE_MISSING\nd5420bbb50 docs: Add reference docs for internal block device structures\n684780ac0d libvirt: Load and cache volume drivers on-demand\n0f558749d1 libvirt: Stop NVMe and ScaleIO unit tests from calling os-brick\n6eae97843a zuul: Skip test_attach_cloned_encrypted_volume in nova-ceph-multistore\naf15488be9 add openstack-python3-wallaby-jobs-arm64 job\n74f5eb2ce9 Add generate schemas tool\ndda179d3f9 Do not inherit os_glance properties on snapshot\nc3804efd42 Refactor ResourceRequest constructor\ne6ff3651e6 Enable mypy on scheduler/utils.py\nbe9dd3d9db Refactor update_pci_request_spec_with_allocated_interface_name\n20fe077343 Factor out port resource deallocation\nb7f133898e Make remove allocation symmetric with add allocation\n54865a17cd Allow extending an existing instance allocation\n672b288324 Refactor _claim_pci_device_for_interface_attach to prepare for qos\n160e229f85 Move query param generation to RequestGroup o.vo\n534b4f3895 api: Log os-resetState as an instance action\n40e0578683 api-ref: Clarify \u0027all_tenants\u0027 command\n7fbd787b1b Cyborg shelve/unshelve support\na72ace1c6c Reuse code from oslo lib for JSON policy migration\n594df9aeb0 use psycopg2 binary instead of source package\n9046f0fff4 utils: Remove periods from instance hostnames\n5f93b680e7 Fix typo in warning message\ndc09bcd311 libvirt: Mock get_arch during some stable rescue unit tests\n7aa7fb94fd tests: Clean up \u0027test_hypervisors\u0027\n1f67ce2496 api: Drop statistics-style fields from os-hypervisors\nef7598ac28 api: Normalize exception handling for os-hypervisors\n5c9c81e0f9 db: Compact Train database migrations\nf0175a346a db: Compact Stein database migrations\n02d6a96588 db: Compact Rocky database migrations\nd83b02c5f6 db: Compact Queens database migrations\n7674fc6cd8 db: Compact Pike database migrations\neb5a1e4d72 db: Compact Ocata database migrations\nc57cc047ee db: Compact Newton database migrations\n76e8653ba1 db: Compact Mitaka database migrations\n1da542ff7a db: Compact Liberty database migrations\nb4fc1c141b db: Compact Kilo database migrations\n15bb50daff db: Compact Juno database migration\n833b7b60dd db: Compact Icehouse database migrations\n4701d28da3 db: Drop \"remove\" of index\nb7ba576cdd db: Merge in unique constraints\nd693d755db db: Clean up base migration\nf86bf703e5 tests: Remove \u0027test_extended_hypervisors\u0027\nc960fee835 Use the non polling notification waiter in func test\n6c6bca0326 Remove dead code from SchedulerReportClient\nb4f560d4e1 Improving the description for unshelve request body\n606a5dfecc api-ref: Add notes about volume attach and detach being async\n800254c0bc Run the db migration tests in the same test worker\n0be00cd58a Trival Change: Remove redundant code in instance delete\n9a0bc6e9cf Revert \"nova-multi-cell: Skip test_cold_migrate_unshelved_instance\"\nb247e69923 stabilize set_host libvirt driver unit test\nc501e23cb7 stabilize unit test asserting log\n1c056c384d Fallback to same-cell resize with qos ports\nf96ade2726 Reproduce bug 1907522 in functional test\n8b33ac0644 only wait for plugtime events in pre-live-migration\n00ad3bb1f1 tests: Merge \u0027test_hypervisor_status\u0027 into \u0027test_hypervisors\u0027\n032bbc3960 Refactoring of functional.regression.test_bug_1702454\n7c9edc02ed Initialize global data separately and run_once in WSGI app init\n0617cebcc9 Fix a deprecation warning of \u0027collections\u0027\nbd0ead2d63 Add a description in the PTL guide\n1cf2431f4b Remove six.text_type (2/2)\n383e2a8bdc Remove six.text_type (1/2)\n07462dd005 Remove six.binary_type/integer_types/string_types\nf02899418d libvirt: make cross cell resize spawn from snapshot image\neaa98e3340 nova-grenade-multinode: Skip test_live_block_migration_paused\n048250a4b7 Disable qos resize tempest test for nova-multi-cell job\n01f0424930 Enhance simple usage API documentation\n1811731424 nova-evacuate: Remove leftover playbook from standalone job\n7a9d4df24b lower-constraints: Bump packaging to 20.4\n610396f8ad tools: Allow check-cherry-picks.sh to be disabled by an env var\n95da142a15 Remove outdated comment from tox.ini\n0b942dcaa5 Omit resource inventories from placement update if zero\n3cb1215f19 Docs: correct cpu_thread_policy explanation\nc11f4e5336 nova-multi-cell: Skip test_cold_migrate_unshelved_instance\nf0ffed8c7f libvirt: check for AMD SEV only on x86-64\na107a5099e libvirt: Skip encryption metadata lookups if secret already exists on host\n4bea68f12b Revert \"zuul: Skip test_attach_scsi_disk_with_config_drive in nova-ceph-multistore\"\n671dc5bfe8 pci: Improve testing of \u0027nova.pci.request\u0027 module\nd81ef45041 pci: Add logging for filtering\nc2357ab9f3 pci: Improve docs for \u0027_filter_*\u0027 functions\n76360e566b nova-live-migration: Disable *all* virt services during negative tests\n7f7a7a163e zuul: Add devstack-plugin-ceph-compute-local-ephemeral to experimental\n26c46a409f compute: Don\u0027t detach volumes when RescheduledException raised without retry\n836e13cd57 zuul: Skip test_attach_scsi_disk_with_config_drive in nova-ceph-multistore\nadc705ee0c pci: Add a \u0027_filter_pools\u0027 helper function\n9c0ea4a901 doc: Fix rendering in the PTL guide\nd38b7e13a6 Support sys.argv in wsgi app\n454b8d30f4 Fix config option default value for sample config file\n9d1683f70f virt: Remove \u0027get_all_bw_counters\u0027 API\n1273c5ee0b Make PCI claim NUMA aware during live migration\n68bc87876f virt: Remove \u0027reset_network\u0027 API\n67454b307b Fix invalid argument formatting in exception messages\ned756d07b1 Fix the instance_uuid for test\n528740128a Revert \"Temporarily disable parts of heal port allocation test\"\n30067be9bd virt: Remove \u0027change_instance_metadata\u0027 API\n160ed6ff65 Add missing exception\ne13e8c8cd4 api-ref: Move \u0027os-agents\u0027 API to obsolete section\n65fb14c36e doc: require openstack client change for every new API microversion\n8c9d6fc8f0 Ignore PCI devices with 32bit domain\n976ac722d3 Reproduce bug 1897528\n2c28b74340 [Trivial]Add missing white space in the log message\na1ec6e69fa doc: Update the PTL guide\n433bee58bc Restore retrying the RPC connection to conductor\n52c8137b4b [docs] Fix a placement client\u0027s command\n1d23b358d6 functional: Wait for revert resize to complete\n3fcaf579a2 Add functional-py39 testing\nbf47afd70a Update contributor guide for Wallaby\n798f06abb2 Remove [database]connection defaulting\n3b44275868 Improve error handling during service level check\n66c7f00e1d Fix unplugging VIF when migrate/resize VM\nb62f9a04a0 Add os-volume_attachments reference docs\nd718adb108 functional: Expand SR-IOV live migration tests with NUMA\nec8732a4b8 functional: Add live migration tests for PCI, SR-IOV servers\nc163205489 Remove compute service level check for qos ops\n4916ab7a4c rbd: Only log import failures when the RbdDriver is used\nbe752b8175 zuul: Reintroduce nova-dsvm-multinode-base\nf6d74eabfc Remove six.byte2int/int2byte\n28ad7c517d Remove six.iteritems/itervalues/iterkeys\n1d0a0e8c20 Remove six.moves\n4ac4a04d18 zuul: Replace nova-live-migration with zuulv3 jobs\nc0fe95fcc5 zuul: Merge nova-evacuate into nova-multinode-live-migration\n8d9f298f4c Fix the compute RPC 5.12 issue\nf2ebe107d1 Bump the lowest eventlet version to 0.26.1\n8f79afd448 Add a regression test for 5.12 compute API issue\n3b8257cd23 Add upgrade check about old computes\n7675964af8 Set instance host and drop migration under lock\n3f348602ae Reproduce bug 1896463 in func env\nf419f1a4b1 Use _evacuate_server helper in func test\naa7c6f8769 Prevent starting services with older than N-1 computes\n033a325578 Remove unused bindir config option\nc82ce37635 Replace md5 with oslo version\ne7c7a2851e Modify glance\u0027s copy_image permission for nova-ceph-multistore\n7bd2bef3b4 tox: Stop linting as part of docs target\nc716e7608e releasenotes: Add fixes note for bug #1882521\n0d2ca53bb8 libvirt: Only ask tpool.Proxy to autowrap vir* classes\nb32a685595 docs: Make JSON valid\n5b9d587e74 trivial: Remove irrelevant comment\nf5a68826c7 [doc]: Fix glance image_metadata link\n2317b92f4e Add regression test for bug #1899835\ne728fe668a Use subqueryload() instead of joinedload() for (system_)metadata\nfbf6178762 Ignore InstanceNotFound exception when host rejoin cluster\n78ad4a4efd Add placeholder migrations for Victoria backports\neeeca4ceff Handle disabled CPU features to fix live migration failures\n2bd8900d9b Fix the vGPU dynamic options race\nb8695de6da Update pci stat pools based on PCI device changes\ne620e45790 CinderFixture: Return a driver_type of `fake` within connection_info\n1f06418549 Remove self.server dependency form _evacuate_server\n770c139edb Add regression test for bug #1899649\n4a4f12678a Fix virsh domifstat to get vhostuser vif statistics\n1c256cf774 Prevent archiving of pci_devices records because of \u0027instance_uuid\u0027\n82528c83ac hardware: Check inventory of shared CPUs for \u0027mixed\u0027 policy\n9e1fe29d76 functional: Add test for #1898272\n583e85622b functional: Add tests for mixed CPU policy\n12cc805d36 objects: Fix issue in exception type\n450213f1f0 Clean up the DynamicVendorData constructor\n1390eecf8d Use cell targeted context to query BDMs for metadata\ncf0e519a0e Allow excluding image type capabilities\nf44700935f Include removal of ephemeral backing files in the image cache manager\n6cf449bdd0 compute: Lock by instance.uuid lock during swap_volume\n383bcbca70 optimize the shelve code flow\n1e10461c71 Follow up for cherry-pick check for merge patch\nc34a17db6f libvirt: Only add a USB controller if it\u0027s necessary\n5b0343d3e1 libvirt: Remove support for \u0027[libvirt] use_usb_tablet\u0027\nf2ca089bce libvirt: \u0027video.vram\u0027 property must be an integer\ndd1e6d4b0c libvirt: Increase incremental and max sleep time during device detach\naa16dd09eb libvirt: Log exception when unable to import rbd or rados\nead034d969 bindep: Install python3 and python3-devel on CentOS 8 and Fedora\n015f7dadcc libvirt: Remove MIN_{LIBVIRT,QEMU}_PMEM_SUPPORT\n0e7cd9d1a9 libvirt: Remove MIN_LIBVIRT_VIDEO_MODEL_VERSIONS\n226250beb6 nova-evacuate: Disable libvirtd service and sockets during negative tests\nd309e3cdf5 Temporarily disable parts of heal port allocation test\n8d3c2ce92b Add a hacking rule for assert_has_calls\n1175081294 Fix a hacking test\n6b22b697f7 libvirt: Remove MIN_LIBVIRT_BETTER_SIGKILL_HANDLING\n6c8e287156 libvirt: Remove MIN_{LIBVIRT,QEMU}_NATIVE_TLS_VERSION\n99850a0481 Add reproducer for bug #1894095\ned4a748bda Add Python3 wallaby unit tests\n851cc12f67 Update master for stable/victoria\nf357d80407 zuul: Introduce nova-evacuate\n73cc85d910 Log stack trace when fails launching a defined domain\n416cd1ab18 libvirt: make mdev types name attribute be optional\n34cc68ac6e libvirt: Remove MIN_LIBVIRT_FILE_BACKED_DISCARD_VERSION\n48123d6321 hyperv: Configures chassis asset tags for VMs.\n50ff47d553 remove python warnning  from tox\nb933312579 Use absolute path during qemu img rebase\n4861eebc91 fixtures: Handle binding of first port\n0fa6f4e7e1 objects: Remove Agent, AgentList\n7ac52e643c api: Remove \u0027os-agents\u0027 API\n8aea747c97 virt: Remove \u0027get_per_instance_usage\u0027 API\n83ae149c72 virt: Remove various aggregate APIs\ncaa5f9ef94 virt: Remove \u0027set_bootable\u0027 API\n39fe22161c virt: Remove \u0027post_interrupted_snapshot_cleanup\u0027 API\n912b717207 virt: Remove \u0027get_console_pool_info\u0027 API\n696c8a95f6 Raise 409 when removing security group from instance\n17184e7d56 vmware: Use cookiejar from oslo.vmware client directly\ndf8a658bd9 vmware tests: Support different moref backend representations\nae8c3067c5 libvirt: Use local variable instead of CONF\nb6e9023751 Fix misleading documentation for live_migration_inbound_addr\nde10c126e9 Improve documentation of (unpin|pin)_cpus_with_siblings()\n5477c0045f Add an online migration for PciDevice.uuid\n9d4725bcb6 Modify PciDevice.uuid generation code\n76c60c25bc Update supported transports for iscsi connector\n\nChange-Id: Idf8d6c4221702c35e835a8dcf6a9cdd868ddbd55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/53e1fe0fbc088f068d42e221cb2e390c52984499"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/53e1fe0fbc088f068d42e221cb2e390c52984499"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
