)]}'
{"id":"openstack%2Freleases~781750","triplet_id":"openstack%2Freleases~master~Id8abfe08d237b587ecae32389c00f22635961c92","project":"openstack/releases","branch":"master","topic":"wallaby-rc1-deadline","hashtags":[],"change_id":"Id8abfe08d237b587ecae32389c00f22635961c92","subject":"Proposing Wallaby RC1 for sahara","status":"MERGED","created":"2021-03-19 14:49:44.000000000","updated":"2021-03-22 09:54:58.000000000","submitted":"2021-03-22 09:54:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"781750-wallaby-rc1-deadline","meta_rev_id":"417e6b20a189884268c980ff0db69556adc3e94e","_number":781750,"virtual_id_number":781750,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-22 09:54:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2021-03-22 08:57:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2021-03-19 19:09:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":2,"date":"2021-03-22 08:22:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2021-03-22 08:57:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-03-19 19:12:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-19 14:51:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-19 15:21:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-19 15:25:17.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"state":"REVIEWER"},{"updated":"2021-03-22 08:57:25.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"6907e5239a3e84dbacdbffb74ab1dc1d1bb7688c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-19 14:49:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f64d7a467cbb651279543a7f8988f1e66d90175","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 15:21:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16c2797a390f4975be8b4e4f4f79c4af : SUCCESS in 6m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/519ba3a3049a489fb44cb4e7f4cd30e0 : SUCCESS in 5m 19s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f3a292720e594f84af3329f1ce4371a8 : SUCCESS in 9m 18s","accounts_in_message":[],"_revision_number":1},{"id":"d0d7c1a69798c797785c9e0f9158f67b59ebe49a","author":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"date":"2021-03-19 19:09:12.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3cfebc9eadceabfdbbb56fc6b551ae325b685775","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 19:12:04.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"e2bda78b7b7d98b9acd1a88cd06926c5b55ad915","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-22 08:22:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d5f282e98bb506bab77a6c5634bdc44efeb78890","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-03-22 08:57:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"13759fab5123ebe742263be8eac9cdb62e7e8eac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 08:59:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7f9b762ed1b12be5078c5d419d4f57093bd38f78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 09:54:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dc3094396a54db780bdb87728fd2abe : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8cfabec375945b588547a7e880c345f : SUCCESS in 6m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7a4e2ef326674db48e3c207a54bfaabd : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":1},{"id":"417e6b20a189884268c980ff0db69556adc3e94e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 09:54:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f8d9d27ce51a5c17a8d37ab68866cafe57f404a2","revisions":{"f8d9d27ce51a5c17a8d37ab68866cafe57f404a2":{"kind":"REWORK","_number":1,"created":"2021-03-19 14:49:44.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/50/781750/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/50/781750/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/50/781750/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/50/781750/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/50/781750/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/50/781750/1"}}},"commit":{"parents":[{"commit":"753fc318dec0b4afaca6a4408e3d5ec7aaa734dc","subject":"Merge \"[OpenstackSDK] Create for all client and non-client lib wallaby branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/753fc318dec0b4afaca6a4408e3d5ec7aaa734dc"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-19 14:49:33.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-19 14:49:33.000000000","tz":60},"subject":"Proposing Wallaby RC1 for sahara","message":"Proposing Wallaby RC1 for sahara\n\nIf you wants to wait for a specific patch to make it to the release,\nsomeone from your team can -1 the patch to have it held, or update that patch\nwith a different commit SHA.\n\nPatches with no response will be merged by Friday March 26.\n\n$ git log --oneline --no-merges 13.0.0..f1861159\nf576d092 Fix lower requirements: bump various packages\nb8ad624e gate fix: no removed tail_log in devstack, disable grenade\nb4b666cb Fix the isolated exception message\nbd46fc6d Replace deprecated UPPER_CONSTRAINTS_FILE variable\n7376520e Add Python3 wallaby unit tests\n107233b4 Update master for stable/victoria\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: Id8abfe08d237b587ecae32389c00f22635961c92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f8d9d27ce51a5c17a8d37ab68866cafe57f404a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f8d9d27ce51a5c17a8d37ab68866cafe57f404a2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
