)]}'
{"id":"openstack%2Freleases~781766","triplet_id":"openstack%2Freleases~master~I811de90c4bc9cbe2d857c0fd556de367d59ab8bf","project":"openstack/releases","branch":"master","topic":"wallaby-rc1-deadline","hashtags":[],"change_id":"I811de90c4bc9cbe2d857c0fd556de367d59ab8bf","subject":"Proposing Wallaby RC1 for watcher","status":"MERGED","created":"2021-03-19 15:08:49.000000000","updated":"2021-03-22 09:56:16.000000000","submitted":"2021-03-22 09:56:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"781766-wallaby-rc1-deadline","meta_rev_id":"49ec519d20bc0ae4fe5cfb6ba534cb51ed056e17","_number":781766,"virtual_id_number":781766,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-22 09:56:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2021-03-22 08:57:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2021-03-22 08:23:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2021-03-21 03:11:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2021-03-22 08:57:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-03-21 03:11:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-19 15:13:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-19 15:21:37.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},"state":"REVIEWER"},{"updated":"2021-03-19 15:21:39.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"state":"REVIEWER"},{"updated":"2021-03-19 15:44:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-22 08:57:40.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"87dd05b7421ed6a853ccac66e5d84ed66d771d19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-19 15:08:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b4c0816aa2616de91d986664d261f441f17ec9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 15:44:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b56b0004fcf476db82238dd368938e1 : SUCCESS in 6m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b1a485f810354d2d8f0685cbd8df3593 : SUCCESS in 5m 12s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/2667cdc834f74c928d21cfcb6dd7dbeb : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":1},{"id":"8751d582fde1835aeae3e02b955118eff8516ae2","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2021-03-21 03:11:10.000000000","message":"Patch Set 1: Code-Review+1\n\nthanks","accounts_in_message":[],"_revision_number":1},{"id":"bcc216b951e0a485ac8287ab8a438b918759564b","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-21 03:11:59.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"24e5eb41f02d3f272e6b33206c350498398a5b3f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-22 08:23:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"aec15cefbdd7ec7081130b2801e02951f470a1e3","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-03-22 08:57:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"10fd4c361d7d62d82e48a298757e84ceeee29c83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 09:00:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c995f6fb84974ee0ef89ee0180fda3c3c5e429f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 09:56:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79f7cb53ce6e48caba2777b8ef59a12b : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8e9950fd98d4c72bd99b223c629ff0e : SUCCESS in 7m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7df0e979b19041a3a917513d8f68a84e : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":1},{"id":"49ec519d20bc0ae4fe5cfb6ba534cb51ed056e17","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 09:56:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8bc74175ddd2c4079645379e63070f949c04fe04","revisions":{"8bc74175ddd2c4079645379e63070f949c04fe04":{"kind":"REWORK","_number":1,"created":"2021-03-19 15:08:49.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/66/781766/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/66/781766/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/781766/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/781766/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/781766/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/66/781766/1"}}},"commit":{"parents":[{"commit":"753fc318dec0b4afaca6a4408e3d5ec7aaa734dc","subject":"Merge \"[OpenstackSDK] Create for all client and non-client lib wallaby branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/753fc318dec0b4afaca6a4408e3d5ec7aaa734dc"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-19 15:08:38.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-19 15:08:38.000000000","tz":60},"subject":"Proposing Wallaby RC1 for watcher","message":"Proposing Wallaby RC1 for watcher\n\nIf you wants to wait for a specific patch to make it to the release,\nsomeone from your team can -1 the patch to have it held, or update that patch\nwith a different commit SHA.\n\nPatches with no response will be merged by Friday March 26.\n\n$ git log --oneline --no-merges 5.0.0..86381515\n86381515 [goal] Deprecate the JSON formatted policy file\n76270c83 Fix gate requirement checks job\nec218989 incorrect name in unit test\ne61f9b5e Use common rpc pattern for all services\n63b6997c Drop lower-constraints\n262edc8c remove bandit B322 check\n204b2766 Fix missing self argument in instances_no_attached\naf02bebc Fix parameter passed to IronicNodeNotFound exception\n3aaa2090 Imported Translations from Zanata\n5097665b Remove the unused coding style modules\n09f6e3bd Remove usage of six\nf488636f Bump py37 to py38 in tox.ini\n16a04866 Remove six\nf8797a7f Add Python3 wallaby unit tests\nda283b49 Update master for stable/victoria\ne21e5f60 [goal] Migrate testing to ubuntu focal\ncca0d9f7 Implements base method for time series metrics\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I811de90c4bc9cbe2d857c0fd556de367d59ab8bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8bc74175ddd2c4079645379e63070f949c04fe04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8bc74175ddd2c4079645379e63070f949c04fe04"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
