)]}'
{"id":"openstack%2Freleases~781783","triplet_id":"openstack%2Freleases~master~I9b644744e9ff462bf42bc7a688fd8a19d3994fd6","project":"openstack/releases","branch":"master","topic":"wallaby-stable-branches","hashtags":[],"change_id":"I9b644744e9ff462bf42bc7a688fd8a19d3994fd6","subject":"Proposing python-openstackclient for wallaby","status":"MERGED","created":"2021-03-19 16:00:11.000000000","updated":"2021-03-20 09:12:52.000000000","submitted":"2021-03-20 09:12:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"781783-wallaby-stable-branches","meta_rev_id":"d3fc1481ab063dd953778f267943105dd55c3918","_number":781783,"virtual_id_number":781783,"owner":{"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-20 09:12:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2021-03-20 09:00:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2021-03-19 16:29:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2021-03-19 16:22:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2021-03-20 09:00:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-03-19 16:24:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-19 16:03:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-19 16:21:02.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2021-03-19 16:22:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-19 16:29:40.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-19 17:51:15.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"72ca5b932e2dc98bbdd3b013fbbfda01689cc3f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},"date":"2021-03-19 16:00:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e4f43df4f4f577736c0337422a3c1fdb114dcae6","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2021-03-19 16:22:07.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e0108361b047217652e3a340c9327ac34217ed13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 16:22:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f10ad3c37f484daf86f1ed0da84bd59f : SUCCESS in 7m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/732599e3d07646b6b9d56b105e664796 : SUCCESS in 6m 45s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f97e80f74b0b4b3eb7468580f6211dc9 : SUCCESS in 6m 28s","accounts_in_message":[],"_revision_number":1},{"id":"4dbbd7682795e1f3fe0bd81375e0d21b37596dff","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 16:24:14.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"cf2ed584ac015a00a431956fb01d6ab4c39f206d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-19 16:29:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cbc732f3bc5cae6f1bde95ea2eab13bf0148354d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-19 17:51:15.000000000","message":"Patch Set 1: Code-Review-1\n\nCan we have a double-check for the version? Reno suggest major version bump [1] based on \u0027release notes\u0027.\n\n[1] https://zuul.opendev.org/t/openstack/build/f97e80f74b0b4b3eb7468580f6211dc9/log/tox/list-changes-results.log#120-122","accounts_in_message":[],"_revision_number":1},{"id":"9df4efe2122772e13a654bbc0febc162e43f7872","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2021-03-19 18:16:09.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e Can we have a double-check for the version? Reno suggest major version bump [1] based on \u0027release notes\u0027.\n\u003e \n\u003e [1] https://zuul.opendev.org/t/openstack/build/f97e80f74b0b4b3eb7468580f6211dc9/log/tox/list-changes-results.log#120-122\n\nUhm, actually it is still ok.\n\nFrom SemVer:\nMINOR version when you add functionality in a backwards compatible manner,\n\nHere the new feature was added not affecting anything else, what we do not consider worth of major release due to absence of breaking changes and any other new features.\n\nMaybe I should have used \"other\" in the reno for the corresponding change to avoid that.","accounts_in_message":[],"_revision_number":1},{"id":"1d7a433910c9e73a11aadb7722539539f339d879","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-19 18:56:41.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1: Code-Review-1\n\u003e \u003e \n\u003e \u003e Can we have a double-check for the version? Reno suggest major version bump [1] based on \u0027release notes\u0027.\n\u003e \u003e \n\u003e \u003e [1] https://zuul.opendev.org/t/openstack/build/f97e80f74b0b4b3eb7468580f6211dc9/log/tox/list-changes-results.log#120-122\n\u003e \n\u003e Uhm, actually it is still ok.\n\u003e \n\u003e From SemVer:\n\u003e MINOR version when you add functionality in a backwards compatible manner,\n\u003e \n\u003e Here the new feature was added not affecting anything else, what we do not consider worth of major release due to absence of breaking changes and any other new features.\n\u003e \n\u003e Maybe I should have used \"other\" in the reno for the corresponding change to avoid that.\n\nThanks Artem, then I think that we can continue a minor version","accounts_in_message":[],"_revision_number":1},{"id":"eadc7d78152a92ee1f4983b64882220606677944","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-19 19:22:06.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1: Code-Review-1\n\u003e \u003e \n\u003e \u003e Can we have a double-check for the version? Reno suggest major version bump [1] based on \u0027release notes\u0027.\n\u003e \u003e \n\u003e \u003e [1] https://zuul.opendev.org/t/openstack/build/f97e80f74b0b4b3eb7468580f6211dc9/log/tox/list-changes-results.log#120-122\n\u003e \n\u003e Uhm, actually it is still ok.\n\u003e \n\u003e From SemVer:\n\u003e MINOR version when you add functionality in a backwards compatible manner,\n\u003e \n\u003e Here the new feature was added not affecting anything else, what we do not consider worth of major release due to absence of breaking changes and any other new features.\n\u003e \n\u003e Maybe I should have used \"other\" in the reno for the corresponding change to avoid that.\n\n@Artem: actually the following reno entry suggested the major bump:\n\nfound breaking change in \u0027upgrade\u0027 section of releasenotes/notes/add-missing-server-list-opts-c41e97e86ff1e1ca.yaml\n\nwhich says \"The ``openstack server list --status`` parameter will now validate the requested status.\".\n\nThis really feels OK to me as only a minor bump, but please give a final confirmation that you wrote about the same change, and I\u0027ll approve the patch. Thanks!","accounts_in_message":[],"_revision_number":1},{"id":"dd98b508d799d1f72dfe3915e10e30f4f8f68aa9","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2021-03-19 19:39:21.000000000","message":"Patch Set 1:\n\nYes, it\u0027s ok","accounts_in_message":[],"_revision_number":1},{"id":"36a34df2dcbe47c856d47546f161ffaf779c8dec","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-20 09:00:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThanks Artem \u0026 sorry for holding back the release.","accounts_in_message":[],"_revision_number":1},{"id":"bbe320a051d31dff1c16a519e3ad6ca193925618","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-20 09:00:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7e866a44b57fd944e9211d3cceef916be4a0c5fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-20 09:12:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86d48a9d6f7444abae057f735a949d07 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc7cca173e084ae8bc2369569665c0ad : SUCCESS in 7m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/29f9ed723c214ed88c531b63194fa510 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":1},{"id":"d3fc1481ab063dd953778f267943105dd55c3918","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-20 09:12:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0248d3a8898b8eadb78590d8b06f5306b3979511","revisions":{"0248d3a8898b8eadb78590d8b06f5306b3979511":{"kind":"REWORK","_number":1,"created":"2021-03-19 16:00:11.000000000","uploader":{"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},"ref":"refs/changes/83/781783/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/83/781783/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/781783/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/781783/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/781783/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/83/781783/1"}}},"commit":{"parents":[{"commit":"753fc318dec0b4afaca6a4408e3d5ec7aaa734dc","subject":"Merge \"[OpenstackSDK] Create for all client and non-client lib wallaby branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/753fc318dec0b4afaca6a4408e3d5ec7aaa734dc"}]}],"author":{"name":"Armstrong Foundjem","email":"foundjem@ieee.org","date":"2021-03-19 16:00:05.000000000","tz":-240},"committer":{"name":"Armstrong Foundjem","email":"foundjem@ieee.org","date":"2021-03-19 16:00:05.000000000","tz":-240},"subject":"Proposing python-openstackclient for wallaby","message":"Proposing python-openstackclient for wallaby\n\nProposing release for cycle-with-intermediary deliverables that have not released yet.\n\nIf you wants to wait for a specific patch to make it to the release, someone from your team can -1 the patch to have it held, or update that patch with a different commit SHA.\n\nPatches with no response will be merged by Friday March 26.\n\n$ git log --oneline --no-merges 5.4.0..86bca18b\n791bed6d Update the file paths mentioned in README.rst\n2ccf7727 compute: Remove \u0027file://\u0027 prefix from \u0027--block-device\u0027\ned731d6c network: Add missing subnet unset --gateway \u003csubnet-id\u003e\nd3bd0146 compute: Add support for loading BDMs from files\n7c1d6f76 compute: Add functional tests for --block-device\n16c72f86 Add --name to port list\ne8509d81 Add \u0027address_group\u0027 type support to rbac commands\n119d2fae project cleanup\n70480fa8 compute: Remove deprecated \u0027server migrate --live\u0027 option\n6f3969a0 compute: Deprecate \u0027server create --file\u0027\n8868c77a compute: Stop silently ignore --(no-)disk-overcommit\n2bdf34dc compute: Auto-configure shared/block live migration\nace4bfb6 compute: Add \u0027server create --block-device\u0027 option\nf2deabb1 compute: Remove references to optional extensions\n4da4b962 compute: Add missing \u0027server create\u0027 options\n074e045c compute: Improve \u0027server create --block-device-mapping\u0027 option parsing\na52beaca compute: Rename \u0027server migrate (confirm|revert)\u0027\nf80fe2d8 compute: Add \u0027server migration show\u0027 command\n32ae1857 Rename FakeServerMigration to FakeMigration\n9ed34aac compute: Add support for \u0027server boot --nic ...,tag\u003d\u003ctag\u003e\u0027\nc7d58237 compute: Improve \u0027server create --nic\u0027 option parsing\nd6c9b7f1 compute: Shuffle options for \u0027server create\u0027\n7ed4f68c compute: Add missing options for \u0027server event list\u0027\n5ec4d4c7 compute: Add missing options for \u0027server group list\u0027\n1a6df700 compute: Add \u0027server * --all-projects\u0027 option\n0cc878e5 Add device profile to ``port``\ndfa869ed compute: Improve \u0027server migration list\u0027 options\n8a164bb0 compute: Add \u0027--force\u0027 option to \u0027server delete\u0027\n262e525a compute: Add missing options for \u0027hypervisor list\u0027\ne01e59ca Support remote-address-group in SG rules\nca7f23d0 compute: Add \u0027server volume update\u0027 command\ndd89efd5 network: Address nits for I3c313fc9329837dde67815901528a34dca98ebcc\nbb15b291 Add reno for change Ic3c555226a220efd9b0f27edffccf6c4c95c2747\nd33eb3e1 Remove retired Karbor support\n2b073c20 Add \u0027server unshelve --wait\u0027 option\n64c2a1a4 Add \u0027server shelve --offload\u0027, \u0027server shelve --wait\u0027 options\nd5026278 compute: Add \u0027server volume list\u0027 command\n95834473 compute: Add missing options for \u0027server image create\u0027\nb3490572 tests: Remove unused fake method\nfc24142e compute: Add missing options for \u0027keypair list\u0027\n8a0f3fc6 compute: Add missing options for \u0027server set\u0027\nd0112a80 compute: Add missing options for \u0027server list\u0027\nbbf7de83 trivial: Use plural for appended parameters\nf2007998 compute: Add \u0027server show --topology\u0027 option\na5c6470f compute: Add \u0027server group create --rule\u0027 option\nbf834f6d compute: Fix \u0027hypervisor show -f yaml\u0027 output\ne2a9a960 compute: Fix \u0027server group * -f yaml\u0027 output\naf5e9d16 compute: Fix \u0027usage * -f yaml\u0027 output\n03776d82 compute: Fix \u0027server * -f yaml\u0027 output\n6f616a29 Add support \u0027--progress\u0027 option for \u0027image create\u0027\nf57e10b9 Support Neutron Address Group CRUD\n29a7c9af image: Unset properties rather than setting to None\n3c80b1b3 Add project field in image list subcommand\n05456223 trivial: Cleanup docs for \u0027server list\u0027\nf9fd3642 compute: Add missing options for \u0027server rebuild\u0027\nda03bd80 Add \u0027flavor list --min-disk\u0027, \u0027--min-ram\u0027 options\nf5b185c3 Make use of comparable \u0027FormattableColumn\u0027 subclasses\na79e7db4 Remove retired Searchlight support\necfda765 Update lower-constraints\n20769cd7 Fix lower-constraints job\nceaba4c5 trivial: Cleanup docs for \u0027server rebuild\u0027\nd688cb58 Add documentation about login with federation\nf36a34b6 Switch compute aggregate functions to SDK\n0f4f42b6 Switch compute flavors from novaclient/direct to SDK\n284c38bc Let autoprogram-cliff know who\u0027s running\n3e8968af Add NODE and HOST parameters in \"server create\" help text\n0f02029d Add option to filter instances by AZ\n5bdcd590 stop image downloads to memory\n8387b114 Add \"fields\" parameter to ListPort query\na5101a41 trivial: Document removal of support for agents\nebaf0eae tests: Remove \u0027agent\u0027 functional tests\n4c0bfb03 Allow to resize in-use volumes\n1a5dd4af Resolve issues with \u0027server migration list\u0027\n512ba114 Switch \u0027openstack keypair\u0027 ops to use SDK\nffd7e939 functional: Remove test for \u0027quota set --force\u0027\n0a7f2692 Remove references to Python 2.7\n311f4130 Add a few selectable fields to the \"openstack server list\" output\n01eb4e83 Add \u0027openstack server evacuate\u0027 command\nad3369ed Fix formatting of the flavor properties\n17f641e1 Compute: Add user id support for keypair\n7c239403 trivial: Rework \u0027CreateServer\u0027 function\n9b02f37e update lower-constraints.txt\n5b04a86d Replace deprecated UPPER_CONSTRAINTS_FILE variable\ne05d39ab Switch console url show operations to SDK\n5f650853 Remove references to setuptools\nf464bba7 Remove the unused coding style modules\n5645fad7 Add support for \u0027keypairs list --project\u0027 parameter\n98a0016c Add support for \u0027keypairs list --user\u0027 parameter\n2f76bfa3 Compute: Add tags support for server\nab0b1fe8 Validate \u0027server group create --policy\u0027 option\nbf35f046 Add \u0027openstack server migration list --type\u0027 option\n5fd399ea Cleanup of \u0027server migration list\u0027 command\n08b0e585 Add \u0027server migration force complete\u0027 command\n1c3cf113 Add \u0027server migration abort\u0027 command\nfd9a235d compute: Add --password option for openstack server create\n9385113d Remove oslo.utils\n671f7369 zuul: Stop testing against Tempest\n415545ab Add an error message when server bind floating IP\n742c80a8 Compute: Add tag support for server add fixed ip\n6f160231 Compute: Add tag support for server add network\n4855fef8 Compute: Add \u0027keypair create --type\u0027 parameter\nf3fbb1b6 Compute: Add tag support for server add port\n1c7fe3b6 Compute: Add tag support for server add volume\nc2df9215 Remove usage of six\n74db8dd6 Switch openstack console log show operation to use OpenStackSDK\nf50bd408 Fix: port attribute name propagate_uplink_status\n17678c9b Restore behavior of image create with same name.\n098a3fe2 Add Python3 wallaby unit tests\n63b46ac5 Update master for stable/victoria\nb77c28d2 Add server migration list CLI\ne9bd4ef0 Remove None valued network quota entries\n58f1c909 Add source_ip_prefix and destination_ip_prefix to metering label rules\n1e053bab Add id and enabled param in ListIdentityProvider parser\n12f1e56e Add \u0027openstack server create --use-config-drive\u0027\n8628e52d Add name and enabled param in ListDomain parser\n001796fa Change \u0027Volume\u0027 to \u0027Block Storage\u0027\ne410e61d Always display direction for security group rules\n\nSigned-off-by: Armstrong Foundjem \u003cfoundjem@ieee.org\u003e\nChange-Id: I9b644744e9ff462bf42bc7a688fd8a19d3994fd6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0248d3a8898b8eadb78590d8b06f5306b3979511"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0248d3a8898b8eadb78590d8b06f5306b3979511"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
