)]}'
{"id":"openstack%2Freleases~782932","triplet_id":"openstack%2Freleases~master~Idfb4c48ae9a8745434d6a5e8a2a0745219f6db36","project":"openstack/releases","branch":"master","topic":"release_osa","hashtags":[],"change_id":"Idfb4c48ae9a8745434d6a5e8a2a0745219f6db36","subject":"Release OpenStack-Ansible Ussuri","status":"MERGED","created":"2021-03-25 09:32:17.000000000","updated":"2021-03-25 16:08:26.000000000","submitted":"2021-03-25 16:08:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"782932-release_osa","meta_rev_id":"e5ec3959aeadf0d9f66342050eaf794572b61aa5","_number":782932,"virtual_id_number":782932,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-25 16:08:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2021-03-25 14:45:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2021-03-25 15:56:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2021-03-25 15:56:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-03-25 14:26:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-25 09:33:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-25 12:31:37.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-25 14:09:55.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2021-03-25 14:45:17.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"1023ef13c03108265caf31691ec1845539433785","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-03-25 09:32:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08481de554143d82b977c1dc37bc70910c5d1911","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 09:33:33.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"87286cbdd774dd0651f5386af43a386d684ff486","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 09:45:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcc3e70c88f64594bf2364deb9d353a0 : SUCCESS in 7m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3155849bc1e143029bf66163c69b6a14 : SUCCESS in 5m 00s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e53be7b3738c42bc9fccc447c07db5f5 : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":1},{"id":"91e97769f040c4752931203b5c2350be77857afc","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-25 12:31:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"86ad6893e3b98ec7f73cc5645111dced5c51a9c6","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-25 14:09:55.000000000","message":"Patch Set 1:\n\n@Dmitriy: is that intentional that the tag is not on the latest merged patch [1]? As I see there are 5 more patch that has merged since the patch you are planning to release as 21.2.4.\n\n[1] https://zuul.opendev.org/t/openstack/build/e53be7b3738c42bc9fccc447c07db5f5/log/tox/list-changes-results.log#220-229","accounts_in_message":[],"_revision_number":1},{"id":"55b9874208dab4f1d936b99b83af7f047d6d9b37","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-03-25 14:19:56.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e @Dmitriy: is that intentional that the tag is not on the latest merged patch [1]? As I see there are 5 more patch that has merged since the patch you are planning to release as 21.2.4.\n\u003e \n\u003e [1] https://zuul.opendev.org/t/openstack/build/e53be7b3738c42bc9fccc447c07db5f5/log/tox/list-changes-results.log#220-229\n\nEventually neither of these changes are affecting users and they are more optimizations for CI. But probably you\u0027re right and it would be more correct to release on 04ccaaca6 which is zuul rebase and actual merge was done.","accounts_in_message":[],"_revision_number":1},{"id":"2ec402dca690c07e035add4d4d03e610a7c0d93f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-03-25 14:20:36.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"915d774533006a3985918ea0a6fb99a71c88e021","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 14:26:50.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"c871e12270e6ce6d20fbe9d67a38ce75b66d88d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 14:36:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5821c215755f40369348b0f039f6c95d : SUCCESS in 8m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8a8bdc51bbc04a6aa23a194f1b0a1123 : SUCCESS in 7m 54s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9e700a6b79624de994f8609fd424cb9e : SUCCESS in 8m 36s","accounts_in_message":[],"_revision_number":2},{"id":"986cb10544dcfbca67fa6fe4a4503d3532d78d9d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-25 14:45:17.000000000","message":"Patch Set 2: Code-Review+2\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e @Dmitriy: is that intentional that the tag is not on the latest merged patch [1]? As I see there are 5 more patch that has merged since the patch you are planning to release as 21.2.4.\n\u003e \u003e \n\u003e \u003e [1] https://zuul.opendev.org/t/openstack/build/e53be7b3738c42bc9fccc447c07db5f5/log/tox/list-changes-results.log#220-229\n\u003e \n\u003e Eventually neither of these changes are affecting users and they are more optimizations for CI. But probably you\u0027re right and it would be more correct to release on 04ccaaca6 which is zuul rebase and actual merge was done.\n\nOk, thanks for the update. LGTM.","accounts_in_message":[],"_revision_number":2},{"id":"49aabf5a0d7e9d9157754a0e4efe7a74006202a0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-25 15:56:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"df8b020ae6d5bce01f9bbadfbc589b666b994976","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 15:57:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c5950468652295c7fe7cd4e651ffd23a80ca7b50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 16:08:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79f4f9fb8a604c6ca13165a51f94593b : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ecf538ce8cd4b6e9ee87cc5095fea62 : SUCCESS in 6m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d89f2769c92f4a10aa843769dea669dd : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":2},{"id":"e5ec3959aeadf0d9f66342050eaf794572b61aa5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 16:08:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e32433dc1fc2b9082dec9235cb7f7144472d4854","revisions":{"79b6c9bd25e3b11c9f01b55af71bc107dbf480e4":{"kind":"REWORK","_number":1,"created":"2021-03-25 09:32:17.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/32/782932/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/32/782932/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/32/782932/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/32/782932/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/32/782932/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/32/782932/1"}}},"commit":{"parents":[{"commit":"6dff229dd3b31e632f8ca89edb11bdc69c5adbe1","subject":"Merge \"Proposing Wallaby RC1 for manila-ui\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6dff229dd3b31e632f8ca89edb11bdc69c5adbe1"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-03-25 09:31:40.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-03-25 09:31:40.000000000","tz":120},"subject":"Release OpenStack-Ansible Ussuri","message":"Release OpenStack-Ansible Ussuri\n\nChange-Id: Idfb4c48ae9a8745434d6a5e8a2a0745219f6db36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/79b6c9bd25e3b11c9f01b55af71bc107dbf480e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/79b6c9bd25e3b11c9f01b55af71bc107dbf480e4"}]},"branch":"refs/heads/master"},"e32433dc1fc2b9082dec9235cb7f7144472d4854":{"kind":"REWORK","_number":2,"created":"2021-03-25 14:20:36.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/32/782932/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/32/782932/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/32/782932/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/32/782932/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/32/782932/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/32/782932/2"}}},"commit":{"parents":[{"commit":"6dff229dd3b31e632f8ca89edb11bdc69c5adbe1","subject":"Merge \"Proposing Wallaby RC1 for manila-ui\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6dff229dd3b31e632f8ca89edb11bdc69c5adbe1"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-03-25 09:31:40.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-03-25 14:20:27.000000000","tz":0},"subject":"Release OpenStack-Ansible Ussuri","message":"Release OpenStack-Ansible Ussuri\n\nChange-Id: Idfb4c48ae9a8745434d6a5e8a2a0745219f6db36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e32433dc1fc2b9082dec9235cb7f7144472d4854"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e32433dc1fc2b9082dec9235cb7f7144472d4854"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
