)]}'
{"id":"openstack%2Freleases~783338","triplet_id":"openstack%2Freleases~master~I666d621cf77a4a8c4d4b68cded14f919404d1702","project":"openstack/releases","branch":"master","topic":"wallaby-rc1-deadline","hashtags":[],"change_id":"I666d621cf77a4a8c4d4b68cded14f919404d1702","subject":"Proposing Wallaby RC1 for octavia","status":"MERGED","created":"2021-03-26 15:41:09.000000000","updated":"2021-03-26 19:02:19.000000000","submitted":"2021-03-26 19:02:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"783338-wallaby-rc1-deadline","meta_rev_id":"21264d7da597310b0f9cdc892e5c16c4f3a6500d","_number":783338,"virtual_id_number":783338,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-26 19:02:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2021-03-26 18:41:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2021-03-26 16:29:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-26 15:42:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2021-03-26 18:41:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-03-26 15:42:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-26 15:41:09.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-26 15:42:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-26 16:25:51.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-03-26 16:25:59.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2021-03-26 18:41:32.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"e4b14e8f0a2c0a8cd7953b0858760dc2bdc6cd5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-26 15:41:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0508665885a4c58743be6275d27b727020157a42","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-26 15:42:33.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"dc675a80f138eb8f8b5fcb5d1271c4392a7c2299","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 15:42:33.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"0bb3449bec324a0f86014a5448f41d4d20c38739","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 15:50:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b94e813ec2d74bbead11e187c30ee432 : SUCCESS in 6m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7d99eaa3af614d459ce22160484f3855 : SUCCESS in 5m 48s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7991b94b36e749de87b46727f34cb8c8 : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":1},{"id":"ef6240f71be83d1fa74d831ac3f5a41a645995fa","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-26 16:29:15.000000000","message":"Patch Set 1: Code-Review+2\n\nAbandoning https://review.opendev.org/c/openstack/releases/+/781729","accounts_in_message":[],"_revision_number":1},{"id":"3467a0df7c0c542c89773e0ac86aab8a00f9bb82","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-26 18:41:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM:\n* version bump is OK\n* uploaded by release liaison\n* test jobs logs don\u0027t show any issue","accounts_in_message":[],"_revision_number":1},{"id":"4fc18597f14775eca0ebe096fa7198fbe5c055bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 18:48:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4aa7070f8b07479cdda030e431bca3d7309fe5ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 19:02:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c6df272417347aba6ce377d00fc4288 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/691b4be94ce54965ab0da0581d7b49ca : SUCCESS in 6m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9cd1737d97734defad88aa85ea093b88 : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":1},{"id":"21264d7da597310b0f9cdc892e5c16c4f3a6500d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 19:02:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f1a0c7e4facdafc1f7d545df245bf4716f7d8ae2","revisions":{"f1a0c7e4facdafc1f7d545df245bf4716f7d8ae2":{"kind":"REWORK","_number":1,"created":"2021-03-26 15:41:09.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/38/783338/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/38/783338/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/783338/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/783338/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/783338/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/38/783338/1"}}},"commit":{"parents":[{"commit":"753fc318dec0b4afaca6a4408e3d5ec7aaa734dc","subject":"Merge \"[OpenstackSDK] Create for all client and non-client lib wallaby branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/753fc318dec0b4afaca6a4408e3d5ec7aaa734dc"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-19 14:20:52.000000000","tz":60},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-03-26 15:41:08.000000000","tz":0},"subject":"Proposing Wallaby RC1 for octavia","message":"Proposing Wallaby RC1 for octavia\n\nIf you wants to wait for a specific patch to make it to the release,\nsomeone from your team can -1 the patch to have it held, or update that patch\nwith a different commit SHA.\n\nPatches with no response will be merged by Friday March 26.\n\n$ git log --oneline --no-merges 7.0.0..3cdac6af\n00e3f834 Fix empty Batch Member Update to unlock objects\nd0ec3aaf Fix using subnets with host_routes in amphorav2 driver\n0513319f Fix LB failover for amphorav2: set security group\n6c54eab5 Make /healthcheck cache results\nfc8ee42d Fix provider driver quota handling\nc5e59b5e [goal] Deprecate the JSON formatted policy file\ne99d7b93 Add aarch64/amd64 support to amphora and devstack\n91ee39a9 Remove some miscellaneous sanity check code\na5f142c5 Add support for scoped tokens and default roles\ndd3edb63 Add HTTP/2 to default lists of ALPN protocols\nd17b2328 Fix grenade resource waiters\ne3ab49c6 Fix LB failover when IP addresses exhausted\n08e1d6d0 Add retry for getting amphora VM\nb95fbe9e Fix overriding default listener timeout values in config file\nc93a76b9 Add HTTP/2 example to the load balancing cookbook\nc37738ba Fix pep8 errors\nce2e0d74 Add test coverage for SCTP health checker script\n3f6fb8af Fix health manager messages about failed failovers\n370aa4e6 Fix pools going into ERROR when updating the pool\nf45092a8 Configure rsyslog on Octavia service nodes in devstack\na822f30e Bump oslo.policy version to 2.1.0\n7d8b0db3 Bump oslo.log version to 4.3.0\na518cefd Disable auto configuration of new interfaces in NetworkManager\ne5f9f670 Fix incorrect ERROR status with IPv6 UDP members\n6a7d8b2c Small fix in loadbalancer POST api-ref\n9b2c6425 Bump the minimum coverage to 92%\n2888f44e Add SCTP support in Amphora\n3730d272 Install HAProxy 2.x in CentOS amphora images\n53f8c28a Add default value for enabled column in l7rule table\nd2d5fc80 Add ALPN support for TLS-enabled pools\ndda1d866 Add validation for VIP network parameters in amphora driver\nfbbc5f90 Validate listener protocol in amphora driver\na4aa03d3 Fix the unit test tree structure\n57f5666f Correct spell error\n1a86e454 Use \u0027bash\u0027 in the keepalived check script\nd81a0556 Fix periodic image builder jobs\ndc876eed Fix gate failure\n1a154839 Add amphora_id in store params for failover_amphora\n2954370e Fix operating status for empty UDP pools\n9653920a Remove useless proto parameter from sample functions\n6aa23358 Fix missing cronie package in RHEL-based image builds\n15a68fd4 Remove re-import of octavia-lib constants\n6d59fec8 Map cloud-guest-utils to cloud-utils-growpart for Red Hat distros.\n8a207a7e Fix setting None in tls_versions and tls_ciphers in pools\n7ad02237 Support deletion of all tags\n57e7bd17 Fix pep8 failures in gate\n17ae2625 Handle listener not found during loadbalancer status update\nf96251c7 Fix load balancers with failed amphora failover\n9a6a9000 Add experimental amphorav2 jobs\nfd4abac1 Replace assertItemsEqual with assertCountEqual\na9a817e4 Use run_flow for failover functions in v2 control worker\n71440bdb Imported Translations from Zanata\n332791ee Fix pool delete race on load balancer cascade delete\nce6b563e Stop install unnecessary packages when run tox -ereleasenotes\n02dc9881 Fix amphorav2 bytes error\n40245366 Fix PROXYV2 pools\n77dc23fc Fix load balancers with failed amphora failover\ndcdf6315 Remove Neutron SDN-specific code\n418ea090 Add Python3 wallaby unit tests\n3d0486a7 Update master for stable/victoria\nb627d7cf Remove unnecessary joinedload\n64a301d4 Fix nf_conntrack_buckets sysctl in Amphora\n03e6c73d Correct a typo in the document\n\nChange-Id: I666d621cf77a4a8c4d4b68cded14f919404d1702\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f1a0c7e4facdafc1f7d545df245bf4716f7d8ae2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f1a0c7e4facdafc1f7d545df245bf4716f7d8ae2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
