)]}'
{"id":"openstack%2Freleases~783905","triplet_id":"openstack%2Freleases~master~Ie30316c000041470977a7f1485f7dff6e8d5ba85","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"Ie30316c000041470977a7f1485f7dff6e8d5ba85","subject":"Release ansible-role-thales-hsm for Wallaby","status":"MERGED","created":"2021-03-30 16:26:56.000000000","updated":"2021-03-31 12:07:47.000000000","submitted":"2021-03-31 12:07:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"783905","meta_rev_id":"3cff6ba553318b0488a42d8380a9bf9bc42f646b","_number":783905,"virtual_id_number":783905,"owner":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-31 12:07:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2021-03-31 11:42:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":2,"date":"2021-03-31 11:30:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2021-03-31 11:42:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-30 16:28:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-30 16:32:10.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-31 11:42:57.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"576cade54e8d9904bfec7bf195fe450dfbdc3937","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-03-30 16:26:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"788c092630d21523020bbc9b8f37b605f216d8e1","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 16:28:08.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"a234d72a3aad644d88f7aa34f3c8cdb4ada9a863","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 16:47:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6373290c98a451cbae318fe19acac6e : SUCCESS in 6m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/559d6e4399fb4d8f84bd5a4732e7dbb3 : FAILURE in 6m 15s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/1eee1018b6e348ada9d65a990565e3aa : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":1},{"id":"bf1d58b03c0a115864ab43039490f6699ad92142","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-30 17:00:12.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"65e7ec3c8500732075beff6eb1c89dd23ed82c67","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-30 17:02:14.000000000","message":"Patch Set 1:\n\nI\u0027m not sure why the get this error:\n\n```\ndeliverables/wallaby/ansible-role-thales-hsm.yaml: validate_release_type: openstack-infra/project-config/zuul.d/projects.yaml no release job specified for openstack/ansible-role-thales-hsm, one of [\u0027publish-to-pypi\u0027] needs to be included in [] or no release will be published\n```","accounts_in_message":[],"_revision_number":1},{"id":"1863b9f482327bc10fecc83b9dc94d729e64926c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-30 17:03:30.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e I\u0027m not sure why the get this error:\n\u003e \n\u003e ```\n\u003e deliverables/wallaby/ansible-role-thales-hsm.yaml: validate_release_type: openstack-infra/project-config/zuul.d/projects.yaml no release job specified for openstack/ansible-role-thales-hsm, one of [\u0027publish-to-pypi\u0027] needs to be included in [] or no release will be published\n\u003e ```\n\nansible-role-lunasa-hsm doesn\u0027t have it and CI doesn\u0027t complain\n\nhttps://review.opendev.org/c/openstack/releases/+/783663","accounts_in_message":[],"_revision_number":1},{"id":"5459e41c6bb3930fdad58d1922e4ef98bd97c015","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-03-30 17:14:58.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e I\u0027m not sure why the get this error:\n\u003e \u003e \n\u003e \u003e ```\n\u003e \u003e deliverables/wallaby/ansible-role-thales-hsm.yaml: validate_release_type: openstack-infra/project-config/zuul.d/projects.yaml no release job specified for openstack/ansible-role-thales-hsm, one of [\u0027publish-to-pypi\u0027] needs to be included in [] or no release will be published\n\u003e \u003e ```\n\u003e \n\u003e ansible-role-lunasa-hsm doesn\u0027t have it and CI doesn\u0027t complain\n\u003e \n\u003e https://review.opendev.org/c/openstack/releases/+/783663\n\nI wonder if the difference is that ansible-role-lunasa-hsm has been previously released to PyPI https://pypi.org/project/ansible-role-lunasa-hsm/\n\nI don\u0027t this repo or ansible-role-atos-hsm have been released on PyPI yet.","accounts_in_message":[],"_revision_number":1},{"id":"a1ae2feea63a4ea659f2244b4c180d0862505f0e","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-03-30 17:18:59.000000000","message":"Patch Set 1: Workflow-1\n\nAs I mentioned in the other ansible-role patch, we moved these two repos to the Barbican project so we could make versioned releases, but they both have been maintained for a while: https://opendev.org/openstack/governance/commit/e7a8fe847847c82c21e9a9eb8d80413c6ecc8b4a\n\nThis one also had a couple of releases before being moved to OpenStack.  See: https://github.com/dmend/ansible-role-thales-hsm/tags\n\nI think 1.0.0 is appropriate for this release as well.","accounts_in_message":[],"_revision_number":1},{"id":"d45dfa6a47f9a91e1e0093830a2bcda874e490e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-03-30 17:21:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"641794158ecfe3399b7451ad0b722c429d45bc9f","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 17:22:11.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"37f073f70ba5ca9bfe2af1b3a50f16d367b67aa3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-30 18:10:09.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 1: Workflow-1\n\u003e \n\u003e As I mentioned in the other ansible-role patch, we moved these two repos to the Barbican project so we could make versioned releases, but they both have been maintained for a while: https://opendev.org/openstack/governance/commit/e7a8fe847847c82c21e9a9eb8d80413c6ecc8b4a\n\u003e \n\u003e This one also had a couple of releases before being moved to OpenStack.  See: https://github.com/dmend/ansible-role-thales-hsm/tags\n\u003e \n\u003e I think 1.0.0 is appropriate for this release as well.\n\nok WFM","accounts_in_message":[],"_revision_number":2},{"id":"b2aae64531c43a75c7cf72b442e32de350eff034","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 18:33:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b3958a92034478da7883669365b00e4 : SUCCESS in 7m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/962aa53fd8754a14ba2c0ece2d4b7f2e : FAILURE in 7m 09s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/91f2091f3ad544879fb58f126f918928 : SUCCESS in 7m 17s","accounts_in_message":[],"_revision_number":2},{"id":"b116d4639ad546b3f609cc0864918d674ddc471a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-30 18:44:53.000000000","message":"Patch Set 2: Code-Review-1\n\nFirst you need to add [1] zuul template (publish-to-pypi [2]) to this role.\n\nWithout that we can\u0027t continue.\n\n[1] https://opendev.org/openstack/project-config/src/branch/master/zuul.d/projects.yaml#L1201\n[2] https://opendev.org/openstack/project-config/src/branch/master/zuul.d/projects.yaml#L1140","accounts_in_message":[],"_revision_number":2},{"id":"27db935dcd8b35af74ef1d2d08ddca84b8ecfd75","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-30 20:10:57.000000000","message":"Patch Set 2:\n\nMind you it will be an issue for you if we start to release these project with Xena?\n\nWe are really close from the requirements branching and many pieces are still missing here, I\u0027m afraid this is a problem.\n\nHopefully we can continue if all the missing elements are in place tomorrow, else I think that we should skip it for Wallaby.","accounts_in_message":[],"_revision_number":2},{"id":"f02a88b5e9b42da3f30cf202c77a4d843bd8c62d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-31 06:55:21.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"b5b74591a318f33823dc25848acfb4ef45ab5344","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 07:10:55.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2a5b2919e904f2da58e6d76aedf49a1 : SUCCESS in 7m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b9415e085c094f72bb423ceaf7041301 : FAILURE in 6m 31s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/18dfbee5967f42dbabf4dc64f54680db : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":3},{"id":"6533d34686a21ab94bcf3be5e27f5d2d1ac4f53c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-31 07:15:05.000000000","message":"Patch Set 3:\n\nI think that the gates will be fixed when the zuul template part will be merged\n\nhttps://review.opendev.org/c/openstack/project-config/+/784011","accounts_in_message":[],"_revision_number":3},{"id":"8ef04259cb43d264888f3bbdaa149ae887d9605c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-31 08:28:34.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ce4e394202299f073b2bc81b7e871019aacdb0bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 08:51:02.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b34817317a6d49e688610726867c66fd : SUCCESS in 7m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7088e98e67ba45de96ad660d89d3f5a1 : FAILURE in 6m 25s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/60ff0d421f1b4bb89ad725bfbac67c38 : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":3},{"id":"26b239e4e556740d3c9bb64fb76c469dac275d7a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-31 08:57:55.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"43817c09b60927b73e08762ddbde0dc4926131c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 09:12:06.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/304d17d362fc4fc68d507e321a0433ed : SUCCESS in 7m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3c94d9fe7dbc41fb82030a7e5adbf88a : FAILURE in 5m 21s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/77686a7c3481408e8b4ca85dd76518b9 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":3},{"id":"fb127b3a5998e004c508a2e9a24ee9f1a8fcca8a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-31 09:37:17.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ba9054468e02b0a0daeb2098af120aa090c8d762","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 09:51:00.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55aafe2a4bf146c58db0d70584982757 : SUCCESS in 7m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c2eb99761a7f4abd93ed7c69a4070f79 : FAILURE in 6m 39s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/bb398951bcae4a199756d279a001ac25 : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":3},{"id":"d7d0ad5966fe191ba95b1095b765cbd7831f58a1","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-31 11:15:43.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"7023f9f4bdca2f0974561bb8d72c557269b84aaf","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-31 11:17:41.000000000","message":"Patch Set 3:\n\nHopefully recent gates issues will be fixed by https://review.opendev.org/c/openstack/releases/+/784068","accounts_in_message":[],"_revision_number":3},{"id":"a622d313b744bc41d89cd7d90344511f036a17cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 11:29:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3d1359f6e0d46ce850f4064cfdd0408 : SUCCESS in 8m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ddad36b4107f4fb2bed9080b5d544c90 : SUCCESS in 5m 35s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/45dcd08fa7f84ddbb6dea8b15a492925 : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":3},{"id":"e87295b962fa47f5cd16488e60e3d24ecd5426b6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-31 11:30:20.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"51260ed0884b221c441cfb73418cb5c766c011e1","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-03-31 11:42:57.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"de7cfd184e3d442d0c58694e96f436e90b4e103c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 11:43:32.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fb396366a36460de9b6f8660e3ac15f6827a6c88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 12:07:36.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2239797aff4c4ef7b22561dbc7c09bd7 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46280f1e5b0e476a92408f9f6cc98ead : SUCCESS in 7m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b4a31335fdc94fc5ac6d81cf690ce217 : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":3},{"id":"3cff6ba553318b0488a42d8380a9bf9bc42f646b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 12:07:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"63d476f70a524c0f30e77de1a0d7cc21ec606810","revisions":{"93384fc6b76dd72d96e512e52a015c94327ed644":{"kind":"REWORK","_number":1,"created":"2021-03-30 16:26:56.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/05/783905/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/05/783905/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/783905/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/783905/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/783905/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/05/783905/1"}}},"commit":{"parents":[{"commit":"1bc4890fef79535ac307335d31cfeb5f90ea5cc1","subject":"Merge \"Ignore trailing projects from RC1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1bc4890fef79535ac307335d31cfeb5f90ea5cc1"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-03-30 16:26:31.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-03-30 16:26:31.000000000","tz":-300},"subject":"Release ansible-role-thales-hsm for Wallaby","message":"Release ansible-role-thales-hsm for Wallaby\n\nChange-Id: Ie30316c000041470977a7f1485f7dff6e8d5ba85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/93384fc6b76dd72d96e512e52a015c94327ed644"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/93384fc6b76dd72d96e512e52a015c94327ed644"}]},"branch":"refs/heads/master"},"0ba75abf1a4fd50de6b73bd1aa9e9c8d07cf874f":{"kind":"REWORK","_number":2,"created":"2021-03-30 17:21:04.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/05/783905/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/05/783905/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/783905/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/783905/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/783905/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/05/783905/2"}}},"commit":{"parents":[{"commit":"1bc4890fef79535ac307335d31cfeb5f90ea5cc1","subject":"Merge \"Ignore trailing projects from RC1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1bc4890fef79535ac307335d31cfeb5f90ea5cc1"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-03-30 16:26:31.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-03-30 17:20:58.000000000","tz":-300},"subject":"Release ansible-role-thales-hsm for Wallaby","message":"Release ansible-role-thales-hsm for Wallaby\n\nChange-Id: Ie30316c000041470977a7f1485f7dff6e8d5ba85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0ba75abf1a4fd50de6b73bd1aa9e9c8d07cf874f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0ba75abf1a4fd50de6b73bd1aa9e9c8d07cf874f"}]},"branch":"refs/heads/master"},"63d476f70a524c0f30e77de1a0d7cc21ec606810":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-03-31 06:55:21.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/05/783905/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/05/783905/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/783905/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/783905/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/783905/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/05/783905/3"}}},"commit":{"parents":[{"commit":"1bc4890fef79535ac307335d31cfeb5f90ea5cc1","subject":"Merge \"Ignore trailing projects from RC1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1bc4890fef79535ac307335d31cfeb5f90ea5cc1"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-03-30 16:26:31.000000000","tz":-300},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-31 06:54:54.000000000","tz":120},"subject":"Release ansible-role-thales-hsm for Wallaby","message":"Release ansible-role-thales-hsm for Wallaby\n\nDepends-On: https://review.opendev.org/c/openstack/project-config/+/784011\nChange-Id: Ie30316c000041470977a7f1485f7dff6e8d5ba85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/63d476f70a524c0f30e77de1a0d7cc21ec606810"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/63d476f70a524c0f30e77de1a0d7cc21ec606810"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
