)]}'
{"id":"openstack%2Freleases~783908","triplet_id":"openstack%2Freleases~master~I2992bfe6ccfcc096f7024eddd904f2235da18aac","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I2992bfe6ccfcc096f7024eddd904f2235da18aac","subject":"Release ansible-role-atos-hsm for Wallaby","status":"MERGED","created":"2021-03-30 16:43:50.000000000","updated":"2021-03-31 12:07:31.000000000","submitted":"2021-03-31 12:07:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"783908","meta_rev_id":"9a4444f0f2b0b3ee645752f049e8510ae66484ae","_number":783908,"virtual_id_number":783908,"owner":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-31 12:07:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2021-03-31 11:42:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2021-03-31 11:44:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2021-03-31 11:30:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2021-03-31 11:42:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2021-03-31 11:44:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-30 16:57:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-30 16:59:12.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-31 11:42:47.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2021-03-31 11:44:53.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"ec5bd9e97bcbc08c7e1e55a235d1e5aff3ffaa1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-03-30 16:43:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"25beb94f831f9fe1515a63b69b364d7eac8fa57f","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 16:57:27.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"b08c92bdeee82227bb088888f20695637a49482c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-30 16:59:12.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eb76211b34933258c818363b9a334f0b7b79865e","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-03-30 17:05:32.000000000","message":"Patch Set 1:\n\nThanks for the review, Hervé.   For context, this repo was previously maintained as part of the OpenStackAnsible deliverables.  We moved it  into Barbican so that we could make versioned releases.  See: https://opendev.org/openstack/governance/commit/e7a8fe847847c82c21e9a9eb8d80413c6ecc8b4a\n\nThis repo did get a few releases before being moved to OpenStack.  See: https://github.com/dmend/ansible-role-atos-hsm/tags\n\nI think it would be best to release as 1.0.0.","accounts_in_message":[],"_revision_number":1},{"id":"95b831d2bc54f14ed852a9b5569aed9ff39eeaaa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-03-30 17:07:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a19dd1e748429cc9f5d647994b6ba3485eff20ad","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 17:09:03.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"8a371d43aee53effd1572ce815c931455833ff3c","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-03-30 17:11:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9142360c2e8a6423226c3a3baa05f1a0b547281c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-30 18:11:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2170c62fe2ce16e4a71a963f163f35c855ff7a9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 18:16:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2936ca467d44fb4b7f2f9bf8b628793 : SUCCESS in 6m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7c004053375f4d9e942bd8dc16967b66 : FAILURE in 5m 17s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6f39e11a44c6482b96e799b493a24fdb : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":2},{"id":"e846636ff43f26e78c12f661c0a55124c866e70b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-30 18:43:58.000000000","message":"Patch Set 2: Code-Review-1\n\nFirst you need to add [1] zuul template (publish-to-pypi [2]) to this role.\n\nWithout that we can\u0027t continue.\n\n[1] https://opendev.org/openstack/project-config/src/branch/master/zuul.d/projects.yaml#L1112\n[2] https://opendev.org/openstack/project-config/src/branch/master/zuul.d/projects.yaml#L1140","accounts_in_message":[],"_revision_number":2},{"id":"57a802a2fc89234a76dd53281e9bb869a3cd6057","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-30 20:11:06.000000000","message":"Patch Set 2:\n\nMind you it will be an issue for you if we start to release these project with Xena?\n\nWe are really close from the requirements branching and many pieces are still missing here, I\u0027m afraid this is a problem.\n\nHopefully we can continue if all the missing elements are in place tomorrow, else I think that we should skip it for Wallaby.","accounts_in_message":[],"_revision_number":2},{"id":"a0b636e806864bf2712046487c3f54759dd71e1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-31 06:54:09.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"be70138f801fc30c3acd5078b71b19373b2e965d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 07:42:23.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/667fe507265f40d4b916030a86da560e : SUCCESS in 7m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ea467be627394b0ebb620d054f66a868 : FAILURE in 5m 38s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9cd9078cc35e4061932ceaf16a328016 : SUCCESS in 6m 44s","accounts_in_message":[],"_revision_number":3},{"id":"627e24999b1cb7dcc016880deb5cbfef99a86cc9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-31 08:28:38.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"3002db6f36f5ceec88d330227448f6de7b3632ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 09:04:32.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ed3b3798aa84aab8b6c28b72b612c6c : SUCCESS in 7m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b5c0374818274e53b58883094019f234 : FAILURE in 5m 40s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e1f974169af1451fbb39c1b0725c3d53 : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":3},{"id":"28e92b411d5da762468c9ebb9075806f9c7b21fb","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-31 09:37:10.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"cd6db5e026493b80cf85d51f0177f459c60e7df0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 09:51:24.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78019c627c3440d4aba8ef20783dc88e : SUCCESS in 7m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f72a8875b48d4eef9b9f6f94ff5636c4 : FAILURE in 5m 30s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/cf95964a2c7d437887adc4ad07560837 : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":3},{"id":"942c567f9e53793a72e4834ef147cfc85c507216","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-31 11:15:37.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ec0e82bc3ba11cba20f269f0e8befcdc539a350b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-31 11:17:46.000000000","message":"Patch Set 3:\n\nHopefully recent gates issues will be fixed by https://review.opendev.org/c/openstack/releases/+/784068","accounts_in_message":[],"_revision_number":3},{"id":"f75a24fcaa30581614cdbf720c37a0754afd8788","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 11:29:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a791a029c5543f98ac48181e8df0315 : SUCCESS in 8m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7d3e5e15f9dc4bdc8f793b94773d9b9f : SUCCESS in 6m 12s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c737cfe0fc2a49fe9426f3d455d76b25 : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":3},{"id":"343a04c6bee57d77c3869a1c615333d9e026daba","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-31 11:30:25.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e9c9520c480728878ff384c5b1bf85d664d418e7","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-03-31 11:42:47.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"695b498538b11d859e81da40b99d9d7345dfc39a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 11:43:20.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7a6219fe50849f6e07f18f685aa5629b51b94920","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-31 11:44:53.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nLGTM. I could have imagine only a minor version bump but according to the PTL\u0027s summary major version bump is OK. Validator issue was resolved, list-changes-results.log doesn\u0027t show any issue.","accounts_in_message":[],"_revision_number":3},{"id":"57fb5318f3c68c2ce65ac8d551e147ae507e2dd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 12:07:29.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d74ef153a5e847b3b49f2a58a961c443 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76336c6ee15b4a1aaa88d3863372e91c : SUCCESS in 7m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/eaff2f4aa3134dc9ac072b3433aa20a7 : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":3},{"id":"9a4444f0f2b0b3ee645752f049e8510ae66484ae","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 12:07:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ce7d9500ede22d873cb57c031f94fd3ae818f17d","revisions":{"a0befc2e5cf583d35b7e3b70e0d96cc2e3b6132c":{"kind":"REWORK","_number":1,"created":"2021-03-30 16:43:50.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/08/783908/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/08/783908/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/783908/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/783908/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/783908/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/08/783908/1"}}},"commit":{"parents":[{"commit":"1bc4890fef79535ac307335d31cfeb5f90ea5cc1","subject":"Merge \"Ignore trailing projects from RC1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1bc4890fef79535ac307335d31cfeb5f90ea5cc1"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-03-30 16:43:34.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-03-30 16:43:34.000000000","tz":-300},"subject":"Release ansible-role-atos-hsm for Wallaby","message":"Release ansible-role-atos-hsm for Wallaby\n\nChange-Id: I2992bfe6ccfcc096f7024eddd904f2235da18aac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a0befc2e5cf583d35b7e3b70e0d96cc2e3b6132c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a0befc2e5cf583d35b7e3b70e0d96cc2e3b6132c"}]},"branch":"refs/heads/master"},"d4703fc8fa387c30d64595fdc9f24671af090fc0":{"kind":"REWORK","_number":2,"created":"2021-03-30 17:07:34.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/08/783908/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/08/783908/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/783908/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/783908/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/783908/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/08/783908/2"}}},"commit":{"parents":[{"commit":"1bc4890fef79535ac307335d31cfeb5f90ea5cc1","subject":"Merge \"Ignore trailing projects from RC1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1bc4890fef79535ac307335d31cfeb5f90ea5cc1"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-03-30 16:43:34.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-03-30 17:07:27.000000000","tz":-300},"subject":"Release ansible-role-atos-hsm for Wallaby","message":"Release ansible-role-atos-hsm for Wallaby\n\nChange-Id: I2992bfe6ccfcc096f7024eddd904f2235da18aac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d4703fc8fa387c30d64595fdc9f24671af090fc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d4703fc8fa387c30d64595fdc9f24671af090fc0"}]},"branch":"refs/heads/master"},"ce7d9500ede22d873cb57c031f94fd3ae818f17d":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-03-31 06:54:09.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/08/783908/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/08/783908/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/783908/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/783908/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/783908/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/08/783908/3"}}},"commit":{"parents":[{"commit":"1bc4890fef79535ac307335d31cfeb5f90ea5cc1","subject":"Merge \"Ignore trailing projects from RC1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1bc4890fef79535ac307335d31cfeb5f90ea5cc1"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-03-30 16:43:34.000000000","tz":-300},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-31 06:53:34.000000000","tz":120},"subject":"Release ansible-role-atos-hsm for Wallaby","message":"Release ansible-role-atos-hsm for Wallaby\n\nDepends-On: https://review.opendev.org/c/openstack/project-config/+/784011\nChange-Id: I2992bfe6ccfcc096f7024eddd904f2235da18aac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ce7d9500ede22d873cb57c031f94fd3ae818f17d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ce7d9500ede22d873cb57c031f94fd3ae818f17d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
