)]}'
{"id":"openstack%2Freleases~784371","triplet_id":"openstack%2Freleases~master~I1a119c751d637260b0ca187c2ea65fa221554759","project":"openstack/releases","branch":"master","topic":"devstack-plugin-victoria","hashtags":[],"change_id":"I1a119c751d637260b0ca187c2ea65fa221554759","subject":"Branch devstack-plugin-* deliverables for Victoria","status":"ABANDONED","created":"2021-04-01 11:11:55.000000000","updated":"2021-04-02 05:35:41.000000000","total_comment_count":3,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"13d42d77bc310041725105179d43ce953fcf4afc","_number":784371,"virtual_id_number":784371,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-04-01 11:59:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"}],"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-01 11:13:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-01 11:14:49.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2021-04-01 11:20:11.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2021-04-01 11:38:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-01 17:11:34.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"CC"},{"updated":"2021-04-01 17:36:11.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"2d4c4f15e7489a19c6b390d0b8fb6d26015b932d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-04-01 11:11:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a180c846b28c195b96a07bd3d7280d1c5c733d41","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-04-01 11:14:49.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"08aedf4429922288bfefca6991bea8c91cecdcef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 11:38:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39056fa39b9a4bc6b6d0cb409708d835 : SUCCESS in 6m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dfeefa77edcb48ea81a6c66ef0873aa9 : FAILURE in 7m 18s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/71e81a043e7e4f9684c15ef7b6faec7d : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":1},{"id":"fe6ec57ebaa2e73cd3343a6583d81ee0f32e7607","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-04-01 11:46:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e4574d30ff38aa00f61ab7c4400ce50cfdc11760","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 11:59:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e5d892053504fcc9b463c60e5430882 : SUCCESS in 6m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2c3e16649b6a43cbae71e3043c0b6fbb : SUCCESS in 5m 03s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/5ea06928cb2447d5a5aa8ba4d002bfb2 : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":2},{"id":"77f66611e519a8abf1d7480ce3f6fedfa337c256","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2021-04-01 17:11:34.000000000","message":"Patch Set 2:\n\nWe intentionally stopped branching these because they had little to no activity over multiple releases, so branching was just extra overhead. I prefer https://review.opendev.org/c/openstack/releases/+/784376 since there\u0027s no reason for the release team to be involved in these repos anymore.\n\nNote: I\u0027m referring only to the amqp1 and kafka plugins. I have no idea about the container plugin as that\u0027s not one of Oslo\u0027s.","accounts_in_message":[],"_revision_number":2},{"id":"a9a9ffe4f91b7aa7edbb3b3070603ce99ebb4631","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-04-02 05:35:27.000000000","message":"Patch Set 2:\n\nThanks for your response Ben.\n\nI\u0027ll abandon this patch and I\u0027ll delete the victoria and the wallaby part in the same patch, but first I think that we need a round of deprecation on the governance side to avoid inconsistencies\nwithin our tooling.","accounts_in_message":[],"_revision_number":2},{"id":"13d42d77bc310041725105179d43ce953fcf4afc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-04-02 05:35:41.000000000","message":"Abandoned\n\nCf. my previous message","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"268ebce0939033e7b01fe1c0503ee68bd60cda39","revisions":{"97f4f9ec2a2e0acf23cbad0f057d44da1c6854cd":{"kind":"REWORK","_number":1,"created":"2021-04-01 11:11:55.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/71/784371/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/71/784371/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/784371/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/784371/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/784371/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/71/784371/1"}}},"commit":{"parents":[{"commit":"decab00bb4a48c0cb691353ab0b4860bd10cdbe4","subject":"Merge \"Make a stable/victoria release for all the tripleo things\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/decab00bb4a48c0cb691353ab0b4860bd10cdbe4"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-04-01 11:11:32.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-04-01 11:11:32.000000000","tz":120},"subject":"Branch devstack-plugin-* deliverables","message":"Branch devstack-plugin-* deliverables\n\nChange-Id: I1a119c751d637260b0ca187c2ea65fa221554759\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/97f4f9ec2a2e0acf23cbad0f057d44da1c6854cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/97f4f9ec2a2e0acf23cbad0f057d44da1c6854cd"}]},"branch":"refs/heads/master"},"268ebce0939033e7b01fe1c0503ee68bd60cda39":{"kind":"REWORK","_number":2,"created":"2021-04-01 11:46:51.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/71/784371/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/71/784371/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/784371/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/784371/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/784371/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/71/784371/2"}}},"commit":{"parents":[{"commit":"decab00bb4a48c0cb691353ab0b4860bd10cdbe4","subject":"Merge \"Make a stable/victoria release for all the tripleo things\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/decab00bb4a48c0cb691353ab0b4860bd10cdbe4"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-04-01 11:11:32.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-04-01 11:46:29.000000000","tz":120},"subject":"Branch devstack-plugin-* deliverables for Victoria","message":"Branch devstack-plugin-* deliverables for Victoria\n\nChange-Id: I1a119c751d637260b0ca187c2ea65fa221554759\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/268ebce0939033e7b01fe1c0503ee68bd60cda39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/268ebce0939033e7b01fe1c0503ee68bd60cda39"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
