)]}'
{"id":"openstack%2Freleases~785166","triplet_id":"openstack%2Freleases~master~Ib59fd198fe74cabcc13bde7464241855a1fbb997","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"Ib59fd198fe74cabcc13bde7464241855a1fbb997","subject":"Cinder team releases for stable/train","status":"MERGED","created":"2021-04-07 12:27:03.000000000","updated":"2021-04-08 10:40:30.000000000","submitted":"2021-04-08 10:40:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"785166","meta_rev_id":"a2f61d0abf5a01736f2b85b7f3ad06041a2f1083","_number":785166,"virtual_id_number":785166,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-08 10:40:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-04-08 10:33:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-04-07 12:56:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2021-04-08 08:35:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2021-04-08 10:33:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-04-07 12:57:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-07 12:32:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-07 12:42:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-07 12:49:07.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2021-04-07 12:56:32.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-04-07 13:18:54.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2021-04-07 13:20:07.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-04-08 08:35:10.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"86ae809191df1e9c8ef61f1a458f9c52d8e75e32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-04-07 12:27:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e83bc80851d52d19cab3a8f0810991d59dd17bed","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-04-07 12:32:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9b56166cf5d11f6b854df1dae611acdc081df69d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 12:42:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcc430efc0114985978fcdcea12c594e : SUCCESS in 6m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/83a1298dd5d04d9b8a3a1effd68f7ec8 : FAILURE in 5m 08s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ea9ee221784647efb4533e1fa03a52f4 : FAILURE in 4m 59s","accounts_in_message":[],"_revision_number":1},{"id":"13b66e4281df37fd32aaf0d4bfc9cf9216a35003","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-04-07 12:47:19.000000000","message":"Patch Set 1:\n\nrecheck\n\ndon\u0027t think this is a related failure\n\n2021-04-07 12:40:46.547830 | ubuntu-focal |   DEBUG: Branch \u0027stable/train\u0027 set up to track remote branch \u0027stable/train\u0027 from \u0027origin\u0027.\n2021-04-07 12:40:46.547848 | ubuntu-focal |   DEBUG: + git pull --ff-only\n2021-04-07 12:40:46.547866 | ubuntu-focal |   DEBUG: fatal: unable to access \u0027https://opendev.org/openstack/cinder/\u0027: server certificate verification failed. CAfile: none CRLfile: none\n2021-04-07 12:40:46.547884 | ubuntu-focal | Traceback (most recent call last):\n2021-04-07 12:40:46.547902 | ubuntu-focal |   File \"/home/zuul/src/opendev.org/openstack/releases/.tox/list-changes/bin/list-changes\", line 10, in \u003cmodule\u003e\n2021-04-07 12:40:46.547959 | ubuntu-focal |     sys.exit(main())\n2021-04-07 12:40:46.547981 | ubuntu-focal |   File \"/home/zuul/src/opendev.org/openstack/releases/openstack_releases/cmds/list_changes.py\", line 386, in main\n2021-04-07 12:40:46.548001 | ubuntu-focal |     gitutils.clone_repo(\n2021-04-07 12:40:46.548019 | ubuntu-focal |   File \"/home/zuul/src/opendev.org/openstack/releases/openstack_releases/gitutils.py\", line 129, in clone_repo\n2021-04-07 12:40:46.548040 | ubuntu-focal |     processutils.check_call(cmd)\n2021-04-07 12:40:46.548058 | ubuntu-focal |   File \"/home/zuul/src/opendev.org/openstack/releases/openstack_releases/processutils.py\", line 57, in check_call\n2021-04-07 12:40:46.548077 | ubuntu-focal |     raise subprocess.CalledProcessError(completed.returncode, cmd)\n2021-04-07 12:40:46.548096 | ubuntu-focal | subprocess.CalledProcessError: Command \u0027[\u0027./tools/clone_repo.sh\u0027, \u0027--workspace\u0027, \u0027/home/zuul/src/opendev.org/openstack/releases/.tox/list-changes/tmp/releases-ooo9njxx\u0027, \u0027--branch\u0027, \u0027stable/train\u0027, \u0027openstack/cinder\u0027]\u0027 returned non-zero exit status 1.\n2021-04-07 12:40:47.450637 | ubuntu-focal | ERROR: InvocationError for command /home/zuul/src/opendev.org/openstack/releases/tools/tox-log-command.sh list-changes (exited with code 1)","accounts_in_message":[],"_revision_number":1},{"id":"9cf60f8a5e76699d1b3368d85df72b2dd7f6bfaf","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-04-07 12:49:07.000000000","message":"Patch Set 1:\n\nrecheck some ssh and cert problems","accounts_in_message":[],"_revision_number":1},{"id":"9439b6dd7d98e5922f46f3f8ff10034e7e7cfbc6","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-04-07 12:56:32.000000000","message":"Patch Set 1: Code-Review+1\n\nVerified the hashes.","accounts_in_message":[],"_revision_number":1},{"id":"158dd96a46f8f8ae39ca098af206aa9bb214c8d4","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 12:57:44.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"44da3a76c80bd7e5fb2776bc5609b9d7c519baea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 13:00:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf42b20e3c884b929147a6dd17bd6ba9 : FAILURE in 4m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/30c7c64025d14050af8c939b7c7e03ab : SUCCESS in 6m 06s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/634a4c14d7994ce9b5fe82a6c572a3c6 : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":1},{"id":"9cb3e4bbe4dee18c0075daf5ee5844a736c71a58","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-04-07 13:18:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4821f89df4b1ed847eb29e8f9535b2d31e18e804","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 13:36:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ae6df7c76b546118be975268afdfb89 : SUCCESS in 7m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d12c639f9cea459da71481ceaf6abd9f : SUCCESS in 6m 45s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c3c4f805ada6474986b1ed520728f365 : SUCCESS in 7m 06s","accounts_in_message":[],"_revision_number":1},{"id":"144366b88ec9572e6042e249f69fbafa46899486","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-04-08 08:35:10.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d506bce89ebf475d427ebd6150909edbd79d366c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-04-08 10:33:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b6c169896391b2bff7cd842af19e85470bed0855","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 10:34:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"490c6f5346b0b6df0f5ffc2728d493bf887d114c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 10:40:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f82aedcac67e4a0e95d1f5c5a62453ce : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0047310559d42d9a2c4a44b9b8b7908 : SUCCESS in 6m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/972b3154b52346f0a646b586f701a67c : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":1},{"id":"a2f61d0abf5a01736f2b85b7f3ad06041a2f1083","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 10:40:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c27fea6ffc5d667196e102da10f9f90142af9370","revisions":{"c27fea6ffc5d667196e102da10f9f90142af9370":{"kind":"REWORK","_number":1,"created":"2021-04-07 12:27:03.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/66/785166/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/66/785166/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/785166/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/785166/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/785166/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/66/785166/1"}}},"commit":{"parents":[{"commit":"00586c5c8aaf9bc095b34424ec3fa37a277b6992","subject":"Merge \"Release tacker RC2 for Wallaby\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/00586c5c8aaf9bc095b34424ec3fa37a277b6992"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-04-07 12:23:54.000000000","tz":-240},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-04-07 12:26:57.000000000","tz":-240},"subject":"Cinder team releases for stable/train","message":"Cinder team releases for stable/train\n\nNot being released at this time:\n- cinderlib (no functional changes in branch)\n\nChange-Id: Ib59fd198fe74cabcc13bde7464241855a1fbb997\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c27fea6ffc5d667196e102da10f9f90142af9370"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c27fea6ffc5d667196e102da10f9f90142af9370"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
