)]}'
{"id":"openstack%2Freleases~785342","triplet_id":"openstack%2Freleases~master~I5cbe54f32b128fce070fe4e3b73e8bcf51dafe7b","project":"openstack/releases","branch":"master","topic":"wallaby-final-rc","hashtags":[],"change_id":"I5cbe54f32b128fce070fe4e3b73e8bcf51dafe7b","subject":"Final RC patch for ceilometer","status":"ABANDONED","created":"2021-04-08 08:42:50.000000000","updated":"2021-04-10 12:21:40.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"81cfe28b966be17dfa43785bb74aeb0d31e4a794","_number":785342,"virtual_id_number":785342,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-04-08 08:51:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-08 08:44:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-08 08:51:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-08 11:21:21.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"}],"messages":[{"id":"5874ba50b9728412e264de5c8873edcd3bff322f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-04-08 08:42:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4641ec39b41b3ea5edb37ac597347d25ed4c1c52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 08:51:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdc6a8dab6e24381b9f1dd73e7020cb7 : SUCCESS in 6m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f5f4b078b665497e8c9af1eb426cd010 : SUCCESS in 5m 23s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/2f0c398f9b8240879aa7025ea0251afc : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":1},{"id":"4b2af884527f81c61e8bc3c6a158371f2d93b436","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-04-09 18:24:59.000000000","message":"Patch Set 1:\n\nThe landed change seems to fix a worrying behavior, indeed, if tenacity is not pin it won\u0027t work as expected and we will try to use undefined method (removed), which that could be unexpected side effects at runtime","accounts_in_message":[],"_revision_number":1},{"id":"5f9183a321ad2a94a29e850d5724cf4778f2460f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-04-10 11:23:21.000000000","message":"Abandoned\n\nNo feedback from PTL/liaison","accounts_in_message":[],"_revision_number":1},{"id":"81cfe28b966be17dfa43785bb74aeb0d31e4a794","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-04-10 12:21:40.000000000","message":"Patch Set 1:\n\nsorry, I have been on pto, and apparently nobody else is looking at telemetry stuff.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1d4f406db96eeddfd68a36a87cde18a2e160459a","revisions":{"1d4f406db96eeddfd68a36a87cde18a2e160459a":{"kind":"REWORK","_number":1,"created":"2021-04-08 08:42:50.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/42/785342/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/42/785342/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/785342/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/785342/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/785342/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/42/785342/1"}}},"commit":{"parents":[{"commit":"12b3ad220af84c5bddef24b6d2520c0460f00303","subject":"Merge \"Release devstack-tools 1.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/12b3ad220af84c5bddef24b6d2520c0460f00303"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-04-08 08:42:42.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-04-08 08:42:42.000000000","tz":120},"subject":"Final RC patch for ceilometer","message":"Final RC patch for ceilometer\n\nThis creates a new RC before the final release candidate deadline to\nrelease the latest fixes and/or translations.\n\nPlease ack with a +1, or if there is some reason to hold off, leave a -1.\n\n$ git log --oneline --no-merges 16.0.0.0rc1..eeffd902\neeffd902 Cap tenacity \u003c 7.0.0\n8d190c8d Update TOX_CONSTRAINTS_FILE for stable/wallaby\ncb91f665 Update .gitreview for stable/wallaby\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I5cbe54f32b128fce070fe4e3b73e8bcf51dafe7b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1d4f406db96eeddfd68a36a87cde18a2e160459a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1d4f406db96eeddfd68a36a87cde18a2e160459a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
