)]}'
{"id":"openstack%2Freleases~785374","triplet_id":"openstack%2Freleases~master~Iff8df80aa6f3ba67c04e0674d7c72474f3a4e73a","project":"openstack/releases","branch":"master","topic":"final-rc","hashtags":[],"change_id":"Iff8df80aa6f3ba67c04e0674d7c72474f3a4e73a","subject":"Allow to change the used branch","status":"MERGED","created":"2021-04-08 11:03:10.000000000","updated":"2021-04-29 09:27:01.000000000","submitted":"2021-04-29 09:27:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"785374-final-rc","meta_rev_id":"e171a5d995a1d9f7707fa9aa11fbfd015a6d1414","_number":785374,"virtual_id_number":785374,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-29 09:27:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2021-04-29 09:19:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2021-04-26 15:03:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2021-04-29 09:19:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-08 11:17:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-12 08:25:40.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2021-04-15 17:56:15.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2021-04-26 15:03:16.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"5de9dd92ca237bd4de09def3827f93da90ed5d46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-04-08 11:03:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b15d2366a05c750842b746a2079018d0013b41fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 11:17:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a93afdf695d497c8d6fbb2b8f772e94 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5b6f79479f0431285844be35f4dd884 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25202fb46861407ab3c811b813ff4af8 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60dc42689e8e494b90487990ac2900cc : SUCCESS in 6m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/232cbbfe457d4d5dabdd0e6dd9c8ac09 : SUCCESS in 5m 36s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/de7dd828e0e84a24a242f301f5a45d7a : SUCCESS in 3m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/950ef3b4a38b4fe7897269b34404e2ec : SUCCESS in 4m 43s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/ba34768b4a9b40439dd9b88d0307cab5 : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":1},{"id":"c99f7af91e90a13c6848a0e4337c1dbf1e19f452","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-04-12 08:25:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3d946f1a445f00793284337779735aa3821e9fa1","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-04-15 17:56:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8f4d13eb0da83a1bc49206ee7954894eaf161a9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-04-26 12:21:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bbe1baaaa75225f7b8b7d11e6b6cc921d8769855","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-04-26 12:22:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"50cc47dea8ffb6d3147b92dec781f0696a4844b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 12:42:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c0f647a0f9646f8a2afe7e7910cc2f3 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f26986cab82c4d2bba7ab484b5c27e89 : FAILURE in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bccb81c912f84a1596bb07c6b94ee99f : FAILURE in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff1202b0004243c181db09c7038f5dc0 : SUCCESS in 6m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c3e219a15d59440c9b15e90bd2d97c8f : SUCCESS in 4m 56s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/4b87f74a55aa4368a4d5f17ce608670d : SUCCESS in 3m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ec3dbb7cdb374eceb79657c47a02d9a2 : FAILURE in 6m 06s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/af9c9675c2a24591b6c48ecdd201d21f : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":2},{"id":"5ffa81c49e21332e53ea190c5cd13e9c90199b73","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-04-26 14:33:25.000000000","message":"Patch Set 2:\n\nrecheck - it seems that creating stable/ocata in release-test repository while testing the branch delete ACL caused some unit test failure. I\u0027ve deleted now stable/ocata from release-test repository. Now the tests should pass. Sorry for that.","accounts_in_message":[],"_revision_number":2},{"id":"77f39625930ca1aaa92cb947a80341bdc9ba845f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-04-26 14:39:13.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e recheck - it seems that creating stable/ocata in release-test repository while testing the branch delete ACL caused some unit test failure. I\u0027ve deleted now stable/ocata from release-test repository. Now the tests should pass. Sorry for that.\n\nnp thanks for the heads up","accounts_in_message":[],"_revision_number":2},{"id":"4a262495657a9a0606470876974e46ef24a5bb5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 14:50:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5a0a920c0c9495baa0b4fae151c81da : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad1c6c5fdd0c4da4a3ec9afe68d394f5 : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/32f9e7151cf946ca9bccb701a602a48e : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a05c88ab89f14390ab52c153f38c0323 : SUCCESS in 10m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/84e4cd7a20e14fb19ca4fff9e368bd02 : SUCCESS in 5m 07s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/791871eccc414432b672599d6350dc30 : SUCCESS in 3m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/be4c5c39364d45ac8e6c147f02cbc838 : SUCCESS in 5m 34s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/8fffbaa0432140e8b6be5ed1b7b74f2f : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":2},{"id":"f801c77fc5679a56266a3a085fa986b1806c24b2","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-04-26 15:03:16.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks for the update, LGTM.","accounts_in_message":[],"_revision_number":2},{"id":"740942d2ec1dc6bd36ac69237dffcb4838ca8ec6","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-04-29 09:19:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9b40ea18df22805fdaebe989be8d708aa11fe2e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 09:19:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"367fa5407e33f46cc60e56bb5a4ebe32bfd8aaba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 09:27:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff01862cbc1c4d318411dfbc102882b7 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e00768619e3844d8868efd3c7fe60973 : SUCCESS in 6m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1096763869914c7d8779861c5b4da8d7 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f969e2c958a4934a3ff1199aec0f332 : SUCCESS in 5m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/915af5cc01f34f998673bb99add85ab0 : SUCCESS in 4m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/46b937fb6d9f46efaa852ecab39d2493 : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":2},{"id":"e171a5d995a1d9f7707fa9aa11fbfd015a6d1414","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 09:27:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"030a2200d23d30f002a57b68f3219bcf3400d978","revisions":{"2c8b7c21e8f6bff91b9cf770456cdf188d9f887a":{"kind":"REWORK","_number":1,"created":"2021-04-08 11:03:10.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/74/785374/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/74/785374/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/785374/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/785374/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/785374/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/74/785374/1"}}},"commit":{"parents":[{"commit":"00586c5c8aaf9bc095b34424ec3fa37a277b6992","subject":"Merge \"Release tacker RC2 for Wallaby\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/00586c5c8aaf9bc095b34424ec3fa37a277b6992"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-04-08 10:52:20.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-04-08 11:01:00.000000000","tz":120},"subject":"Allow to change the used branch","message":"Allow to change the used branch\n\nThe branch to use in process_auto_release is determined by the\nbranch linked to a series\u0027s status, however, at some point during\na series projects are branched and the status isn\u0027t yet updated\nso used branch refer to master, and so, to the next series.\n\nDuring the final-rc delta of changes between RC1 and RC2 is\nretrieved from master, which is wrong.\n\nThese changes allow to choose another branch if needed.\n\nThe default used branch will remain the series retrieved\nby status.\n\nChange-Id: Iff8df80aa6f3ba67c04e0674d7c72474f3a4e73a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2c8b7c21e8f6bff91b9cf770456cdf188d9f887a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2c8b7c21e8f6bff91b9cf770456cdf188d9f887a"}]},"branch":"refs/heads/master"},"030a2200d23d30f002a57b68f3219bcf3400d978":{"kind":"REWORK","_number":2,"created":"2021-04-26 12:21:30.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/74/785374/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/74/785374/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/785374/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/785374/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/785374/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/74/785374/2"}}},"commit":{"parents":[{"commit":"15570d163452edf7602281aeb07f07e721ae3261","subject":"Merge \"Release neutron-lib 2.11.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/15570d163452edf7602281aeb07f07e721ae3261"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-04-08 10:52:20.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-04-26 12:21:01.000000000","tz":120},"subject":"Allow to change the used branch","message":"Allow to change the used branch\n\nThe branch to use in process_auto_release is determined by the\nbranch linked to a series\u0027s status, however, at some point during\na series projects are branched and the status isn\u0027t yet updated\nso used branch refer to master, and so, to the next series.\n\nDuring the final-rc delta of changes between RC1 and RC2 is\nretrieved from master, which is wrong.\n\nThese changes allow to choose another branch if needed.\n\nThe default used branch will remain the series retrieved\nby status.\n\nChange-Id: Iff8df80aa6f3ba67c04e0674d7c72474f3a4e73a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/030a2200d23d30f002a57b68f3219bcf3400d978"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/030a2200d23d30f002a57b68f3219bcf3400d978"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
