)]}'
{"id":"openstack%2Freleases~785565","triplet_id":"openstack%2Freleases~master~I25c15fcfa426124e252eee6650d6a0f8d4425cb2","project":"openstack/releases","branch":"master","topic":"wallaby-final-rc","hashtags":[],"change_id":"I25c15fcfa426124e252eee6650d6a0f8d4425cb2","subject":"Proposing wallaby RC2 for Cyborg","status":"ABANDONED","created":"2021-04-09 07:08:05.000000000","updated":"2021-04-09 08:46:57.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a53ea3bfccd76371e1a6a704415f53fa1f51c0c4","_number":785565,"virtual_id_number":785565,"owner":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-04-09 07:16:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":-1,"date":"2021-04-09 07:34:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":-1,"date":"2021-04-09 07:08:50.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-09 07:08:50.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"state":"REVIEWER"},{"updated":"2021-04-09 07:09:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-09 07:16:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-09 07:34:26.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"df5df3267d4dceeb094ad5f6ab9e7da60c443dae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2021-04-09 07:08:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"860d7e42cf39e38e8f39f6598a6514d7229f36f2","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2021-04-09 07:08:50.000000000","message":"Patch Set 1: Workflow-1\n\n-1 W Until https://review.opendev.org/c/openstack/cyborg/+/785314 merged","accounts_in_message":[],"_revision_number":1},{"id":"0bb35ba0b3f943f333d9a20e828a4504014d33e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 07:16:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9673951787304f9cbd57bbcef155e3ae : SUCCESS in 7m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4d63f5d435db4a1c853d4521e4188ef0 : FAILURE in 6m 11s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/bdca6c06551e4deb9b9d6b20977fdd96 : FAILURE in 5m 47s","accounts_in_message":[],"_revision_number":1},{"id":"14c3d3300daf9ce8ddd850e34c09dcbef7ccdabe","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-04-09 07:34:26.000000000","message":"Patch Set 1: Code-Review-1\n\nHello,\n\nFirst thing, this patch is duplicated with this one where zuul is ok \u003d\u003e https://review.opendev.org/c/openstack/releases/+/785344\n\nSecondly, I can\u0027t find the given hash in the cyborg repo, it don\u0027t seems to exist.\n\nThird, the waited patch is against master and not stable/wallaby so it should be backported to stable/wallaby once it will be merged (https://review.opendev.org/c/openstack/cyborg/+/785314).\nRC2 must be based on stable/wallaby.\n\nAlso notice that the other patch is already pointed on the latest changes available for stable/wallaby https://review.opendev.org/c/openstack/releases/+/785344\n\nThanks","accounts_in_message":[],"_revision_number":1},{"id":"1a7e6dd108ba37adf5e5fd3cce83c70a4a78e10e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-04-09 07:37:51.000000000","message":"Patch Set 1:\n\nAlso notice that today we will propose the final release (dropping RC) for RC projects, so we can\u0027t wait too long.","accounts_in_message":[],"_revision_number":1},{"id":"a53ea3bfccd76371e1a6a704415f53fa1f51c0c4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2021-04-09 08:46:57.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8299772ffe97ad3e70b703e4bc8d0208a2d55318","revisions":{"8299772ffe97ad3e70b703e4bc8d0208a2d55318":{"kind":"REWORK","_number":1,"created":"2021-04-09 07:08:05.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/65/785565/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/65/785565/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/785565/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/785565/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/785565/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/65/785565/1"}}},"commit":{"parents":[{"commit":"c1056dfbd7ed75657e96542ea12fb18c8166a060","subject":"Release manila-ui rc2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c1056dfbd7ed75657e96542ea12fb18c8166a060"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2021-04-09 07:04:09.000000000","tz":480},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2021-04-09 07:05:04.000000000","tz":480},"subject":"Proposing wallaby RC2 for Cyborg","message":"Proposing wallaby RC2 for Cyborg\n\nRelease 6.0.0.0rc2\n\nWe merged some issue fixes as below:\nroot@ubuntu:/home/brinzhang/cyborg# git log --oneline --no-merges 6.0.0.0rc1..a8979b2\n9fc6a30 Switch using Launchpad instead of Storyboard\nabe9f40 Using pop method to avoid incompatible data type\nd9233bc Switch using Launchpad instead of Storyboard\n0e1bb83 Add backup and snapshot support matrix\n008c366 Revert device_profile key to show device profile API\n891c1de tools: Allow check-cherry-picks.sh to be disabled by an env var\n4fac92a Add Python3 xena unit tests\n851dcfc Update master for stable/wallaby\n8cddcf8 (master1) Travil Change: Correct comment errors\n\nChange-Id: I25c15fcfa426124e252eee6650d6a0f8d4425cb2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8299772ffe97ad3e70b703e4bc8d0208a2d55318"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8299772ffe97ad3e70b703e4bc8d0208a2d55318"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
