)]}'
{"id":"openstack%2Freleases~788241","triplet_id":"openstack%2Freleases~master~I7b2bc5d74394994706978cd32ad5f5a0f03281c5","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I7b2bc5d74394994706978cd32ad5f5a0f03281c5","subject":"Add glance-tempest-plugin to wallaby deliverables","status":"MERGED","created":"2021-04-27 10:36:38.000000000","updated":"2021-05-05 15:52:22.000000000","submitted":"2021-05-05 15:52:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"788241","meta_rev_id":"00012ff74a561a49538773cecc01ee713d56d79a","_number":788241,"virtual_id_number":788241,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-05 15:52:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2021-05-05 15:40:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2021-05-05 06:51:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2021-05-04 11:31:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2021-05-05 15:40:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-05-05 06:56:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-27 10:37:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-27 10:56:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-27 12:11:22.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-05-05 06:23:37.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-05-05 15:40:38.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"92f76c6a7d5dcfe8e8ffa92ccddf5e4a73ef014a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-04-27 10:36:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e8c6a1c2e45d69133833166a4d1a90a1c86d64c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 10:56:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29dd2d5766c24499aabedd7ff92985c1 : SUCCESS in 6m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/344211578f32406aabe3da9c252d1272 : FAILURE in 5m 50s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/5f3968d9231d4332b540bade8d65a992 : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":1},{"id":"bbdbdbd21be186e4eb739edb45aec89efa0a5849","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-04-27 12:11:22.000000000","message":"Patch Set 1: Code-Review-1\n\nHello,\n\nSome pieces are missing.\n\nYou must first append `publish-to-pypi` to the templates of your project zuul\u0027s config [1].\n\nSee an example [2] with barbican their tempest plugin.\n\nAlso, when you\u0027ll have proposed a patch to update your config, you can add a `depends-on` this patch here (in the commit message). The CI will be relaunched and tests will pass.\n\n[1] https://opendev.org/openstack/project-config/src/branch/master/zuul.d/projects.yaml#L2560\n[2] https://opendev.org/openstack/project-config/src/branch/master/zuul.d/projects.yaml#L1284-L1287","accounts_in_message":[],"_revision_number":1},{"id":"a315e0012841365717ba5d954d06e6c1356a8056","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-04-27 12:18:02.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9706698b50158316ce1dcdb47146bbb23706b6a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 12:36:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aecb20148b1c458db6c36de936c190cb : SUCCESS in 6m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8a6aae60468141cc97ef1d6e339da36e : FAILURE in 4m 59s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/72d822c919844f909cf013a96f60cd27 : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":2},{"id":"baa169a289d090b9587218f08d0d1aa6893b3c68","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-05-03 04:58:32.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"88f597cc2fb45e4daa56bc455502cceff252f6f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 05:09:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/094f8f2206434426b98a52edbdc27b44 : SUCCESS in 6m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dd62808a466f4849ba299c45620d9c04 : SUCCESS in 5m 18s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ec98f997980944988d33ddac1e416286 : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":2},{"id":"1a1c12371c17864da8b2a55b2eefa6e4e80bbadd","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-05-04 11:31:55.000000000","message":"Patch Set 2: Code-Review+2\n\nThe missing pieces seems to have been added, tests are now ok, need PTL/liaison validation","accounts_in_message":[],"_revision_number":2},{"id":"01d5860af3d8749b1d4ffd43adc22e24a53f2718","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-05-05 06:51:59.000000000","message":"Patch Set 2: Code-Review+1\n\nDependent patch is merged, so this is good to go.","accounts_in_message":[],"_revision_number":2},{"id":"c02bb5fdad3112049b87d040968f5fd6adce2229","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 06:56:29.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"81a17b6b13f5fba0943b3fa521dae800dbce8401","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-05-05 15:40:38.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLooks OK to me + as the former comments say. Let\u0027s do the release.","accounts_in_message":[],"_revision_number":2},{"id":"223606b1d6bab8d3328b4dd74039d1d1dafa9d3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 15:43:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"635a9ca3e8397b5d6b8b93777318e31b1621d227","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 15:52:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/399797bf7c20489f80fbc58cfa40e373 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b12c489301c46b9b1f264226f1e8b71 : SUCCESS in 7m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2d3c8f339ea64b70b046e6d4798df3a3 : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":2},{"id":"00012ff74a561a49538773cecc01ee713d56d79a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 15:52:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"59bbfa88f8d1c7012f1878115cf6f4ef779cdc2a","revisions":{"ec88ad193c0cadfb2f24816a1c05eaf5bde725b9":{"kind":"REWORK","_number":1,"created":"2021-04-27 10:36:38.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/41/788241/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/41/788241/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/788241/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/788241/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/788241/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/41/788241/1"}}},"commit":{"parents":[{"commit":"f1215e0b79f222fa8213665f1d3048331b66e782","subject":"Merge \"pbr 5.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f1215e0b79f222fa8213665f1d3048331b66e782"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-04-27 09:38:34.000000000","tz":330},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-04-27 10:34:23.000000000","tz":330},"subject":"Add glance-tempest-plugin to wallaby deliverables","message":"Add glance-tempest-plugin to wallaby deliverables\n\nAdding glance-tempest-plugin deliverables for wallaby cycle and\ninitial release\n\nChange-Id: I7b2bc5d74394994706978cd32ad5f5a0f03281c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ec88ad193c0cadfb2f24816a1c05eaf5bde725b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ec88ad193c0cadfb2f24816a1c05eaf5bde725b9"}]},"branch":"refs/heads/master"},"59bbfa88f8d1c7012f1878115cf6f4ef779cdc2a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-04-27 12:18:02.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/41/788241/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/41/788241/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/788241/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/788241/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/788241/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/41/788241/2"}}},"commit":{"parents":[{"commit":"f1215e0b79f222fa8213665f1d3048331b66e782","subject":"Merge \"pbr 5.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f1215e0b79f222fa8213665f1d3048331b66e782"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-04-27 09:38:34.000000000","tz":330},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-04-27 12:17:29.000000000","tz":0},"subject":"Add glance-tempest-plugin to wallaby deliverables","message":"Add glance-tempest-plugin to wallaby deliverables\n\nAdding glance-tempest-plugin deliverables for wallaby cycle and\ninitial release\n\nDepends-On: I61d93fdf0132a2bd0aea731f43ecbbbbf0210ea9\nChange-Id: I7b2bc5d74394994706978cd32ad5f5a0f03281c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/59bbfa88f8d1c7012f1878115cf6f4ef779cdc2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/59bbfa88f8d1c7012f1878115cf6f4ef779cdc2a"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
