)]}'
{"id":"openstack%2Freleases~789064","triplet_id":"openstack%2Freleases~master~I012696f62c41915b71dfc12f3c658decc26e9644","project":"openstack/releases","branch":"master","topic":"artifact-signing","hashtags":[],"change_id":"I012696f62c41915b71dfc12f3c658decc26e9644","subject":"List the Xena cycle signing key as current","status":"MERGED","created":"2021-05-01 19:37:29.000000000","updated":"2021-05-06 07:47:12.000000000","submitted":"2021-05-06 07:47:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"789064-artifact-signing","meta_rev_id":"22f79f3052e5434c3aa2ea9a3ca854674d49076d","_number":789064,"virtual_id_number":789064,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-06 07:47:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2021-05-06 07:37:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2021-05-06 07:37:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-01 19:50:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-03 07:50:12.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-05-03 08:38:07.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"9fc6adcb9cf682cac96394bec5c2ff93dae43dad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-05-01 19:37:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"88e07321f592ede6779150c574e0a857ce81e440","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-01 19:50:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dbc2276c5fe41b48d38609ecedf3a33 : SUCCESS in 6m 41s","accounts_in_message":[],"_revision_number":1},{"id":"1cd61ff9a70a0adac2dd431a0a531f4de0c26c8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-05-03 07:50:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b69f9629b0791fc79ecb738ccd202a4cb2d37665","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-05-03 07:51:13.000000000","message":"Patch Set 2: Code-Review+2\n\nDates updated with PS2 as suggested","accounts_in_message":[],"_revision_number":2},{"id":"5a78b081a84fce7044b21eeb69317d8f233edc31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 07:57:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4ad863b784f49dbbbee55d7d5479030 : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":2},{"id":"80aa2b6b07ff8b3dab610a9e35bc30eb5d0b0815","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-05-03 08:38:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fa121dc99354f08642158e894ac3377ce28345d6","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-05-03 16:43:54.000000000","message":"Patch Set 2:\n\nNote that the dependency for this has not merged yet.","accounts_in_message":[],"_revision_number":2},{"id":"c72765f7e5846356f6f80b53607f1fa0a258c8e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-05-06 01:11:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a680ac3e550d96fddfbe7883f5e6d53f9379886e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-05-06 01:11:51.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"cfd6f6d220502d01adf23d969f2a5a1c7f42f4d2","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-05-06 01:13:03.000000000","message":"Patch Set 4:\n\nI\u0027ve now revised this to reflect the date the key rotation merged, and cleaned up the commit message accordingly.","accounts_in_message":[],"_revision_number":4},{"id":"a7172a8177007a546b140f0a1fca0c2b3607115a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 01:19:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/778bc5eeddbb4969adea724535819855 : SUCCESS in 6m 40s","accounts_in_message":[],"_revision_number":4},{"id":"8aada674bef34883937163341ab816ad0834f1cf","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-05-06 07:37:03.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"9754bbcfb0906ec44da0ce481b0c5e7fec8e7ac5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 07:37:21.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"4707c8322e71c61d122c3a3caf958b1f8df18dfe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 07:47:09.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/845e3cbe28b94bc7b86ff12d6fc8526b : SUCCESS in 6m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e54cfdfa4f294b6fb880708327c9c603 : SUCCESS in 8m 13s","accounts_in_message":[],"_revision_number":4},{"id":"22f79f3052e5434c3aa2ea9a3ca854674d49076d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 07:47:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e01fd273dcea92164655543bc32c7cbcf3b174a6","revisions":{"7c4f99f1d2299632b9b7214199223e30e69f6013":{"kind":"REWORK","_number":1,"created":"2021-05-01 19:37:29.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/64/789064/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/64/789064/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/64/789064/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/64/789064/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/64/789064/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/64/789064/1"}}},"commit":{"parents":[{"commit":"f67b410744e2cfbd66d7cf28f9903be201de61c5","subject":"Merge \"Update Telemetry release liaison\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f67b410744e2cfbd66d7cf28f9903be201de61c5"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-05-01 19:35:45.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-05-01 19:35:45.000000000","tz":0},"subject":"List the Xena cycle signing key as current","message":"List the Xena cycle signing key as current\n\nThis sets the end date for the Wallaby key, adds the Xena key\nstarting today and links to a minimal export of the corresponding\npublic key and its details on the keyserver network. This can be\namended with a new patchset or a follow-up change to correct the\nincluded dates after the key material change (Depends-On below)\nmerges and goes into effect.\n\nDepends-On: https://review.opendev.org/789063\nChange-Id: I012696f62c41915b71dfc12f3c658decc26e9644\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7c4f99f1d2299632b9b7214199223e30e69f6013"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7c4f99f1d2299632b9b7214199223e30e69f6013"}]},"branch":"refs/heads/master"},"7bb5f066be8e920c7d62dc27b7e8fd3de1178813":{"kind":"REWORK","_number":2,"created":"2021-05-03 07:50:12.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/64/789064/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/64/789064/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/64/789064/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/64/789064/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/64/789064/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/64/789064/2"}}},"commit":{"parents":[{"commit":"f67b410744e2cfbd66d7cf28f9903be201de61c5","subject":"Merge \"Update Telemetry release liaison\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f67b410744e2cfbd66d7cf28f9903be201de61c5"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-05-01 19:35:45.000000000","tz":0},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-05-03 07:50:01.000000000","tz":120},"subject":"List the Xena cycle signing key as current","message":"List the Xena cycle signing key as current\n\nThis sets the end date for the Wallaby key, adds the Xena key\nstarting today and links to a minimal export of the corresponding\npublic key and its details on the keyserver network. This can be\namended with a new patchset or a follow-up change to correct the\nincluded dates after the key material change (Depends-On below)\nmerges and goes into effect.\n\nDepends-On: https://review.opendev.org/789063\nChange-Id: I012696f62c41915b71dfc12f3c658decc26e9644\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7bb5f066be8e920c7d62dc27b7e8fd3de1178813"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7bb5f066be8e920c7d62dc27b7e8fd3de1178813"}]},"branch":"refs/heads/master"},"be79927be3a5b5b4482fc6931a4811f51bc0aaa2":{"kind":"REWORK","_number":3,"created":"2021-05-06 01:11:21.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/64/789064/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/64/789064/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/64/789064/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/64/789064/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/64/789064/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/64/789064/3"}}},"commit":{"parents":[{"commit":"f67b410744e2cfbd66d7cf28f9903be201de61c5","subject":"Merge \"Update Telemetry release liaison\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f67b410744e2cfbd66d7cf28f9903be201de61c5"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-05-01 19:35:45.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-05-06 01:11:04.000000000","tz":0},"subject":"List the Xena cycle signing key as current","message":"List the Xena cycle signing key as current\n\nThis sets the end date for the Wallaby key, adds the Xena key\nstarting today and links to a minimal export of the corresponding\npublic key and its details on the keyserver network. This can be\namended with a new patchset or a follow-up change to correct the\nincluded dates after the key material change (Depends-On below)\nmerges and goes into effect.\n\nDepends-On: https://review.opendev.org/789063\nChange-Id: I012696f62c41915b71dfc12f3c658decc26e9644\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/be79927be3a5b5b4482fc6931a4811f51bc0aaa2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/be79927be3a5b5b4482fc6931a4811f51bc0aaa2"}]},"branch":"refs/heads/master"},"e01fd273dcea92164655543bc32c7cbcf3b174a6":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-05-06 01:11:51.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/64/789064/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/64/789064/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/64/789064/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/64/789064/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/64/789064/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/64/789064/4"}}},"commit":{"parents":[{"commit":"f67b410744e2cfbd66d7cf28f9903be201de61c5","subject":"Merge \"Update Telemetry release liaison\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f67b410744e2cfbd66d7cf28f9903be201de61c5"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-05-01 19:35:45.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2021-05-06 01:11:26.000000000","tz":0},"subject":"List the Xena cycle signing key as current","message":"List the Xena cycle signing key as current\n\nThis sets the end date for the Wallaby key, adds the Xena key\nstarting today and links to a minimal export of the corresponding\npublic key and its details on the keyserver network.\n\nChange-Id: I012696f62c41915b71dfc12f3c658decc26e9644\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e01fd273dcea92164655543bc32c7cbcf3b174a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e01fd273dcea92164655543bc32c7cbcf3b174a6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
