)]}'
{"id":"openstack%2Freleases~789587","triplet_id":"openstack%2Freleases~master~Ia899b39dbc68b245e1b84518363de5420af500c0","project":"openstack/releases","branch":"master","topic":"move-ipa-builder","hashtags":[],"change_id":"Ia899b39dbc68b245e1b84518363de5420af500c0","subject":"ironic-python-agent-builder release model change","status":"MERGED","created":"2021-05-04 14:48:33.000000000","updated":"2021-05-10 10:56:49.000000000","submitted":"2021-05-10 10:56:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"789587-move-ipa-builder","meta_rev_id":"7c0051f238499dded6544bbd6fa6937a32d235f5","_number":789587,"virtual_id_number":789587,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-10 10:56:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2021-05-10 09:03:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2021-05-10 10:49:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2021-05-07 11:01:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2021-05-10 10:49:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-04 14:50:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-04 14:59:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-04 15:12:36.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"CC"},{"updated":"2021-05-04 15:31:03.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"CC"},{"updated":"2021-05-04 15:48:54.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2021-05-04 16:05:32.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"CC"},{"updated":"2021-05-05 15:20:11.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2021-05-06 09:38:46.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2021-05-06 11:16:01.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2021-05-06 12:20:11.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2021-05-10 10:49:19.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"4d10eb754a6e3deea5694f9530bf9970fcb3af43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-05-04 14:48:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a5923a9dcd5e147d7e862f19a32a274718cb157","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 14:59:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7419a17f1bf49d8899a7781554607d5 : SUCCESS in 6m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c786b1f095b84ddcb51e918a3471b2c3 : FAILURE in 9m 07s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d9d04635aa184bb5aaea9d7ec6a6c53d : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":1},{"id":"9fc69fcb3f480ba683ecd767947e600a62457e7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-05-04 15:09:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1831014b31798953d217df705c3793c39a194909","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 15:19:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba57237e9e7d4903a7463055f5fe110a : SUCCESS in 8m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8f2038875ee34ef8bed9d8bf44e5dcde : SUCCESS in 4m 55s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/fbb5ce13f65847f48a94f94c15483690 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":2},{"id":"10360b7fe5389edb20f675bd3c6cb2512f0f974a","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-05-04 15:48:54.000000000","message":"Patch Set 2: Code-Review+1\n\nFrom SemVer standpoint this should likely by 2.6.1, but okay.","accounts_in_message":[],"_revision_number":2},{"id":"f4ffd389aca7a3713585ba438e65cc4627e76c28","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 15:49:54.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"ce5fba7cd027efc5b147bdb8e856f8afe6eae088","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-05-04 16:05:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6d2f9170842e58ec46e51c14cabd1424e91d0726","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-05-04 19:27:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"db3f377fd4652e5908ddd3eca78de9a90bdac274","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-05-05 08:55:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3c1dc7c1499f0e0a06f0dced7a6a299c0782e3ca","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-05-05 13:12:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9c080fc069358897b5c6ff638f3f85c9b54c2fca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-05-05 13:21:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"32eef8b86dff41a8cdcb34c2fd09ba5ae55a7f43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 13:32:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f14db64a2d3d4c9f944253a38cf1053c : SUCCESS in 6m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b057a3b8c2da4b5699ebc3df223e4f1b : SUCCESS in 6m 47s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/25e0abe4a8cc4496a8ea18cc974310a3 : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":3},{"id":"a929b760f9cb2e57cef088b0970327d86b917f05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-05-05 13:54:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"37386779ff67d612ed63c3e2a4b89505d19b1e83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 14:02:36.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34f95c3d0bbc4d9792873cf1453ee646 : SUCCESS in 6m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1fd36da8016246b3aaed1d154559e9bb : SUCCESS in 5m 22s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/3c5a082d9084479f8dcc461083c9f63d : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":4},{"id":"cb8dfaf7e16a99e5c3502f6d77083f00354171a0","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-05-06 09:38:46.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0c0ffde0af5a67be5ea7e5c89faaec1da5282134","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-05-06 11:16:01.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"c0ed6ea99d31f9b17230974281c92995b9021621","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-05-06 11:19:29.000000000","message":"Patch Set 4: -Code-Review\n\nWe usually remove the old file because the way the history is displayed on releases.o.o ends up to be more confusing than not displaying any history at all (see how it shows an old version as current if you look at https://releases.openstack.org/independent.html#ironic-python-agent-builder )","accounts_in_message":[],"_revision_number":4},{"id":"41f3ebb73f3bb3e1555e04fcd5eadb0227932fcc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-05-07 09:38:15.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6cf161dfc9ca53864430bb067a3ee93afa676ae4","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-05-07 09:40:07.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"36942a13474afeaab8886399359d2f4b1e39f63b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-05-07 09:44:48.000000000","message":"Patch Set 5:\n\n\u003e Patch Set 4: -Code-Review\n\u003e \n\u003e We usually remove the old file because the way the history is displayed on releases.o.o ends up to be more confusing than not displaying any history at all (see how it shows an old version as current if you look at https://releases.openstack.org/independent.html#ironic-python-agent-builder )\n\nThanks Thierry for your comment, I think you\u0027re right. I was not totally sure if we should kept this file or not, as usually we keep previous series files when we transition deliverables to independent I thought we should do the same here.\nFile removed with PS5.","accounts_in_message":[],"_revision_number":5},{"id":"832721a30346200d30d53801a8a1059fabbb4fe7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 09:48:40.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9f95c50f4984054a1ac4a7fcaf74512 : SUCCESS in 8m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d617b0e616024978891e70cf9984a8d1 : SUCCESS in 4m 55s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f82ce64605ab4933ad2d34b3c91f3455 : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":5},{"id":"43dc97d5b0f882d7f06f0f8ad593b436fe890add","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-05-07 10:54:51.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"f77cbea6c727c28830e25ef6aeb72ce420d96ac7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 11:01:51.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e277f6be6c6402e889bc9cea6f89130 : SUCCESS in 5m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ce1441ddfaba455a9602b6579bc59ad0 : SUCCESS in 4m 23s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d42a2ec983124fc5b4b7ecc0e5b6e126 : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":6},{"id":"ed5a040c5ba25154c4de5fd6147f6e03c54dcd78","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-05-07 11:01:56.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"c94c5b92130f4226c6495bb00194a0bb00e8e98b","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-05-10 09:03:14.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"96265f34b3abc16f9b9027be0bd0f22990ee7648","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-05-10 09:37:52.000000000","message":"Patch Set 6:\n\nRiccardo is release liaison https://review.opendev.org/c/openstack/releases/+/789381","accounts_in_message":[],"_revision_number":6},{"id":"2b9d4c8cb8d9ad14bdf3bb2338e028dfab258776","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-05-10 10:49:19.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\nReading the long commit message and the comments this change looks OK to me. Also, approved by release liaison.","accounts_in_message":[],"_revision_number":6},{"id":"0dcc1111d318d7fe6c3e3235b2a70fe61295a857","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 10:49:37.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"2de44ff3900b473ad4230e502758082b56b7a392","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 10:56:49.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98d88b273ecc433fa09fb6497aa88883 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad5abbc134ca4a3ea7d01b1bf03488b2 : SUCCESS in 6m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0de151d0d8c947f28b387415cfdb9a34 : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":6},{"id":"7c0051f238499dded6544bbd6fa6937a32d235f5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 10:56:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"e96978a74ad7bac2175cd1b4069b07c6f3641133","revisions":{"0449b786b773d83caa9d213fcae50734018138b7":{"kind":"REWORK","_number":1,"created":"2021-05-04 14:48:33.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/87/789587/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/87/789587/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/789587/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/789587/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/789587/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/87/789587/1"}}},"commit":{"parents":[{"commit":"632b30df6d3bfe5dba93813f087b1e1b7532a486","subject":"Fix list_eol_stale_branches.sh to list only stale branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/632b30df6d3bfe5dba93813f087b1e1b7532a486"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-05-04 14:27:11.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-05-04 14:45:49.000000000","tz":120},"subject":"ironic-python-agent-builder release model change","message":"ironic-python-agent-builder release model change\n\nDuring the recent xena ptg, the ironic community had a discussion about\nthe need to move the ironic-python-agent-builder project from an\nindependent model to the standard release model.\n\nWhen we initially split the builder from ironic-python-agent, we decided\nagainst it, but considering some problems we encountered during the\nroad, the ironic community seems to be in favor of the change.\nThe reasons for this are mainly to strictly align the image building\nproject to ironic-python-agent releases, and ease dealing with the\noccasional upgrade of tinycore linux, the base image used to build the\n\"tinyipa\" ironic-python-agent ramdisk.\n\nWe decided to split them at some point, and we\u0027d like to keep them\nseparated, but we realized that it would make sense to keep them synced\nby branch.\n\nThe ironic contributor suffered CI pain over the\npast year where code handling distributions has broken several times.\nLargely these breaks have had to be navigated around with specific\nversion pinning, and settings changes, because we were unable to fix\nsomething a release or two back inside the tooling, in this case\nironic-python-agent-builder. In other words, if the ironic team had that\ncapability and those branches/tags, we would have been less in a world of\npain because we wouldn\u0027t have had to make changes on every job/branch\nacross every repository where we were hitting issues with ipa-b. Reduction\nof pain is why this change is being proposed.\n\nWe should also notice that the Kolla team also use ipa-builder and they\nfaced similar issues with Kayobe.\n\nThese changes propose to cut ipa-builder for Wallaby in a retro active\nmanner (Wallaby is already out at this point).\n\nThese changes use the same model that the one used by\nironic-python-agent.\n\nThese changes also init ipa-builder for Xena.\n\nhttp://lists.openstack.org/pipermail/openstack-discuss/2021-April/022032.html\n\nChange-Id: Ia899b39dbc68b245e1b84518363de5420af500c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0449b786b773d83caa9d213fcae50734018138b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0449b786b773d83caa9d213fcae50734018138b7"}]},"branch":"refs/heads/master"},"e65682caa4592dd105c319d461bc6bd4a0f0eaea":{"kind":"REWORK","_number":2,"created":"2021-05-04 15:09:22.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/87/789587/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/87/789587/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/789587/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/789587/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/789587/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/87/789587/2"}}},"commit":{"parents":[{"commit":"632b30df6d3bfe5dba93813f087b1e1b7532a486","subject":"Fix list_eol_stale_branches.sh to list only stale branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/632b30df6d3bfe5dba93813f087b1e1b7532a486"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-05-04 14:27:11.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-05-04 15:08:09.000000000","tz":120},"subject":"ironic-python-agent-builder release model change","message":"ironic-python-agent-builder release model change\n\nDuring the recent xena ptg, the ironic community had a discussion about\nthe need to move the ironic-python-agent-builder project from an\nindependent model to the standard release model.\n\nWhen we initially split the builder from ironic-python-agent, we decided\nagainst it, but considering some problems we encountered during the\nroad, the ironic community seems to be in favor of the change.\nThe reasons for this are mainly to strictly align the image building\nproject to ironic-python-agent releases, and ease dealing with the\noccasional upgrade of tinycore linux, the base image used to build the\n\"tinyipa\" ironic-python-agent ramdisk.\n\nWe decided to split them at some point, and we\u0027d like to keep them\nseparated, but we realized that it would make sense to keep them synced\nby branch.\n\nThe ironic contributor suffered CI pain over the\npast year where code handling distributions has broken several times.\nLargely these breaks have had to be navigated around with specific\nversion pinning, and settings changes, because we were unable to fix\nsomething a release or two back inside the tooling, in this case\nironic-python-agent-builder. In other words, if the ironic team had that\ncapability and those branches/tags, we would have been less in a world of\npain because we wouldn\u0027t have had to make changes on every job/branch\nacross every repository where we were hitting issues with ipa-b. Reduction\nof pain is why this change is being proposed.\n\nWe should also notice that the Kolla team also use ipa-builder and they\nfaced similar issues with Kayobe.\n\nThese changes propose to cut ipa-builder for Wallaby in a retro active\nmanner (Wallaby is already out at this point).\n\nThese changes use the same model that the one used by\nironic-python-agent.\n\nThese changes also init ipa-builder for Xena.\n\nThese changes keep the independent files for the sake of a proper\nhistory (previous releases history of ipa-builder).\n\nhttp://lists.openstack.org/pipermail/openstack-discuss/2021-April/022032.html\n\nChange-Id: Ia899b39dbc68b245e1b84518363de5420af500c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e65682caa4592dd105c319d461bc6bd4a0f0eaea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e65682caa4592dd105c319d461bc6bd4a0f0eaea"}]},"branch":"refs/heads/master"},"9ee65fbbc2a8487705bef313ff90ae6e6b34f098":{"kind":"REWORK","_number":3,"created":"2021-05-05 13:21:06.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/87/789587/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/87/789587/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/789587/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/789587/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/789587/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/87/789587/3"}}},"commit":{"parents":[{"commit":"632b30df6d3bfe5dba93813f087b1e1b7532a486","subject":"Fix list_eol_stale_branches.sh to list only stale branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/632b30df6d3bfe5dba93813f087b1e1b7532a486"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-05-04 14:27:11.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-05-05 13:19:58.000000000","tz":120},"subject":"ironic-python-agent-builder release model change","message":"ironic-python-agent-builder release model change\n\nDuring the recent xena ptg, the ironic community had a discussion about\nthe need to move the ironic-python-agent-builder project from an\nindependent model to the standard release model.\n\nWhen we initially split the builder from ironic-python-agent, we decided\nagainst it, but considering some problems we encountered during the\nroad, the ironic community seems to be in favor of the change.\nThe reasons for this are mainly to strictly align the image building\nproject to ironic-python-agent releases, and ease dealing with the\noccasional upgrade of tinycore linux, the base image used to build the\n\"tinyipa\" ironic-python-agent ramdisk.\n\nWe decided to split them at some point, and we\u0027d like to keep them\nseparated, but we realized that it would make sense to keep them synced\nby branch.\n\nThe ironic contributor suffered CI pain over the\npast year where code handling distributions has broken several times.\nLargely these breaks have had to be navigated around with specific\nversion pinning, and settings changes, because we were unable to fix\nsomething a release or two back inside the tooling, in this case\nironic-python-agent-builder. In other words, if the ironic team had that\ncapability and those branches/tags, we would have been less in a world of\npain because we wouldn\u0027t have had to make changes on every job/branch\nacross every repository where we were hitting issues with ipa-b. Reduction\nof pain is why this change is being proposed.\n\nWe should also notice that the Kolla team also use ipa-builder and they\nfaced similar issues with Kayobe.\n\nThese changes propose to cut ipa-builder for Wallaby in a retro active\nmanner (Wallaby is already out at this point).\n\nThese changes use the same model that the one used by\nironic-python-agent.\n\nThese changes create a new tag (2.7.0) based on the same hash that the\nprevious independent version (2.6.0), mostly to properly init the series\nwith a meaningful version number, avoid misleading with future bugfix\nreleases versionned with a minor bump, create the wallaby branch from the\nright point, avoid errors with our tooling related to missing tags for a\nseries.\n\nThese changes also init ipa-builder for Xena.\n\nThese changes keep the independent files for the sake of a proper\nhistory (previous releases history of ipa-builder).\n\nhttp://lists.openstack.org/pipermail/openstack-discuss/2021-April/022032.html\n\nChange-Id: Ia899b39dbc68b245e1b84518363de5420af500c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9ee65fbbc2a8487705bef313ff90ae6e6b34f098"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9ee65fbbc2a8487705bef313ff90ae6e6b34f098"}]},"branch":"refs/heads/master"},"4ae8d6e46b162e40fe5b1962e99053ef2d0f5ddb":{"kind":"REWORK","_number":4,"created":"2021-05-05 13:54:38.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/87/789587/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/87/789587/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/789587/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/789587/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/789587/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/87/789587/4"}}},"commit":{"parents":[{"commit":"632b30df6d3bfe5dba93813f087b1e1b7532a486","subject":"Fix list_eol_stale_branches.sh to list only stale branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/632b30df6d3bfe5dba93813f087b1e1b7532a486"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-05-04 14:27:11.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-05-05 13:54:22.000000000","tz":120},"subject":"ironic-python-agent-builder release model change","message":"ironic-python-agent-builder release model change\n\nDuring the recent xena ptg, the ironic community had a discussion about\nthe need to move the ironic-python-agent-builder project from an\nindependent model to the standard release model.\n\nWhen we initially split the builder from ironic-python-agent, we decided\nagainst it, but considering some problems we encountered during the\nroad, the ironic community seems to be in favor of the change.\nThe reasons for this are mainly to strictly align the image building\nproject to ironic-python-agent releases, and ease dealing with the\noccasional upgrade of tinycore linux, the base image used to build the\n\"tinyipa\" ironic-python-agent ramdisk.\n\nWe decided to split them at some point, and we\u0027d like to keep them\nseparated, but we realized that it would make sense to keep them synced\nby branch.\n\nThe ironic contributor suffered CI pain over the\npast year where code handling distributions has broken several times.\nLargely these breaks have had to be navigated around with specific\nversion pinning, and settings changes, because we were unable to fix\nsomething a release or two back inside the tooling, in this case\nironic-python-agent-builder. In other words, if the ironic team had that\ncapability and those branches/tags, we would have been less in a world of\npain because we wouldn\u0027t have had to make changes on every job/branch\nacross every repository where we were hitting issues with ipa-b. Reduction\nof pain is why this change is being proposed.\n\nWe should also notice that the Kolla team also use ipa-builder and they\nfaced similar issues with Kayobe.\n\nThese changes propose to cut ipa-builder for Wallaby in a retro active\nmanner (Wallaby is already out at this point).\n\nThese changes use the same model that the one used by\nironic-python-agent.\n\nThese changes create a new tag (2.7.0) based on the same hash that the\nprevious independent version (2.6.0), mostly to properly init the series\nwith a meaningful version number, avoid misleading with future bugfix\nreleases versionned with a minor bump, create the wallaby branch from the\nright point, avoid errors with our tooling related to missing tags for a\nseries.\n\nThese changes also init ipa-builder for Xena.\n\nThese changes keep the independent files for the sake of a proper\nhistory (previous releases history of ipa-builder).\n\nhttp://lists.openstack.org/pipermail/openstack-discuss/2021-April/022032.html\n\nChange-Id: Ia899b39dbc68b245e1b84518363de5420af500c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4ae8d6e46b162e40fe5b1962e99053ef2d0f5ddb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4ae8d6e46b162e40fe5b1962e99053ef2d0f5ddb"}]},"branch":"refs/heads/master"},"5f91b982b4c039b87dddf271eb8958dd28c3cf3c":{"kind":"REWORK","_number":5,"created":"2021-05-07 09:38:15.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/87/789587/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/87/789587/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/789587/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/789587/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/789587/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/87/789587/5"}}},"commit":{"parents":[{"commit":"632b30df6d3bfe5dba93813f087b1e1b7532a486","subject":"Fix list_eol_stale_branches.sh to list only stale branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/632b30df6d3bfe5dba93813f087b1e1b7532a486"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-05-04 14:27:11.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-05-07 09:38:00.000000000","tz":120},"subject":"ironic-python-agent-builder release model change","message":"ironic-python-agent-builder release model change\n\nDuring the recent xena ptg, the ironic community had a discussion about\nthe need to move the ironic-python-agent-builder project from an\nindependent model to the standard release model.\n\nWhen we initially split the builder from ironic-python-agent, we decided\nagainst it, but considering some problems we encountered during the\nroad, the ironic community seems to be in favor of the change.\nThe reasons for this are mainly to strictly align the image building\nproject to ironic-python-agent releases, and ease dealing with the\noccasional upgrade of tinycore linux, the base image used to build the\n\"tinyipa\" ironic-python-agent ramdisk.\n\nWe decided to split them at some point, and we\u0027d like to keep them\nseparated, but we realized that it would make sense to keep them synced\nby branch.\n\nThe ironic contributor suffered CI pain over the\npast year where code handling distributions has broken several times.\nLargely these breaks have had to be navigated around with specific\nversion pinning, and settings changes, because we were unable to fix\nsomething a release or two back inside the tooling, in this case\nironic-python-agent-builder. In other words, if the ironic team had that\ncapability and those branches/tags, we would have been less in a world of\npain because we wouldn\u0027t have had to make changes on every job/branch\nacross every repository where we were hitting issues with ipa-b. Reduction\nof pain is why this change is being proposed.\n\nWe should also notice that the Kolla team also use ipa-builder and they\nfaced similar issues with Kayobe.\n\nThese changes propose to cut ipa-builder for Wallaby in a retro active\nmanner (Wallaby is already out at this point).\n\nThese changes use the same model that the one used by\nironic-python-agent.\n\nThese changes create a new tag (2.7.0) based on the same hash that the\nprevious independent version (2.6.0), mostly to properly init the series\nwith a meaningful version number, avoid misleading with future bugfix\nreleases versionned with a minor bump, create the wallaby branch from the\nright point, avoid errors with our tooling related to missing tags for a\nseries.\n\nThese changes also init ipa-builder for Xena.\n\nThese changes keep the independent files for the sake of a proper\nhistory (previous releases history of ipa-builder).\n\nhttp://lists.openstack.org/pipermail/openstack-discuss/2021-April/022032.html\n\nChange-Id: Ia899b39dbc68b245e1b84518363de5420af500c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5f91b982b4c039b87dddf271eb8958dd28c3cf3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5f91b982b4c039b87dddf271eb8958dd28c3cf3c"}]},"branch":"refs/heads/master"},"e96978a74ad7bac2175cd1b4069b07c6f3641133":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2021-05-07 10:54:51.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/87/789587/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/87/789587/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/789587/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/789587/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/789587/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/87/789587/6"}}},"commit":{"parents":[{"commit":"632b30df6d3bfe5dba93813f087b1e1b7532a486","subject":"Fix list_eol_stale_branches.sh to list only stale branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/632b30df6d3bfe5dba93813f087b1e1b7532a486"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-05-04 14:27:11.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-05-07 10:53:36.000000000","tz":120},"subject":"ironic-python-agent-builder release model change","message":"ironic-python-agent-builder release model change\n\nDuring the recent xena ptg, the ironic community had a discussion about\nthe need to move the ironic-python-agent-builder project from an\nindependent model to the standard release model.\n\nWhen we initially split the builder from ironic-python-agent, we decided\nagainst it, but considering some problems we encountered during the\nroad, the ironic community seems to be in favor of the change.\nThe reasons for this are mainly to strictly align the image building\nproject to ironic-python-agent releases, and ease dealing with the\noccasional upgrade of tinycore linux, the base image used to build the\n\"tinyipa\" ironic-python-agent ramdisk.\n\nWe decided to split them at some point, and we\u0027d like to keep them\nseparated, but we realized that it would make sense to keep them synced\nby branch.\n\nThe ironic contributor suffered CI pain over the\npast year where code handling distributions has broken several times.\nLargely these breaks have had to be navigated around with specific\nversion pinning, and settings changes, because we were unable to fix\nsomething a release or two back inside the tooling, in this case\nironic-python-agent-builder. In other words, if the ironic team had that\ncapability and those branches/tags, we would have been less in a world of\npain because we wouldn\u0027t have had to make changes on every job/branch\nacross every repository where we were hitting issues with ipa-b. Reduction\nof pain is why this change is being proposed.\n\nWe should also notice that the Kolla team also use ipa-builder and they\nfaced similar issues with Kayobe.\n\nThese changes propose to cut ipa-builder for Wallaby in a retro active\nmanner (Wallaby is already out at this point).\n\nThese changes use the same model that the one used by\nironic-python-agent.\n\nThese changes create a new tag (2.7.0) based on the same hash that the\nprevious independent version (2.6.0), mostly to properly init the series\nwith a meaningful version number, avoid misleading with future bugfix\nreleases versionned with a minor bump, create the wallaby branch from the\nright point, avoid errors with our tooling related to missing tags for a\nseries.\n\nThese changes also init ipa-builder for Xena.\n\nhttp://lists.openstack.org/pipermail/openstack-discuss/2021-April/022032.html\n\nChange-Id: Ia899b39dbc68b245e1b84518363de5420af500c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e96978a74ad7bac2175cd1b4069b07c6f3641133"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e96978a74ad7bac2175cd1b4069b07c6f3641133"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
