)]}'
{"id":"openstack%2Freleases~790053","triplet_id":"openstack%2Freleases~master~Id1eff47d03da09648e4200d036ff71b8b8d74e78","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"Id1eff47d03da09648e4200d036ff71b8b8d74e78","subject":"Release wallaby python-tripleoclient and tripleo-common","status":"ABANDONED","created":"2021-05-06 12:23:23.000000000","updated":"2021-05-06 14:53:33.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4efb55546da328991f5a5c53de33eb47507b969f","_number":790053,"virtual_id_number":790053,"owner":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-05-06 12:30:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"disliked":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":-1,"date":"2021-05-06 14:36:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2021-05-06 12:24:32.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":1,"date":"2021-05-06 12:24:31.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-05-06 12:24:58.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}],"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-06 12:24:27.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-05-06 12:24:27.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-05-06 12:24:31.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2021-05-06 12:24:32.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2021-05-06 12:24:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-06 13:20:10.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2021-05-06 14:33:07.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"CC"}],"messages":[{"id":"19ba9d88774f190932fea1fc18d0d40723f870b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-05-06 12:23:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f378f8c24b861b5326807ac7be3278d2b743bbb4","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-05-06 12:24:31.000000000","message":"Patch Set 1: Code-Review+1\n\nThank you!","accounts_in_message":[],"_revision_number":1},{"id":"6372cec8f0b82975317b231d0038b97545f130c7","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2021-05-06 12:24:32.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5a3e4aff6819204efda12fb9ec0f29ff9875443f","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 12:24:58.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"0917b0cbec5663d697d9680b3fd94042a1b4a6b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 12:30:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60bae2b3e12a4ab19696b818e2ec3b57 : SUCCESS in 5m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a2cf1608bbbd4d59b19c336d0b27615b : SUCCESS in 5m 15s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/1041eec7d1db4353b8a999fe4868e15d : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":1},{"id":"28170a14b46f2b25f7ec0aeea4a7a030862c591e","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-05-06 12:55:40.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM.","accounts_in_message":[],"_revision_number":1},{"id":"522c10a9c2e9956d8d87fba7a383ce413d43f540","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-05-06 14:33:07.000000000","message":"Patch Set 1:\n\nIs this going to work as those commit hashes are only in \u0027master\u0027 and not in \u0027stable/wallaby\u0027 branch of the repos?","accounts_in_message":[],"_revision_number":1},{"id":"26e013ef9dd6da382b97021fae1cdc8a84013b81","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-05-06 14:36:57.000000000","message":"Patch Set 1: Code-Review-1\n\n\u003e Patch Set 1:\n\u003e \n\u003e Is this going to work as those commit hashes are only in \u0027master\u0027 and not in \u0027stable/wallaby\u0027 branch of the repos?\n\nRabi is right, the hashes are wrong.","accounts_in_message":[],"_revision_number":1},{"id":"bbe5f4fb45f56d5dd6a48fa4c282cd44ec8da792","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-05-06 14:44:29.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e Is this going to work as those commit hashes are only in \u0027master\u0027 and not in \u0027stable/wallaby\u0027 branch of the repos?\n\nthanks Elod and Rabi - this is why i was pointing to it in the commit message.\n\nWe were trying to come up with a solution to the drift between stable/wallaby for tripleoclient and tripleo-common vs the rest of the repos. Since we branched client/common first/earlier they diverged from what we called stable/wallaby in tripleo-ansible/tripleo-heat-templates etc.\n\nSince we can\u0027t go back and tag something else as stable/wallaby we thought perhaps tagging a new release from master would be ok - but not sure if legal, for the reason you point to.","accounts_in_message":[],"_revision_number":1},{"id":"4efb55546da328991f5a5c53de33eb47507b969f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-05-06 14:53:33.000000000","message":"Abandoned\n\nsorry for wasting time, we discussed this in a call today and weren\u0027t sure if it would work. We thought to post the review to find out. thank you for your very useful reviews.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7e0a4f33022ee30d5b31d1f6a4cd3098117f5e6e","revisions":{"7e0a4f33022ee30d5b31d1f6a4cd3098117f5e6e":{"kind":"REWORK","_number":1,"created":"2021-05-06 12:23:23.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/53/790053/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/53/790053/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/790053/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/790053/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/790053/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/53/790053/1"}}},"commit":{"parents":[{"commit":"98143a9159069b3396e6949476de4af2bcfd5b10","subject":"Merge \"Release ansible-role-redhat-subscription\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/98143a9159069b3396e6949476de4af2bcfd5b10"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-05-06 12:22:16.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-05-06 12:22:16.000000000","tz":180},"subject":"Release wallaby python-tripleoclient and tripleo-common","message":"Release wallaby python-tripleoclient and tripleo-common\n\nWe released a version with [1] but did not get a new enough\nhash causing drift with other tripleo repos. This creates a new\nrelease for each from the top commit on master branch.\n\n[1] https://review.opendev.org/c/openstack/releases/+/789558\n\nChange-Id: Id1eff47d03da09648e4200d036ff71b8b8d74e78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7e0a4f33022ee30d5b31d1f6a4cd3098117f5e6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7e0a4f33022ee30d5b31d1f6a4cd3098117f5e6e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
