)]}'
{"id":"openstack%2Freleases~790763","triplet_id":"openstack%2Freleases~master~I3e1f927cb9ab5152e6e080c2713f1321f52b123f","project":"openstack/releases","branch":"master","topic":"train-em","hashtags":[],"change_id":"I3e1f927cb9ab5152e6e080c2713f1321f52b123f","subject":"[OpenStackAnsible] Transition Train to EM","status":"MERGED","created":"2021-05-11 15:14:47.000000000","updated":"2021-05-18 13:55:30.000000000","submitted":"2021-05-18 13:55:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"790763-train-em","meta_rev_id":"a1c9c0127670cfec0620a65ccb9d76ea9e32c755","_number":790763,"virtual_id_number":790763,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-18 13:55:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2021-05-17 14:58:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2021-05-18 13:44:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2021-05-18 13:44:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-05-15 04:49:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-11 15:16:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-11 15:28:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-11 16:24:59.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2021-05-18 13:44:49.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"81803ca679368fc58897d2a42d23ac28d74d50d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-05-11 15:14:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5ed8b33cdaa161356f030d72df8c151f85afa541","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 15:28:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c979a6fc701407fb55f0a4a26223b32 : SUCCESS in 7m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1fa02c43a6914691b41412d142724948 : FAILURE in 8m 35s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d70c75a197364c178832b47c60bc044b : SUCCESS in 11m 36s","accounts_in_message":[],"_revision_number":1},{"id":"cc42368da37fa15c42eade0d11710dca0f6fdf25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-05-11 15:43:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1907324a4f686d29fdde2d8d63557baf7e68b550","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 15:55:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3e224eb480b4d0cb4fb548b9a03065a : SUCCESS in 7m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a10bdf33502743c9912e570ad41d8ed0 : SUCCESS in 6m 56s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/4f5856459d1c4607ae299077c97cd641 : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":2},{"id":"95048d48d319e6366531b47e502eafddbb64e7d7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-05-12 13:51:12.000000000","message":"Patch Set 2: Code-Review-1\n\nWe need to merge https://review.opendev.org/c/openstack/openstack-ansible/+/790655 and base EM on that SHA","accounts_in_message":[],"_revision_number":2},{"id":"c09b6b7db3ccf382eb5a6b5a67d5ae89e4f09ccb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-05-14 15:20:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"140eb22056e1ea998981b388694d959359dbd6fb","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 15:23:47.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"f11e3c3c7f76bbc3bbf570294dfadd92ccb73423","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 15:32:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ec251238c03402f80d94652d3517091 : SUCCESS in 7m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/58899586b7364e79847d39b2ebe21e77 : FAILURE in 8m 45s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d758ab315bf4419891d86b9795c0ee73 : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":3},{"id":"d6a39d8c9442551d2eb2251998cf2a150f0edd31","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-05-14 15:52:25.000000000","message":"Patch Set 3: Code-Review-1\n\n@Dmitriy: train-em tag must be applied against a released version, as validator test log says [1]. Please prepare a separate release patch and then rebase this patch on top of that.\n\n[1] https://zuul.opendev.org/t/openstack/build/58899586b7364e79847d39b2ebe21e77/log/tox/validate-request-results.log#5692","accounts_in_message":[],"_revision_number":3},{"id":"78356fcda63bb018ebebfc3555c56bb6d2ce08cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-05-15 04:48:44.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"70eb5afe3a9bb5608162f87650f0808401ca31c7","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-15 04:49:44.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"38ef5eec44b93e8463b9b1fc989edb98768be92b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-05-15 04:53:15.000000000","message":"Patch Set 4:\n\n\u003e Patch Set 3: Code-Review-1\n\u003e \n\u003e @Dmitriy: train-em tag must be applied against a released version, as validator test log says [1]. Please prepare a separate release patch and then rebase this patch on top of that.\n\u003e \n\u003e [1] https://zuul.opendev.org/t/openstack/build/58899586b7364e79847d39b2ebe21e77/log/tox/validate-request-results.log#5692\n\nWell, yeah, fair. But a bit unfortunate actually. As we use \"bumps\" of service SHAs for releases - we update used versions of services and playbooks. And for EM that would make sense to switch just to table/train, so that users would get all emergency backports for EM tag that could be backported after EM.\nBut, it doesn\u0027t make sense to create a regular tag for this, as it might be super confusing for users, as they will expect it to be a normal \"stable\" release, rather then getting all latest stuff.","accounts_in_message":[],"_revision_number":4},{"id":"73f467f4c0d7f944b73409bebe17c7d40179ad5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-15 05:00:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/326365b266c04cdd878567e17a867887 : SUCCESS in 5m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/704e2a7ca884452bac260be8cbcf3e94 : FAILURE in 9m 48s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c335e63babbb43739b6a0bd1abcdcf80 : SUCCESS in 8m 08s","accounts_in_message":[],"_revision_number":4},{"id":"6b9971f6a863d492fad846ed2e00337b9076e22f","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-05-17 14:47:07.000000000","message":"Patch Set 4:\n\n\u003e Patch Set 4:\n\u003e \n\u003e \u003e Patch Set 3: Code-Review-1\n\u003e \u003e \n\u003e \u003e @Dmitriy: train-em tag must be applied against a released version, as validator test log says [1]. Please prepare a separate release patch and then rebase this patch on top of that.\n\u003e \u003e \n\u003e \u003e [1] https://zuul.opendev.org/t/openstack/build/58899586b7364e79847d39b2ebe21e77/log/tox/validate-request-results.log#5692\n\u003e \n\u003e Well, yeah, fair. But a bit unfortunate actually. As we use \"bumps\" of service SHAs for releases - we update used versions of services and playbooks. And for EM that would make sense to switch just to table/train, so that users would get all emergency backports for EM tag that could be backported after EM.\n\u003e But, it doesn\u0027t make sense to create a regular tag for this, as it might be super confusing for users, as they will expect it to be a normal \"stable\" release, rather then getting all latest stuff.\n\nI think the option to leave the last release as the final train release could be a solution in such cases, where it does not make sense to release the latest merged patches. But I might misunderstood your point.","accounts_in_message":[],"_revision_number":4},{"id":"054ff200b92e85d6c93b2f1b8c2af30a0739fb43","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-05-17 14:47:32.000000000","message":"Patch Set 4:\n\nrecheck - parent has merged, zuul should be happy now","accounts_in_message":[],"_revision_number":4},{"id":"cd22440040dddbbfa7c98e345fae427955245b96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 14:55:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/166c992574ee4531831bc88d68f95f68 : SUCCESS in 6m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/979dcddfba7a4afa94ee1220c2d4f762 : SUCCESS in 4m 58s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/5daf74c758dc4b90b84cac1ae82f0440 : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":4},{"id":"df15d90041eaf0cbc009bf7e2a76eeb702ba2148","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-05-17 14:58:39.000000000","message":"Patch Set 4: Code-Review+2\n\nThanks for the update Dmitriy, LGTM.","accounts_in_message":[],"_revision_number":4},{"id":"8aa8fedd45552babdddb41ad7c437dcd839c6e68","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-05-18 13:44:49.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"afa690ef9098f256cb879c681e61edddfc9df171","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 13:45:08.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1d4e20623c55987af4a622d3539913bda33466eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 13:55:25.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f0f8d27290c4d889c8850028a1ac521 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23d12dcfe7224b0fa2b670f927c3d444 : SUCCESS in 7m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cdc2266570de4a7f9ee6b29a3a0c4825 : SUCCESS in 7m 17s","accounts_in_message":[],"_revision_number":4},{"id":"a1c9c0127670cfec0620a65ccb9d76ea9e32c755","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 13:55:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f2f6bfbc5575e982f13e630b3afaa982d13b0680","revisions":{"03d540af006e74c1621282848a3910170a7182d1":{"kind":"REWORK","_number":1,"created":"2021-05-11 15:14:47.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/63/790763/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/63/790763/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/63/790763/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/63/790763/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/63/790763/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/63/790763/1"}}},"commit":{"parents":[{"commit":"4852e0a77a512f60179aee0e4aa992e860214668","subject":"Merge \"ironic-python-agent-builder release model change\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4852e0a77a512f60179aee0e4aa992e860214668"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-05-11 15:14:05.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-05-11 15:14:05.000000000","tz":120},"subject":"[OpenStackAnsible] Transition Train to EM","message":"[OpenStackAnsible] Transition Train to EM\n\nThis transition the train branch to extended maintenance.\nChanges for bugfixes and things the team deems important are\nstill encouraged, but there will no longer be official releases\noff of the branch.\n\nPlease +1 if the team is ready for us to proceed with this\ntransition, or -1 if there are any final backports currently in\nflight that we should wait for. For the latter case, please\nupdate the patch with the new commit hash after doing a final\nrelease to get those changes out so we know to proceed with the\ntransition.\n\nChange-Id: I3e1f927cb9ab5152e6e080c2713f1321f52b123f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/03d540af006e74c1621282848a3910170a7182d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/03d540af006e74c1621282848a3910170a7182d1"}]},"branch":"refs/heads/master"},"5d02f409d2669aa740acaf740be3f4384ce55af1":{"kind":"REWORK","_number":2,"created":"2021-05-11 15:43:57.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/63/790763/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/63/790763/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/63/790763/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/63/790763/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/63/790763/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/63/790763/2"}}},"commit":{"parents":[{"commit":"4852e0a77a512f60179aee0e4aa992e860214668","subject":"Merge \"ironic-python-agent-builder release model change\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4852e0a77a512f60179aee0e4aa992e860214668"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-05-11 15:14:05.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-05-11 15:43:09.000000000","tz":120},"subject":"[OpenStackAnsible] Transition Train to EM","message":"[OpenStackAnsible] Transition Train to EM\n\nThis transition the train branch to extended maintenance.\nChanges for bugfixes and things the team deems important are\nstill encouraged, but there will no longer be official releases\noff of the branch.\n\nPlease +1 if the team is ready for us to proceed with this\ntransition, or -1 if there are any final backports currently in\nflight that we should wait for. For the latter case, please\nupdate the patch with the new commit hash after doing a final\nrelease to get those changes out so we know to proceed with the\ntransition.\n\nChange-Id: I3e1f927cb9ab5152e6e080c2713f1321f52b123f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5d02f409d2669aa740acaf740be3f4384ce55af1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5d02f409d2669aa740acaf740be3f4384ce55af1"}]},"branch":"refs/heads/master"},"4e32052ede96472ed2b56a8402abf74fb595682f":{"kind":"REWORK","_number":3,"created":"2021-05-14 15:20:28.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/63/790763/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/63/790763/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/63/790763/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/63/790763/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/63/790763/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/63/790763/3"}}},"commit":{"parents":[{"commit":"c01cd02976130ff6ff3005c670ecfb9d295ed8ec","subject":"Release OpenStack-Ansible Train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c01cd02976130ff6ff3005c670ecfb9d295ed8ec"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-05-11 15:14:05.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-05-14 15:20:13.000000000","tz":180},"subject":"[OpenStackAnsible] Transition Train to EM","message":"[OpenStackAnsible] Transition Train to EM\n\nThis transition the train branch to extended maintenance.\nChanges for bugfixes and things the team deems important are\nstill encouraged, but there will no longer be official releases\noff of the branch.\n\nPlease +1 if the team is ready for us to proceed with this\ntransition, or -1 if there are any final backports currently in\nflight that we should wait for. For the latter case, please\nupdate the patch with the new commit hash after doing a final\nrelease to get those changes out so we know to proceed with the\ntransition.\n\nChange-Id: I3e1f927cb9ab5152e6e080c2713f1321f52b123f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4e32052ede96472ed2b56a8402abf74fb595682f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4e32052ede96472ed2b56a8402abf74fb595682f"}]},"branch":"refs/heads/master"},"f2f6bfbc5575e982f13e630b3afaa982d13b0680":{"kind":"REWORK","_number":4,"created":"2021-05-15 04:48:44.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/63/790763/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/63/790763/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/63/790763/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/63/790763/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/63/790763/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/63/790763/4"}}},"commit":{"parents":[{"commit":"c01cd02976130ff6ff3005c670ecfb9d295ed8ec","subject":"Release OpenStack-Ansible Train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c01cd02976130ff6ff3005c670ecfb9d295ed8ec"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-05-11 15:14:05.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-05-15 04:48:39.000000000","tz":0},"subject":"[OpenStackAnsible] Transition Train to EM","message":"[OpenStackAnsible] Transition Train to EM\n\nThis transition the train branch to extended maintenance.\nChanges for bugfixes and things the team deems important are\nstill encouraged, but there will no longer be official releases\noff of the branch.\n\nPlease +1 if the team is ready for us to proceed with this\ntransition, or -1 if there are any final backports currently in\nflight that we should wait for. For the latter case, please\nupdate the patch with the new commit hash after doing a final\nrelease to get those changes out so we know to proceed with the\ntransition.\n\nChange-Id: I3e1f927cb9ab5152e6e080c2713f1321f52b123f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f2f6bfbc5575e982f13e630b3afaa982d13b0680"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f2f6bfbc5575e982f13e630b3afaa982d13b0680"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
