)]}'
{"id":"openstack%2Freleases~790767","triplet_id":"openstack%2Freleases~master~Ib0ea7e3976c48fc5d5334b3b0a7b37a3dae23d66","project":"openstack/releases","branch":"master","topic":"train-em","hashtags":[],"change_id":"Ib0ea7e3976c48fc5d5334b3b0a7b37a3dae23d66","subject":"[Puppet OpenStack] Transition Train to EM","status":"MERGED","created":"2021-05-11 15:19:56.000000000","updated":"2021-07-07 15:04:54.000000000","submitted":"2021-07-07 15:04:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"790767-train-em","meta_rev_id":"c43609d934fd0bcdc832673ab2ef33754e7a8bb7","_number":790767,"virtual_id_number":790767,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-07 15:04:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2021-07-07 14:53:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2021-07-06 13:35:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-06-29 08:10:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2021-07-07 14:53:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-11 15:21:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-11 15:22:25.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2021-05-11 15:35:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-12 11:24:27.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-06-16 18:51:51.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2021-06-29 08:10:49.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-07-03 07:10:56.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"4c242a2883e2b168d512fc4b42308325fcd42b90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-05-11 15:19:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9d41434518e4c7e77a61698e09d197e424ccd614","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 15:35:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1108860c0701454ea549fea841234404 : SUCCESS in 6m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1abd1a516fe847afbad004106084259a : FAILURE in 6m 03s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/32cfc5fbd1b5402fb522ef20e3248aa4 : SUCCESS in 9m 19s","accounts_in_message":[],"_revision_number":1},{"id":"540b316cf96da0b07533f2a17e9ea4bef135ee6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-05-11 15:55:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"239c5cdc7c27066ca1e39852f8efe41616011b11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 16:06:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b1077389b9a427a8468c91002844289 : SUCCESS in 7m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/53a716e5a7b04d478e39f7df9f1f28e0 : SUCCESS in 6m 20s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d3a76768cf284f18bb41a872337decae : SUCCESS in 9m 46s","accounts_in_message":[],"_revision_number":2},{"id":"4acaf25b2b1d26b8db786133298025f52fef9a17","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-12 11:24:27.000000000","message":"Patch Set 2: Code-Review-1\n\nI\u0027ll check each repos and create new releases if we have any changes since the last release. This would be important especially for some distros like RDO which is relying on the released versions.","accounts_in_message":[],"_revision_number":2},{"id":"e3621cbdc3210898f6db17ec620be4a34869cba5","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-05-26 18:55:44.000000000","message":"Patch Set 2:\n\nHi @Takashi, any update on this?","accounts_in_message":[],"_revision_number":2},{"id":"d6dacd3f1225bab118342ed0229e3add9e59b8cb","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-26 23:35:39.000000000","message":"Patch Set 2:\n\nSorry about delay. I was waiting for several reviews to make the last stable/train release.\nI proposed a change to create the final releases. Once it is merged then we can move all repos to EM based on that.\n https://review.opendev.org/c/openstack/releases/+/792820","accounts_in_message":[],"_revision_number":2},{"id":"bd63a732882f6494b79c37e705b80403e96c05fd","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-16 18:51:51.000000000","message":"Patch Set 2:\n\nShould be rebased over the latest train releases","accounts_in_message":[],"_revision_number":2},{"id":"21aef4e98a896a752ef31ac6916964f262925ab8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-06-17 18:37:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"82e7ad6f9326f7145df1d722d593fb17a62b18f5","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-06-17 18:41:19.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nI\u0027ve rebased the patch, now it should be up to date.","accounts_in_message":[],"_revision_number":3},{"id":"05dbdc9126ae0b93f47b908013fe761bb28932a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 18:51:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e92e576c76194da1bd3bb386c220d875 : SUCCESS in 6m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/65e3a916ef834e18a9b2cf925417ce88 : SUCCESS in 7m 44s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/130bc0edc6904401971c5731a29eaf7b : SUCCESS in 12m 34s","accounts_in_message":[],"_revision_number":3},{"id":"86012ccdf0fffd3a8fc36d7f666e3a5912194326","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-18 00:01:01.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f59de51ded9d5dd1e5926adcb535c112c60767d6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-18 00:01:24.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"244bdc1b1f03a7c5de75ff39e4c852191d81a9d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 00:11:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc6b879fd917442ab3862f3bce06fb1d : SUCCESS in 6m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/691cd47a70c24bc3909cec37366f712d : FAILURE in 6m 44s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/781cd2fb3beb4071903f7186b38407d5 : SUCCESS in 9m 28s","accounts_in_message":[],"_revision_number":4},{"id":"a1bae2d2ba94dbd5956a52e5f43591ea3f0955d7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-18 15:02:13.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"923f4803a8aa748e8b4f7c0bb0a241dffa8108dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 15:16:14.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fbf5f3c28e74ea1820ebbab872172d0 : SUCCESS in 7m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/997fda341b2c44e7b7dbec236ec3b4f6 : FAILURE in 6m 39s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/124081f0b0884795819c5e125839f0ec : SUCCESS in 9m 44s","accounts_in_message":[],"_revision_number":4},{"id":"50b0e9f1870aa56b1e6d94eebb131d6843f63d18","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-21 12:11:54.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"7ab1bc44336de6904223e62a39026500836dabd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 12:23:25.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d302822e6b24948b9c41abf6eb4961b : SUCCESS in 7m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/066c5c26a1ba4d78955d258fd6a22f8a : FAILURE in 7m 02s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b424706d2773485f81181d681751d2bc : SUCCESS in 9m 32s","accounts_in_message":[],"_revision_number":4},{"id":"498d965f3084f10fd41a5bff258459fcf0fcac2a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-06-21 13:13:05.000000000","message":"Patch Set 4:\n\ntests should pass as soon as parent patch merges.","accounts_in_message":[],"_revision_number":4},{"id":"c0d685b2af452c628e017301ffd8f3d303823ab0","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-06-21 15:08:07.000000000","message":"Patch Set 4:\n\nrecheck - parent patch has merged","accounts_in_message":[],"_revision_number":4},{"id":"1ad211c6bce68c28d366f845f31a2848436d9ebc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 15:22:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b282e3a630a4019808b4d1f0ec82128 : SUCCESS in 6m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/deb5a99162e243529ed3060a7935f2b7 : SUCCESS in 6m 46s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/302c2a3ced944c77aaddff3a984aaacc : SUCCESS in 9m 12s","accounts_in_message":[],"_revision_number":4},{"id":"dec4bbf856ce3db5659dde270aa0ed53f2ee1057","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-29 08:10:49.000000000","message":"Patch Set 4: Code-Review+2\n\nI think that we are good to go.","accounts_in_message":[],"_revision_number":4},{"id":"fc49250063eede5b15671000dc1f0980866a92c7","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-07-02 12:57:33.000000000","message":"Patch Set 4:\n\n@Takasi @zhongshengping can we merge this?","accounts_in_message":[],"_revision_number":4},{"id":"95af0f94554465b80d8c4a461acd018c9949ba81","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-07-06 13:05:37.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"de63ac85d340899dd5a4b93553eb025f5a8aef0f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-07-06 13:06:50.000000000","message":"Patch Set 4: Code-Review-1\n\nactually we probably should check with tripleo if we want this right now.  I don\u0027t think tripleo is planning to EM train at this time so this would be problematic.","accounts_in_message":[],"_revision_number":4},{"id":"fe60db4b44558f05548a180e606a71f2465e7ed3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-06 13:19:42.000000000","message":"Patch Set 4:\n\n\u003e Patch Set 4: Code-Review-1\n\u003e \n\u003e actually we probably should check with tripleo if we want this right now.  I don\u0027t think tripleo is planning to EM train at this time so this would be problematic.\n\nstable/train of TripleO has entered to EM status by [1].\nTheoretically any change required by the final tripleo train release should be included in the final train release of puppet repos and I don\u0027t see any problem with making strable/train of puppet repos EM now.\n\n[1] https://review.opendev.org/c/openstack/releases/+/790778","accounts_in_message":[],"_revision_number":4},{"id":"8674aedf1fe7302bb530ce97a38223c330df5094","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-07-06 13:35:37.000000000","message":"Patch Set 4: Code-Review+1\n\noh i missed that. I wouldn\u0027t have transitioned tripleo at this time. oh well","accounts_in_message":[],"_revision_number":4},{"id":"fe1c2fec49cefdaf5be174ed3555b20a1eb74214","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-07-07 14:53:06.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nThanks for the approval Alex (Alex is release liaison for Puppet OpenStack). Let\u0027s merge this!","accounts_in_message":[],"_revision_number":4},{"id":"3ab97e12b10182d8ffe543acb0f65601f21c9588","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 14:53:23.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"355d42c302b8d02a3d6b409cbad4f6e2dba314ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 15:04:53.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaecf24b1c634061a50baed43ef3a1ef : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d31f3f2de914ad7be1a432b109db2f8 : SUCCESS in 6m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d62c8108510e4c60ac07bd474717fd14 : SUCCESS in 6m 24s","accounts_in_message":[],"_revision_number":4},{"id":"c43609d934fd0bcdc832673ab2ef33754e7a8bb7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 15:04:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"9ad46a4e2ed40c1b726405abdf3c2c97683e1296","revisions":{"c2509d9cb8066267896b8a20bb3aba2fafd58222":{"kind":"REWORK","_number":1,"created":"2021-05-11 15:19:56.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/67/790767/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/67/790767/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/790767/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/790767/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/790767/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/67/790767/1"}}},"commit":{"parents":[{"commit":"4852e0a77a512f60179aee0e4aa992e860214668","subject":"Merge \"ironic-python-agent-builder release model change\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4852e0a77a512f60179aee0e4aa992e860214668"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-05-11 15:19:17.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-05-11 15:19:17.000000000","tz":120},"subject":"[Puppet OpenStack] Transition Train to EM","message":"[Puppet OpenStack] Transition Train to EM\n\nThis transition the train branch to extended maintenance.\nChanges for bugfixes and things the team deems important are\nstill encouraged, but there will no longer be official releases\noff of the branch.\n\nPlease +1 if the team is ready for us to proceed with this\ntransition, or -1 if there are any final backports currently in\nflight that we should wait for. For the latter case, please\nupdate the patch with the new commit hash after doing a final\nrelease to get those changes out so we know to proceed with the\ntransition.\n\nChange-Id: Ib0ea7e3976c48fc5d5334b3b0a7b37a3dae23d66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c2509d9cb8066267896b8a20bb3aba2fafd58222"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c2509d9cb8066267896b8a20bb3aba2fafd58222"}]},"branch":"refs/heads/master"},"fa6521b696ff6147a035b6aec39af3727cd824c1":{"kind":"REWORK","_number":2,"created":"2021-05-11 15:55:18.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/67/790767/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/67/790767/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/790767/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/790767/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/790767/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/67/790767/2"}}},"commit":{"parents":[{"commit":"4852e0a77a512f60179aee0e4aa992e860214668","subject":"Merge \"ironic-python-agent-builder release model change\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4852e0a77a512f60179aee0e4aa992e860214668"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-05-11 15:19:17.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-05-11 15:55:00.000000000","tz":120},"subject":"[Puppet OpenStack] Transition Train to EM","message":"[Puppet OpenStack] Transition Train to EM\n\nThis transition the train branch to extended maintenance.\nChanges for bugfixes and things the team deems important are\nstill encouraged, but there will no longer be official releases\noff of the branch.\n\nPlease +1 if the team is ready for us to proceed with this\ntransition, or -1 if there are any final backports currently in\nflight that we should wait for. For the latter case, please\nupdate the patch with the new commit hash after doing a final\nrelease to get those changes out so we know to proceed with the\ntransition.\n\nChange-Id: Ib0ea7e3976c48fc5d5334b3b0a7b37a3dae23d66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fa6521b696ff6147a035b6aec39af3727cd824c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fa6521b696ff6147a035b6aec39af3727cd824c1"}]},"branch":"refs/heads/master"},"812112202d199d8436dffc79dc0cacfa3dec4775":{"kind":"REWORK","_number":3,"created":"2021-06-17 18:37:24.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/67/790767/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/67/790767/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/790767/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/790767/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/790767/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/67/790767/3"}}},"commit":{"parents":[{"commit":"b42430299331b734ec1c5327b2b4221742e2cfce","subject":"Merge \"[Telemetry] Transition Ocata to End of Life\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b42430299331b734ec1c5327b2b4221742e2cfce"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-05-11 15:19:17.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-06-17 18:36:58.000000000","tz":120},"subject":"[Puppet OpenStack] Transition Train to EM","message":"[Puppet OpenStack] Transition Train to EM\n\nThis transition the train branch to extended maintenance.\nChanges for bugfixes and things the team deems important are\nstill encouraged, but there will no longer be official releases\noff of the branch.\n\nPlease +1 if the team is ready for us to proceed with this\ntransition, or -1 if there are any final backports currently in\nflight that we should wait for. For the latter case, please\nupdate the patch with the new commit hash after doing a final\nrelease to get those changes out so we know to proceed with the\ntransition.\n\nChange-Id: Ib0ea7e3976c48fc5d5334b3b0a7b37a3dae23d66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/812112202d199d8436dffc79dc0cacfa3dec4775"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/812112202d199d8436dffc79dc0cacfa3dec4775"}]},"branch":"refs/heads/master"},"9ad46a4e2ed40c1b726405abdf3c2c97683e1296":{"kind":"REWORK","_number":4,"created":"2021-06-18 00:01:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/790767/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/67/790767/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/790767/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/790767/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/790767/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/67/790767/4"}}},"commit":{"parents":[{"commit":"6639de7b8fb98f782b9475cadd3d744db26c645c","subject":"[Puppet OpenStack] Fix some bugs in the last train release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6639de7b8fb98f782b9475cadd3d744db26c645c"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-05-11 15:19:17.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-18 00:00:32.000000000","tz":540},"subject":"[Puppet OpenStack] Transition Train to EM","message":"[Puppet OpenStack] Transition Train to EM\n\nThis transition the train branch to extended maintenance.\nChanges for bugfixes and things the team deems important are\nstill encouraged, but there will no longer be official releases\noff of the branch.\n\nPlease +1 if the team is ready for us to proceed with this\ntransition, or -1 if there are any final backports currently in\nflight that we should wait for. For the latter case, please\nupdate the patch with the new commit hash after doing a final\nrelease to get those changes out so we know to proceed with the\ntransition.\n\nChange-Id: Ib0ea7e3976c48fc5d5334b3b0a7b37a3dae23d66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9ad46a4e2ed40c1b726405abdf3c2c97683e1296"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9ad46a4e2ed40c1b726405abdf3c2c97683e1296"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
