)]}'
{"id":"openstack%2Freleases~790778","triplet_id":"openstack%2Freleases~master~Ief89d3034de2bee7a5ec781d91cf038615fbe9c2","project":"openstack/releases","branch":"master","topic":"train-em","hashtags":[],"change_id":"Ief89d3034de2bee7a5ec781d91cf038615fbe9c2","subject":"[tripleo] Transition Train to EM","status":"MERGED","created":"2021-05-11 15:27:46.000000000","updated":"2021-06-04 19:31:29.000000000","submitted":"2021-06-04 19:31:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"790778-train-em","meta_rev_id":"e01344aacdd4ab4c5b7349d95d06854c1cef9913","_number":790778,"virtual_id_number":790778,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-04 19:31:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"recommended":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":1,"date":"2021-06-04 15:15:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":2,"date":"2021-06-04 19:17:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2021-06-04 18:44:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":1,"date":"2021-06-04 19:17:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-06-03 12:55:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"}],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-11 15:29:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-11 15:46:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-11 16:24:33.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2021-05-11 16:24:34.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2021-06-02 11:40:27.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2021-06-04 18:44:41.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"74103a1d41b650cf18990068b358dfed0ca00b56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-05-11 15:27:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4c7abb84655b523d14d208ac4b7db0f85eebc5a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 15:46:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79720776ee0e498da9e8558dfc08a047 : SUCCESS in 6m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/408e5b7ea338459081bdcfe90dabd987 : FAILURE in 6m 23s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f34660b92732493a9b270f16da3b43d1 : SUCCESS in 7m 58s","accounts_in_message":[],"_revision_number":1},{"id":"f11bc2320024c2f993557fd067091dbe1f3ca7b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-05-11 15:57:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5fbe59e197d49c0a516528eef9a4a7946bbfe421","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 16:07:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60c082b0f7df44df9e077df26a2930e6 : SUCCESS in 8m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d40219f96d4d423780cfdac6df6d5f3d : SUCCESS in 7m 24s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/0e22cd162f3041c285b16144d68884d6 : SUCCESS in 8m 34s","accounts_in_message":[],"_revision_number":2},{"id":"e981f749aeca64ea971f4e697dd16ba5100ca4a4","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-05-13 07:08:53.000000000","message":"Patch Set 2: Code-Review-1\n\nthanks, I am setting -1 for now because I want to send a note about that to the mailing list and generally socialise it a bit/bring to our next irc meeting etc to allow folks to post things they want in before we go em\n\nAfter this I will update the patch with the latest\n\nthank you","accounts_in_message":[],"_revision_number":2},{"id":"030932d06bccbbb4f57e02453d179d63ebb7ace3","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-05-13 13:17:36.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2: Code-Review-1\n\u003e \n\u003e thanks, I am setting -1 for now because I want to send a note about that to the mailing list and generally socialise it a bit/bring to our next irc meeting etc to allow folks to post things they want in before we go em\n\u003e \n\u003e After this I will update the patch with the latest\n\u003e \n\u003e thank you\n\n        * http://lists.openstack.org/pipermail/openstack-discuss/2021-May/022428.html\n        * [TripleO] tripleo repos going Extended Maintenance stable/train OK? (not yet IMO)\nMarios Andreou marios at redhat.com\nThu May 13 12:47:04 UTC 2021","accounts_in_message":[],"_revision_number":2},{"id":"b11c15e6f2552be677d047fe6b68d3aacd0b2574","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-02 11:40:27.000000000","message":"Patch Set 2:\n\nHello Marios,\n\nDo we need updates on this patch?\n\nWe plan to finalize the EM transition on Friday.","accounts_in_message":[],"_revision_number":2},{"id":"85ab2cd76caf170af7bbba58c7f1bc8b57330ca3","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-06-02 15:35:39.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e Hello Marios,\n\u003e \n\u003e Do we need updates on this patch?\n\u003e \n\u003e We plan to finalize the EM transition on Friday.\n\n\nHi Herve thanks for reaching out about that.\n\nDoes it mean we don\u0027t really have a choice, i.e. we *have* to transition these repos to extended maintenance this friday? For tripleo the train branch is kinda important as it is one of our fast-forward upgrade branches (queens-\u003etrain). \n\nAssuming we have a choice, keeping it as active would be more technically correct for us as the branch *is* active (e.g. see tripleo-heat-templates activity since the latest tag at https://github.com/openstack/tripleo-heat-templates/compare/11.5.0...stable/train). \n\nHowever, if we have to go extended maintenance then we can also do that, it should not cause any major problems for us in the sense that we can/will just keep merging things to the train branch anyway.\n\nIn the case that we are going to EM this friday it would probably be more correct to make a train release before that and then update this review with the latest hashes as outlined in the commit message here. I can do that tomorrow if needed,\n\nthank you for your help\n\nmarios","accounts_in_message":[],"_revision_number":2},{"id":"1a0acbc77aed564e42acfb303101dd54b669b83f","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-06-02 16:05:02.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e \u003e Patch Set 2:\n\u003e \u003e \n\u003e \u003e Hello Marios,\n\u003e \u003e \n\u003e \u003e Do we need updates on this patch?\n\u003e \u003e \n\u003e \u003e We plan to finalize the EM transition on Friday.\n\u003e \n\u003e \n\u003e Hi Herve thanks for reaching out about that.\n\u003e \n\u003e Does it mean we don\u0027t really have a choice, i.e. we *have* to transition these repos to extended maintenance this friday? For tripleo the train branch is kinda important as it is one of our fast-forward upgrade branches (queens-\u003etrain). \n\u003e \n\u003e Assuming we have a choice, keeping it as active would be more technically correct for us as the branch *is* active (e.g. see tripleo-heat-templates activity since the latest tag at https://github.com/openstack/tripleo-heat-templates/compare/11.5.0...stable/train). \n\u003e \n\u003e However, if we have to go extended maintenance then we can also do that, it should not cause any major problems for us in the sense that we can/will just keep merging things to the train branch anyway.\n\u003e \n\u003e In the case that we are going to EM this friday it would probably be more correct to make a train release before that and then update this review with the latest hashes as outlined in the commit message here. I can do that tomorrow if needed,\n\u003e \n\u003e thank you for your help\n\u003e \n\u003e marios\n\nHi Marios,\n\nconsidering a couple of things:\n- Tripleo is not following stable-policy, I understand that, but Extended Maintenance is beyond that\n- Extended Maintenance means that a stable release (in this case Train) reached its end of Maintenance phase and the teams + release team won\u0027t support any further releases in general!\n- I appreciate that stable/train of TripleO is so heavily active even now that we reached the Extended Maintenance transition deadline, but\n- the branch remains open for bugfix backports in Extended Maintenance (according to Extended Maintenance concept), so that should not mean any problem for the team\n- the bugfix backports that will be merged in TripleO\u0027s stable/train should be consumed downstream via downstream releases\n\nSo in short: the transition is kind of mandatory, but that shouldn\u0027t hinder the team to be active in TripleO\u0027s stable/train branch :)","accounts_in_message":[],"_revision_number":2},{"id":"6578278d7dbc4fefa5efa8eda08a07a2fc98e039","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-03 09:01:23.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e \u003e Patch Set 2:\n\u003e \u003e \n\u003e \u003e \u003e Patch Set 2:\n\u003e \u003e \u003e \n\u003e \u003e \u003e Hello Marios,\n\u003e \u003e \u003e \n\u003e \u003e \u003e Do we need updates on this patch?\n\u003e \u003e \u003e \n\u003e \u003e \u003e We plan to finalize the EM transition on Friday.\n\u003e \u003e \n\u003e \u003e \n\u003e \u003e Hi Herve thanks for reaching out about that.\n\u003e \u003e \n\u003e \u003e Does it mean we don\u0027t really have a choice, i.e. we *have* to transition these repos to extended maintenance this friday? For tripleo the train branch is kinda important as it is one of our fast-forward upgrade branches (queens-\u003etrain). \n\u003e \u003e \n\u003e \u003e Assuming we have a choice, keeping it as active would be more technically correct for us as the branch *is* active (e.g. see tripleo-heat-templates activity since the latest tag at https://github.com/openstack/tripleo-heat-templates/compare/11.5.0...stable/train). \n\u003e \u003e \n\u003e \u003e However, if we have to go extended maintenance then we can also do that, it should not cause any major problems for us in the sense that we can/will just keep merging things to the train branch anyway.\n\u003e \u003e \n\u003e \u003e In the case that we are going to EM this friday it would probably be more correct to make a train release before that and then update this review with the latest hashes as outlined in the commit message here. I can do that tomorrow if needed,\n\u003e \u003e \n\u003e \u003e thank you for your help\n\u003e \u003e \n\u003e \u003e marios\n\u003e \n\u003e Hi Marios,\n\u003e \n\u003e considering a couple of things:\n\u003e - Tripleo is not following stable-policy, I understand that, but Extended Maintenance is beyond that\n\u003e - Extended Maintenance means that a stable release (in this case Train) reached its end of Maintenance phase and the teams + release team won\u0027t support any further releases in general!\n\u003e - I appreciate that stable/train of TripleO is so heavily active even now that we reached the Extended Maintenance transition deadline, but\n\u003e - the branch remains open for bugfix backports in Extended Maintenance (according to Extended Maintenance concept), so that should not mean any problem for the team\n\u003e - the bugfix backports that will be merged in TripleO\u0027s stable/train should be consumed downstream via downstream releases\n\u003e \n\u003e So in short: the transition is kind of mandatory, but that shouldn\u0027t hinder the team to be active in TripleO\u0027s stable/train branch :)\n\nThanks Elod for your well detailed response.\n\nIndeed the transition is mandatory, all stable branches reach EM after 18 months, however these stable branches remain open and maintainable, so, backporting patches and fixes is still possible.\n\nEM simply mean that he release team stop producing release for these branches.\n\nFrom a tripleo point of view, if I\u0027m right, you have an entire machinery and mechanismes that pulling and backport patches downstream, so I don\u0027t think that it will be an issue for the projects under your umbrella.\n\nIf delta exist between proposed hashes and the latest available hashes not yet release, I agree with you, it would be worth releasing them before transitioning.","accounts_in_message":[],"_revision_number":2},{"id":"e8ee1f6febb4780ccbb703bf378bcfc08776a49a","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-06-03 11:50:16.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nOK thank you Elod and Herve for your replies, I am going to push a last train release then rebase this ontop of it with updated hashes.\n\nthank you!\n\n\n\u003e Patch Set 2:\n\u003e \n\u003e \u003e Patch Set 2:\n\u003e \u003e \n\u003e \u003e \u003e Patch Set 2:\n\u003e \u003e \u003e \n\u003e \u003e \u003e \u003e Patch Set 2:\n\u003e \u003e \u003e \u003e \n\u003e \u003e \u003e \u003e Hello Marios,\n\u003e \u003e \u003e \u003e \n\u003e \u003e \u003e \u003e Do we need updates on this patch?\n\u003e \u003e \u003e \u003e \n\u003e \u003e \u003e \u003e We plan to finalize the EM transition on Friday.\n\u003e \u003e \u003e \n\u003e \u003e \u003e \n\u003e \u003e \u003e Hi Herve thanks for reaching out about that.\n\u003e \u003e \u003e \n\u003e \u003e \u003e Does it mean we don\u0027t really have a choice, i.e. we *have* to transition these repos to extended maintenance this friday? For tripleo the train branch is kinda important as it is one of our fast-forward upgrade branches (queens-\u003etrain). \n\u003e \u003e \u003e \n\u003e \u003e \u003e Assuming we have a choice, keeping it as active would be more technically correct for us as the branch *is* active (e.g. see tripleo-heat-templates activity since the latest tag at https://github.com/openstack/tripleo-heat-templates/compare/11.5.0...stable/train). \n\u003e \u003e \u003e \n\u003e \u003e \u003e However, if we have to go extended maintenance then we can also do that, it should not cause any major problems for us in the sense that we can/will just keep merging things to the train branch anyway.\n\u003e \u003e \u003e \n\u003e \u003e \u003e In the case that we are going to EM this friday it would probably be more correct to make a train release before that and then update this review with the latest hashes as outlined in the commit message here. I can do that tomorrow if needed,\n\u003e \u003e \u003e \n\u003e \u003e \u003e thank you for your help\n\u003e \u003e \u003e \n\u003e \u003e \u003e marios\n\u003e \u003e \n\u003e \u003e Hi Marios,\n\u003e \u003e \n\u003e \u003e considering a couple of things:\n\u003e \u003e - Tripleo is not following stable-policy, I understand that, but Extended Maintenance is beyond that\n\u003e \u003e - Extended Maintenance means that a stable release (in this case Train) reached its end of Maintenance phase and the teams + release team won\u0027t support any further releases in general!\n\u003e \u003e - I appreciate that stable/train of TripleO is so heavily active even now that we reached the Extended Maintenance transition deadline, but\n\u003e \u003e - the branch remains open for bugfix backports in Extended Maintenance (according to Extended Maintenance concept), so that should not mean any problem for the team\n\u003e \u003e - the bugfix backports that will be merged in TripleO\u0027s stable/train should be consumed downstream via downstream releases\n\u003e \u003e \n\u003e \u003e So in short: the transition is kind of mandatory, but that shouldn\u0027t hinder the team to be active in TripleO\u0027s stable/train branch :)\n\u003e \n\u003e Thanks Elod for your well detailed response.\n\u003e \n\u003e Indeed the transition is mandatory, all stable branches reach EM after 18 months, however these stable branches remain open and maintainable, so, backporting patches and fixes is still possible.\n\u003e \n\u003e EM simply mean that he release team stop producing release for these branches.\n\u003e \n\u003e From a tripleo point of view, if I\u0027m right, you have an entire machinery and mechanismes that pulling and backport patches downstream, so I don\u0027t think that it will be an issue for the projects under your umbrella.\n\u003e \n\u003e If delta exist between proposed hashes and the latest available hashes not yet release, I agree with you, it would be worth releasing them before transitioning.","accounts_in_message":[],"_revision_number":2},{"id":"335c996c46603770f7dd33eb5ac58558d249f608","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-06-03 12:53:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"808cfefd7b53d535405f548eb5b2035110d94f27","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 12:55:06.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"5d3624faf33c92297cca8c8d453a9234006fcb2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 13:02:57.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5aadcc465ce4b85bcaf037ca478262f : SUCCESS in 6m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7d159ea3664a4b279bcd4d30af36cdc9 : FAILURE in 7m 28s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/4b75751117a64496b6afe36684f85675 : FAILURE in 6m 55s","accounts_in_message":[],"_revision_number":3},{"id":"ed8309aef95e909212c4460b2375e3b20a6d06f3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-03 13:36:46.000000000","message":"Patch Set 3:\n\nThank you Marios, we will recheck this patch when the other one will be merged.","accounts_in_message":[],"_revision_number":3},{"id":"3d642debe7a9f3b7ad0a3efb228fdbf90cd30d0d","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-06-03 16:28:48.000000000","message":"Patch Set 3: Code-Review-1\n\nanyway this is blocked on the parent but i am -1 this waiting for a final sanity check reply from some folks i asked about packaging \n\ni will update tomorrow and remove my -1","accounts_in_message":[],"_revision_number":3},{"id":"16a41774f6ad3f9e9146faa9bf9430682415b880","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-06-04 15:15:14.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"a9ff5d3f9c6c7a6b53d689a6c76dbd7d499686fe","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-04 18:20:16.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"daf95ff29aa0dba309a68d92ef05c16e216f51ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 18:29:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c88419ec95e4894957de16007acc86a : SUCCESS in 7m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c43f5b97028a4207bd8c97b2b903e758 : SUCCESS in 5m 46s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7d79913b7c4f4c04b3902d8b05532b85 : SUCCESS in 6m 59s","accounts_in_message":[],"_revision_number":3},{"id":"fba118fa08d3ae83fb1ceaee7960691c8cb0ce28","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-04 18:44:41.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"39dd9c9e28fc58148f10577da18f8d5a59a2e58f","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-06-04 19:17:47.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nFinal Train release has merged + this patch has been updated by the PTL. Thanks, let\u0027s merge!","accounts_in_message":[],"_revision_number":3},{"id":"10230fa462f17662eae4930c73917b12bc898129","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 19:18:10.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b43b611ca9bd184fe205d1159cfb2bc0eff943a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 19:31:28.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17fe6e0f73124129af8ca8065dd8887a : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47355157b49d4674aa19a6d3602a55d2 : SUCCESS in 7m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dd9c7c54cee549d5954ce159c1a72cca : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":3},{"id":"e01344aacdd4ab4c5b7349d95d06854c1cef9913","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 19:31:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"33e27362ff38ae98d14985664d885811f6c6bdb4","revisions":{"0ea065379b13997a9a24df4da7536af247d15d18":{"kind":"REWORK","_number":1,"created":"2021-05-11 15:27:46.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/78/790778/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/78/790778/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/790778/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/790778/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/790778/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/78/790778/1"}}},"commit":{"parents":[{"commit":"4852e0a77a512f60179aee0e4aa992e860214668","subject":"Merge \"ironic-python-agent-builder release model change\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4852e0a77a512f60179aee0e4aa992e860214668"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-05-11 15:26:39.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-05-11 15:26:39.000000000","tz":120},"subject":"[tripleo] Transition Train to EM","message":"[tripleo] Transition Train to EM\n\nThis transition the train branch to extended maintenance.\nChanges for bugfixes and things the team deems important are\nstill encouraged, but there will no longer be official releases\noff of the branch.\n\nPlease +1 if the team is ready for us to proceed with this\ntransition, or -1 if there are any final backports currently in\nflight that we should wait for. For the latter case, please\nupdate the patch with the new commit hash after doing a final\nrelease to get those changes out so we know to proceed with the\ntransition.\n\nChange-Id: Ief89d3034de2bee7a5ec781d91cf038615fbe9c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0ea065379b13997a9a24df4da7536af247d15d18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0ea065379b13997a9a24df4da7536af247d15d18"}]},"branch":"refs/heads/master"},"21cdb997619574cfb6914db227021e7aebd968b3":{"kind":"REWORK","_number":2,"created":"2021-05-11 15:57:37.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/78/790778/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/78/790778/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/790778/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/790778/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/790778/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/78/790778/2"}}},"commit":{"parents":[{"commit":"4852e0a77a512f60179aee0e4aa992e860214668","subject":"Merge \"ironic-python-agent-builder release model change\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4852e0a77a512f60179aee0e4aa992e860214668"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-05-11 15:26:39.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-05-11 15:57:19.000000000","tz":120},"subject":"[tripleo] Transition Train to EM","message":"[tripleo] Transition Train to EM\n\nThis transition the train branch to extended maintenance.\nChanges for bugfixes and things the team deems important are\nstill encouraged, but there will no longer be official releases\noff of the branch.\n\nPlease +1 if the team is ready for us to proceed with this\ntransition, or -1 if there are any final backports currently in\nflight that we should wait for. For the latter case, please\nupdate the patch with the new commit hash after doing a final\nrelease to get those changes out so we know to proceed with the\ntransition.\n\nChange-Id: Ief89d3034de2bee7a5ec781d91cf038615fbe9c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/21cdb997619574cfb6914db227021e7aebd968b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/21cdb997619574cfb6914db227021e7aebd968b3"}]},"branch":"refs/heads/master"},"33e27362ff38ae98d14985664d885811f6c6bdb4":{"kind":"REWORK","_number":3,"created":"2021-06-03 12:53:56.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/78/790778/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/78/790778/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/790778/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/790778/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/790778/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/78/790778/3"}}},"commit":{"parents":[{"commit":"9ba448f0745b81c7443aff403a0a587c38ccc4bd","subject":"Make final train tags tripleo repos before extended maintenance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9ba448f0745b81c7443aff403a0a587c38ccc4bd"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-05-11 15:26:39.000000000","tz":120},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-06-03 12:53:19.000000000","tz":180},"subject":"[tripleo] Transition Train to EM","message":"[tripleo] Transition Train to EM\n\nThis transition the train branch to extended maintenance.\nChanges for bugfixes and things the team deems important are\nstill encouraged, but there will no longer be official releases\noff of the branch.\n\nPlease +1 if the team is ready for us to proceed with this\ntransition, or -1 if there are any final backports currently in\nflight that we should wait for. For the latter case, please\nupdate the patch with the new commit hash after doing a final\nrelease to get those changes out so we know to proceed with the\ntransition.\n\nCo-Authored-By: Marios Andreou \u003cmarios@redhat.com\u003e\nChange-Id: Ief89d3034de2bee7a5ec781d91cf038615fbe9c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/33e27362ff38ae98d14985664d885811f6c6bdb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/33e27362ff38ae98d14985664d885811f6c6bdb4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
