)]}'
{"id":"openstack%2Freleases~792820","triplet_id":"openstack%2Freleases~master~Ie632abd60936ce8cb382612c0bb96109b4d11617","project":"openstack/releases","branch":"master","topic":"puppet-t-final","hashtags":[],"change_id":"Ie632abd60936ce8cb382612c0bb96109b4d11617","subject":"Puppet OpenStack: Create final stable/train release","status":"MERGED","created":"2021-05-24 14:34:53.000000000","updated":"2021-06-16 15:33:37.000000000","submitted":"2021-06-16 15:33:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"792820-puppet-t-final","meta_rev_id":"a8cfe970482b90d42f14ef8c1e30f19aace181b6","_number":792820,"virtual_id_number":792820,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-16 15:33:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2021-06-16 14:51:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2021-06-16 15:06:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2021-06-16 14:55:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-06-16 15:15:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-06-16 15:15:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-06-16 14:59:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}],"reviewers":{"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-24 14:35:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-24 14:43:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-24 16:20:10.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2021-06-04 13:38:52.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-06-07 12:13:27.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2021-06-07 13:08:00.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-06-08 05:35:39.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"27bcabe88f4660e220cf8944fc034535e9b27fdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-24 14:34:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4bf1f4fab598c18d6cda018cd0497e5e86fb533b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-24 14:35:53.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"a816d59664948c500e32bd7719a8f1fc37899249","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 14:43:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dc85bb549374c48bfd9be2a915a0d39 : SUCCESS in 7m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/17d9d153a7ce49edb3a55f0e86952c53 : FAILURE in 7m 42s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/49c76d51fadc4a469afb9d3b50f24a99 : FAILURE in 5m 22s","accounts_in_message":[],"_revision_number":1},{"id":"b50c970e772e94762270813c35724f2a7c820a43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-25 13:05:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5b0ca84942d598d38514d70a757059ab647bbe61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 13:13:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dba57ad012af4d68971bdb2362eb506e : SUCCESS in 5m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1589e9af45f04695b7899017e1b443f1 : FAILURE in 6m 48s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/71f42edd1c3342a1b8134fb163341d02 : FAILURE in 4m 56s","accounts_in_message":[],"_revision_number":2},{"id":"f67089596b40bdb83c411e87e61c31cadc52a950","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-25 13:17:27.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"17fb372c3ecfb5d72eeee12c547bcb79ec57a0f0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-26 23:34:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b1918dc6029691474446b77ccfef0df08faf5822","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 23:43:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62a6185c3c544a7592e916f7dc620b24 : SUCCESS in 6m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e7d65c25af314e28abc56c9d8b889d62 : SUCCESS in 6m 58s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/4d417c3b55564b4fa439cf2b883dc8a9 : SUCCESS in 7m 11s","accounts_in_message":[],"_revision_number":3},{"id":"90c294f7a01be375abb9b18f8175a60bb005cdcb","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-27 00:19:48.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"e3403e396fda94981b6de82e3c4a4636b460e88b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-04 13:38:52.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"09253ca02af8fd25a59b6afbf9373332fae6c398","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-06-07 12:13:27.000000000","message":"Patch Set 3: Code-Review+2\n\nLGTM but missing PTL approval","accounts_in_message":[],"_revision_number":3},{"id":"94b4afb934115cba12b07e9c2e5b879cb5f3ad28","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-06-07 13:08:00.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"0373400f4e8341a899fea3eb222cdc009b30615b","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 13:09:17.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"5240dcbc35a8a40c278a56ab23ff1c721efb0870","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-07 13:15:38.000000000","message":"Patch Set 3: Workflow+1\n\nValidated by Alex who is the release liaison. Thanks Alex","accounts_in_message":[],"_revision_number":3},{"id":"06995383627fe53cc2cd3565e5c857edd748cd37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 13:15:55.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4a2d2ecadb5986e61149c7d36ed19b08d14eb47c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 13:24:38.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40663559c11d4b2292a05d995a14af92 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a90a3eb146741d581f9c6cafebee61a : SUCCESS in 7m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fdaf2cd94e9043f782b80c3efdfc01bb : FAILURE in 8m 04s","accounts_in_message":[],"_revision_number":3},{"id":"f68c7522b2f5247b449cba6ad67c123eaf1e70d3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-07 14:38:56.000000000","message":"Patch Set 3:\n\nIt seems we are blocked by https://review.opendev.org/c/openstack/project-config/+/793891 ...","accounts_in_message":[],"_revision_number":3},{"id":"5a219644b5204d0ea4fd301d7ecf0e7c2a19fdbc","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-06-08 05:35:39.000000000","message":"Patch Set 3:\n\nKeystone \u0027final train release\u0027 was published yesterday. Is that included in this release patch? And if not, do you need it to be included?","accounts_in_message":[],"_revision_number":3},{"id":"c6063fa9a1bd2dc6106ab9035694f18c8ade4bde","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-08 09:18:39.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3:\n\u003e \n\u003e Keystone \u0027final train release\u0027 was published yesterday. Is that included in this release patch? And if not, do you need it to be included?\n\nSorry but I don\u0027t understand how keystone would be related to this patch for Puppet OpenStack ?","accounts_in_message":[],"_revision_number":3},{"id":"0904400872e372716f18b10a1fe06387b5b89b74","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-09 13:23:05.000000000","message":"Patch Set 3: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b56683f6f0a9c8caf3b8a13c2e78fc0e446a924b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-09 13:24:07.000000000","message":"Patch Set 3: Workflow-1\n\nholding until we are sure about the keystone part","accounts_in_message":[],"_revision_number":3},{"id":"35dfe384dc1cf54bf577f09841e0710e185b7d23","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-06-09 13:30:25.000000000","message":"Patch Set 3:\n\nnew releases of the services do not necessarily impact puppet modules.","accounts_in_message":[],"_revision_number":3},{"id":"c3f6b2687bc2f3853d6e3b9d19b62c72abcac682","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-09 13:39:53.000000000","message":"Patch Set 3:\n\nI went through the release note of keystone and keystonemiddleware but found nothing special from puppet\u0027s perspective.\n\nIn general puppet is a configuration tool and we don\u0027t expect breaking change made by service side in stable branches so we rarely need to care about release of keystone.","accounts_in_message":[],"_revision_number":3},{"id":"d32e4cfa2b835dc6cdf1a28ba085d7e97744692d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-06-09 14:49:50.000000000","message":"Patch Set 3:\n\nSorry, yes, I thought that Puppet OpenStack, as a trailing project depends on the project releases.\nThanks everyone for the clarification.","accounts_in_message":[],"_revision_number":3},{"id":"f94725600a8734326d6aac110d8b04c8fb02358f","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-06-09 15:07:21.000000000","message":"Patch Set 3:\n\nrecheck -  I guess the validator failure [1] is a false negative as it passed before.\n\n[1] https://zuul.opendev.org/t/openstack/build/fdaf2cd94e9043f782b80c3efdfc01bb/log/job-output.txt#28787","accounts_in_message":[],"_revision_number":3},{"id":"727f920b315b6656ef085db29747f9326f5da9f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 15:20:52.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59d744285c094d45a9dd8c3d51167d63 : SUCCESS in 7m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8000a4f915834b65958b67a05fed04e6 : FAILURE in 9m 11s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b4e343dd71bf41cc8651ed6a34228959 : SUCCESS in 10m 15s","accounts_in_message":[],"_revision_number":3},{"id":"31747165cc89d5d3dca33b319863e5a107cb9d24","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-06-09 15:32:48.000000000","message":"Patch Set 3:\n\nlooks like the retirement of puppet-panko has broken stuff","accounts_in_message":[],"_revision_number":3},{"id":"15ee8fbd39cd7216b16b13ed48c89771fb735863","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-09 15:47:26.000000000","message":"Patch Set 3: -Workflow\n\nThanks everybody for your confirmation about keystone.","accounts_in_message":[],"_revision_number":3},{"id":"c7ab6678732448f3afa2db59aa04236f78af451c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-09 15:51:45.000000000","message":"Patch Set 3: -Code-Review\n\n\u003e Patch Set 3:\n\u003e \n\u003e looks like the retirement of puppet-panko has broken stuff\n\nYes I think that these changes broke the gates [1].\nI\u0027ll try re-adding it temporarly to allow us to merge these changes, and then I\u0027ll propose to revert it.\n\n[1] https://opendev.org/openstack/project-config/commit/a5938e0bc94b8ec28e92fd8475eaefbfffe68d30","accounts_in_message":[],"_revision_number":3},{"id":"0adf644e8a7951d0c72135e6622628dace5d49a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-09 15:59:09.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"e1beda0b84e692bd7260eafad378af2eab2eb7dd","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-09 16:07:05.000000000","message":"Patch Set 4:\n\nWell I proposed to revert the openstack/project-config changes that led us to this error.\nI added a depends-on on my revert to merge it first.\nAlso, in a second time, I proposed to revert my revert (inception) when puppet will be transitioned to EM by adding a depends-on https://review.opendev.org/c/openstack/releases/+/790767 on the project-config revert https://review.opendev.org/c/openstack/project-config/+/795586\n\nHopefully my comment is understandable.","accounts_in_message":[],"_revision_number":4},{"id":"f6d1ce6173a4aca0203fd9caed949d02203fca63","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-09 16:11:52.000000000","message":"Patch Set 4:\n\nAlso I wonder if the project-config changes should be left in-place until Wallaby isn\u0027t transitioned to extended maintenance, else we will face similar issues for ussuri, victoria and wallaby too in few months","accounts_in_message":[],"_revision_number":4},{"id":"2770da7f41405e5e2bd0613b4d455297846b555d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-06-09 16:12:47.000000000","message":"Patch Set 4: Code-Review+1\n\nyea seems like it needs to continue until we\u0027re done releasing all the things","accounts_in_message":[],"_revision_number":4},{"id":"bd666c9cd980b88a58852c9b363786a90f6ca006","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 16:12:50.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/149e89ab5d9d4216aaabe60161a19d78 : SUCCESS in 8m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5f5fb0c9042e49afb695573ca4f17936 : FAILURE in 9m 25s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/4cb56fd5175b486da2ef6dd3314bd098 : SUCCESS in 11m 24s","accounts_in_message":[],"_revision_number":4},{"id":"7ffd1ab3d2a61c477f9f6b63cccff0094c3b3fea","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 16:13:36.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"0eded16c3116d1b511854ccb22313d23ff35ed8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-09 16:24:43.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"3f775a2b282234ab1b0bc9809fa65ec537a28e3e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-09 16:26:29.000000000","message":"Patch Set 5:\n\nUsing Matthias\u0027s patch as depends-on https://review.opendev.org/c/openstack/project-config/+/793891","accounts_in_message":[],"_revision_number":5},{"id":"180caa9309927cab6936ddb8e076c5b17e12531e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 16:34:55.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b85a323b71d4251a86b261ed2e404aa : SUCCESS in 7m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/61dc1207d8704a298528296882c76383 : FAILURE in 8m 09s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d3135f379aa54e02be65ef03f0a2390e : SUCCESS in 8m 20s","accounts_in_message":[],"_revision_number":5},{"id":"0f690bc2a6985ab1eac0228ea1ca4ec3f29eaeb9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-15 15:01:56.000000000","message":"Patch Set 5:\n\nrecheck\n\nproject-config patch is now merged https://review.opendev.org/c/openstack/project-config/+/793891","accounts_in_message":[],"_revision_number":5},{"id":"d44c257e59985180255387d2c07c3b7bd7fc11e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 15:15:52.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09727c09cd984bdeae209bb85422d7e5 : SUCCESS in 7m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/754a16e4c292437cb3550a56b0fa9278 : FAILURE in 8m 49s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d7cc4dc931ba40df9097c424dca46810 : SUCCESS in 11m 22s","accounts_in_message":[],"_revision_number":5},{"id":"449afc61c510e7d8956565905d5788c6e3cc44ba","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-15 15:52:29.000000000","message":"Patch Set 5:\n\nrecheck\n\nIt\u0027s weird templates have been readded but our validation still fail.","accounts_in_message":[],"_revision_number":5},{"id":"02f7881ec0a0f339ea32be3b4fbcdc6aa168efbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 16:01:42.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbd5dc54baee44d386f405fb16cb78f3 : SUCCESS in 6m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6e9cd1791f9c43ad8cafa09d59c5a20c : FAILURE in 8m 09s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e136012322f54500837ca019b6c55bc0 : SUCCESS in 6m 46s","accounts_in_message":[],"_revision_number":5},{"id":"96c65af7313a30ae4fa7378f0d976f43525bbd20","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-06-16 08:56:47.000000000","message":"Patch Set 5:\n\nrecheck in case this was a sync issue","accounts_in_message":[],"_revision_number":5},{"id":"1c682ead36c9f7b28eb37bdfeed39dbde88e2a1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 09:13:43.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c598b56760345d5aedf5da6fde42e44 : SUCCESS in 8m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5bd631f96753468c9a74c3ce35db8c88 : FAILURE in 8m 34s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ef2042d77e9141189752df64de7c62cf : SUCCESS in 8m 42s","accounts_in_message":[],"_revision_number":5},{"id":"a5ee0c0e326e8f0276d54cf6fafe959ddc0048a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-06-16 14:24:10.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"09b248db5b0aeff6777168328bd71a06b46e1ea3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 14:39:11.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/806d896121cb47f49663f2de9012c7b0 : SUCCESS in 6m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d0ef04c3dcaf41718490650ca09e50d7 : SUCCESS in 8m 12s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/412f15c92ea7440c9a41b4040fb68e81 : SUCCESS in 8m 06s","accounts_in_message":[],"_revision_number":6},{"id":"cdcaa10bd4d7ac01300a4381dd582b1bc59408d3","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-06-16 14:51:03.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"45afdd43bf31d4ea42ee01725f3727b51adf16e7","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-06-16 14:55:51.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"c35f7c7d65fe39e24e352f459cb6cb1c34c867f5","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 14:59:22.000000000","message":"Patch Set 6: PTL-Approved+1","accounts_in_message":[],"_revision_number":6},{"id":"5a39bba29686b0c0aad36b74c7531b442d057301","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-16 15:06:09.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d5c1d8e5f860bb01b60edfb1657c5db5c549ce08","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-16 15:15:27.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"bb65d54f4b8822a36e80b58bae7364a8096d18fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 15:15:50.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"bb545a91674bed4dba5daa71eabf8ebdcf6d9cc4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 15:33:36.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70603a9427a94abc92d0c91479ebbcbc : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78dead8ac74c4a5bab70da9da2f9707a : SUCCESS in 7m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bcde05e5ce594391aa8a9105700279eb : SUCCESS in 10m 34s","accounts_in_message":[],"_revision_number":6},{"id":"a8cfe970482b90d42f14ef8c1e30f19aace181b6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 15:33:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"c88e55c5413c4fe063e1d04bc6b211dc39b1ee44","revisions":{"56b5d30968062811ade6f352f6f0217f2b972b3d":{"kind":"REWORK","_number":1,"created":"2021-05-24 14:34:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/20/792820/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/20/792820/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/20/792820/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/20/792820/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/20/792820/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/20/792820/1"}}},"commit":{"parents":[{"commit":"bb612045e95599a6e9cc589c50bbf5e32c120ebe","subject":"Merge \"Release osc-lib 2.4.0 for Xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bb612045e95599a6e9cc589c50bbf5e32c120ebe"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-24 14:34:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-24 14:34:39.000000000","tz":540},"subject":"Puppet OpenStack: Create final stable/train release","message":"Puppet OpenStack: Create final stable/train release\n\nChange-Id: Ie632abd60936ce8cb382612c0bb96109b4d11617\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/56b5d30968062811ade6f352f6f0217f2b972b3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/56b5d30968062811ade6f352f6f0217f2b972b3d"}]},"branch":"refs/heads/master"},"0c5e72804b23a9a5f9ce3377753eac699e0cd3a8":{"kind":"REWORK","_number":2,"created":"2021-05-25 13:05:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/20/792820/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/20/792820/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/20/792820/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/20/792820/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/20/792820/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/20/792820/2"}}},"commit":{"parents":[{"commit":"bb612045e95599a6e9cc589c50bbf5e32c120ebe","subject":"Merge \"Release osc-lib 2.4.0 for Xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bb612045e95599a6e9cc589c50bbf5e32c120ebe"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-24 14:34:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-25 13:04:02.000000000","tz":540},"subject":"Puppet OpenStack: Create final stable/train release","message":"Puppet OpenStack: Create final stable/train release\n\nChange-Id: Ie632abd60936ce8cb382612c0bb96109b4d11617\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0c5e72804b23a9a5f9ce3377753eac699e0cd3a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0c5e72804b23a9a5f9ce3377753eac699e0cd3a8"}]},"branch":"refs/heads/master"},"2bdbfdc35c7bbce8c332ab66b8f56cbfe4cf8c0f":{"kind":"REWORK","_number":3,"created":"2021-05-26 23:34:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/20/792820/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/20/792820/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/20/792820/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/20/792820/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/20/792820/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/20/792820/3"}}},"commit":{"parents":[{"commit":"bb612045e95599a6e9cc589c50bbf5e32c120ebe","subject":"Merge \"Release osc-lib 2.4.0 for Xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bb612045e95599a6e9cc589c50bbf5e32c120ebe"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-24 14:34:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-26 23:29:52.000000000","tz":540},"subject":"Puppet OpenStack: Create final stable/train release","message":"Puppet OpenStack: Create final stable/train release\n\nChange-Id: Ie632abd60936ce8cb382612c0bb96109b4d11617\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2bdbfdc35c7bbce8c332ab66b8f56cbfe4cf8c0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2bdbfdc35c7bbce8c332ab66b8f56cbfe4cf8c0f"}]},"branch":"refs/heads/master"},"1a04b7653270124f5add73362115cc8d2d321d6f":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-06-09 15:59:09.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/20/792820/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/20/792820/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/20/792820/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/20/792820/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/20/792820/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/20/792820/4"}}},"commit":{"parents":[{"commit":"bb612045e95599a6e9cc589c50bbf5e32c120ebe","subject":"Merge \"Release osc-lib 2.4.0 for Xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bb612045e95599a6e9cc589c50bbf5e32c120ebe"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-24 14:34:19.000000000","tz":540},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-09 15:58:40.000000000","tz":120},"subject":"Puppet OpenStack: Create final stable/train release","message":"Puppet OpenStack: Create final stable/train release\n\nDepends-On: https://review.opendev.org/c/openstack/project-config/+/795583\nChange-Id: Ie632abd60936ce8cb382612c0bb96109b4d11617\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1a04b7653270124f5add73362115cc8d2d321d6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1a04b7653270124f5add73362115cc8d2d321d6f"}]},"branch":"refs/heads/master"},"31a4eadf5228ac65fb6cd8d2fb93e95dcbdcb637":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2021-06-09 16:24:43.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/20/792820/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/20/792820/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/20/792820/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/20/792820/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/20/792820/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/20/792820/5"}}},"commit":{"parents":[{"commit":"bb612045e95599a6e9cc589c50bbf5e32c120ebe","subject":"Merge \"Release osc-lib 2.4.0 for Xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bb612045e95599a6e9cc589c50bbf5e32c120ebe"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-24 14:34:19.000000000","tz":540},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-09 16:24:12.000000000","tz":120},"subject":"Puppet OpenStack: Create final stable/train release","message":"Puppet OpenStack: Create final stable/train release\n\nDepends-On: https://review.opendev.org/c/openstack/project-config/+/793891\nChange-Id: Ie632abd60936ce8cb382612c0bb96109b4d11617\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/31a4eadf5228ac65fb6cd8d2fb93e95dcbdcb637"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/31a4eadf5228ac65fb6cd8d2fb93e95dcbdcb637"}]},"branch":"refs/heads/master"},"c88e55c5413c4fe063e1d04bc6b211dc39b1ee44":{"kind":"REWORK","_number":6,"created":"2021-06-16 14:24:10.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/20/792820/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/20/792820/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/20/792820/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/20/792820/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/20/792820/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/20/792820/6"}}},"commit":{"parents":[{"commit":"bb612045e95599a6e9cc589c50bbf5e32c120ebe","subject":"Merge \"Release osc-lib 2.4.0 for Xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bb612045e95599a6e9cc589c50bbf5e32c120ebe"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-24 14:34:19.000000000","tz":540},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2021-06-16 14:23:25.000000000","tz":120},"subject":"Puppet OpenStack: Create final stable/train release","message":"Puppet OpenStack: Create final stable/train release\n\nDepends-On: https://review.opendev.org/c/openstack/project-config/+/793891\nChange-Id: Ie632abd60936ce8cb382612c0bb96109b4d11617\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c88e55c5413c4fe063e1d04bc6b211dc39b1ee44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c88e55c5413c4fe063e1d04bc6b211dc39b1ee44"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
