)]}'
{"id":"openstack%2Freleases~792935","triplet_id":"openstack%2Freleases~master~I98005d186a4137417d3f131adffc8965c91e365b","project":"openstack/releases","branch":"master","topic":"xena-milestone-1","hashtags":[],"change_id":"I98005d186a4137417d3f131adffc8965c91e365b","subject":"Release python-manilaclient for Xena-1 milestone","status":"MERGED","created":"2021-05-25 10:28:25.000000000","updated":"2021-06-02 21:31:25.000000000","submitted":"2021-06-02 21:31:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"792935-xena-milestone-1","meta_rev_id":"46531a4570bdaa71bd86320a723d24fb31266326","_number":792935,"virtual_id_number":792935,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-02 21:31:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2021-05-31 10:36:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2021-06-02 11:19:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2021-05-28 13:47:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2021-05-31 10:36:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"date":"2021-06-02 11:10:44.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-05-27 22:37:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-25 10:29:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-25 10:40:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-25 11:20:42.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2021-05-25 11:20:43.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2021-05-28 12:10:37.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-05-31 10:36:08.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"d8f437667dd7bbe46e38b7b4037b9030f94ae9fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-05-25 10:28:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa915b0a97e0f28b6c99b46f43853f4c1f538ed7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 10:40:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6330b8ca420d4468bf186433a75eeb36 : SUCCESS in 6m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/15a5edf40bcc4b7485e9c567e981be10 : SUCCESS in 5m 23s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e4d92efcee8242c4a48b7c7ecf453d85 : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":1},{"id":"0bbb58cc288a6758fcb99f948f77b3261468f6d6","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-05-27 16:21:03.000000000","message":"Patch Set 1: Code-Review-1\n\nWe have some changes to include and will be updating the hash.","accounts_in_message":[],"_revision_number":1},{"id":"49c053a4422c808a9c7ca9c670c5312158424ddf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-05-27 22:36:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1deeeaf32f50c418ef98ec28a7a28cc87f9e87c5","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 22:37:13.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"18a6adb6137b8d9b712632267dc87660a051d5b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 22:47:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90b9f375c2ef4e7daa84b43d7d5ac3d6 : SUCCESS in 8m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e5898fb5198f40bbb0be46ada1fa0f29 : SUCCESS in 6m 16s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c9b3ca5cebef4f12b3e8ee9ecc1e974f : SUCCESS in 6m 46s","accounts_in_message":[],"_revision_number":2},{"id":"fb54b61e4d218f1e5615fe62707d5f9121612281","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-05-28 12:10:37.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)\n\nI think this is OK, but there is one question regarding the version bump, so giving just +1.","accounts_in_message":[],"_revision_number":2},{"id":"4ffa693bea9bb8e0011d627c840a5515b177e4f4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-05-28 13:47:10.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6e81a034a750ce8d3c2ba0772a04a592c6a60134","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-05-31 10:36:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f31ee688a2d4da699960748307625cbdbbca80f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 10:36:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d4a94d4a14ce0a8765855f4d0be46ee8879403b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 10:48:46.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/710c10dca74e4a10a05720c5c877db68 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d66af6ef8c224fb78bc17975300fef5c : SUCCESS in 9m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/69019cad61e04c96a2c8fe3f6e7e6983 : FAILURE in 5m 52s","accounts_in_message":[],"_revision_number":2},{"id":"b728c84d17969460d14979a9bfc0549b37b25adb","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-05-31 13:52:11.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4cb4dd9651cb585f9f4324545e0d084d420efebb","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-05-31 13:53:11.000000000","message":"Patch Set 2: Workflow-1\n\nAs discussed on IRC we want to wait a bit for Goutham\u0027s response","accounts_in_message":[],"_revision_number":2},{"id":"89c88244e0fdca748d0383b68f68c8ebc7e660d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 14:00:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31dfc0462314425c8967f26502a00304 : SUCCESS in 6m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/73b40b40a9524932bd6ccc37ade84bd9 : SUCCESS in 6m 37s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/fa55d7e331e7479a9009b19561978d8f : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":2},{"id":"82abaf2268c3907a72bf3b9a75df09b533f0eb98","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-02 10:06:51.000000000","message":"Patch Set 2:\n\n@Goutham: Please can you answer Elod\u0027s comment?","accounts_in_message":[],"_revision_number":2},{"id":"5323447bb6fa31eb4ab1aa9ca6e8f9daeb99e134","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-06-02 10:58:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"45f5df0b59456b74ba688533a989b8d443ad4d01","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-06-02 11:04:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dca256365cc35ec0396e433f4e860bca25779e7c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-02 11:10:44.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"f8d59a4d495dd3e3532e1df06f15d23e5c9b6393","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-06-02 11:19:48.000000000","message":"Patch Set 2: Code-Review+2\n\nAs we discussed on IRC [1]: no need for the major version bump, so this patch is good to go\n\n[1] http://eavesdrop.openstack.org/irclogs/%23openstack-release/%23openstack-release.2021-06-02.log.html#t2021-06-02T11:00:37","accounts_in_message":[],"_revision_number":2},{"id":"80b655f0ffea08999ef8992ef6332c395f08f1df","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-06-02 18:23:17.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"cdd57899c82b18c73bb13c04b84f840e4082422d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 18:34:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2879e4bc9f64b599fe73814e25bfc0f : SUCCESS in 8m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c62bae70542b434ea745279f21671313 : SUCCESS in 6m 08s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6b05aa92b961477998a9dfa1bf1a6301 : SUCCESS in 6m 51s","accounts_in_message":[],"_revision_number":2},{"id":"f0806706c40b5c65dd5143b3affe5955c0396b38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 18:36:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"18df358ad0b14fc6571d2f1695b0c63b5d95dd59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 18:45:31.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a2df08a0abd4a9f87fc76a8941c33c6 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abd814d634d845489265598626c0c15c : FAILURE in 7m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/968561d8dabb45d5bb2d1ca37661584d : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":2},{"id":"7652154d044c156aa16ec9b71895094d36414c3a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-06-02 20:57:43.000000000","message":"Patch Set 2:\n\nrecheck\n\nseems like a sporadic error","accounts_in_message":[],"_revision_number":2},{"id":"fe436d24a1f628194204d20da6887393541bac90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 21:17:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e65c5372ab934adab02c482d8ba6660d : SUCCESS in 6m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a807c8e9d7044366b40b6111fe89a812 : SUCCESS in 5m 40s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b3d29e81128e44839387989ed50328f7 : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":2},{"id":"61177f2d41d294b268b5bf36c06303f69ceb732a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 21:18:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"dbaaa901bcff98b863ff2a70c2d2e64e4a35afaf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 21:31:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08187f4f4932440480178a4596cf14e9 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f0c955e633c40bcbb9d5805b5783faa : SUCCESS in 6m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a20a9ad4a072405d946e569e90f52258 : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":2},{"id":"46531a4570bdaa71bd86320a723d24fb31266326","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 21:31:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"65b79de8c839a9cfc477a7f5ececec2ce8ee6714","revisions":{"7e202148a8fd581c3b2b02fbc30bc5032a568a05":{"kind":"REWORK","_number":1,"created":"2021-05-25 10:28:25.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/35/792935/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/35/792935/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/792935/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/792935/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/792935/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/35/792935/1"}}},"commit":{"parents":[{"commit":"1354f4acab5b77e927fdbbcae6be7e6ed18be786","subject":"Merge \"[watcher] Transition Train to EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1354f4acab5b77e927fdbbcae6be7e6ed18be786"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-05-25 10:28:10.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-05-25 10:28:10.000000000","tz":120},"subject":"Release python-manilaclient for Xena-1 milestone","message":"Release python-manilaclient for Xena-1 milestone\n\nThis is a library release for ospython-manilaclient for the Xena-1\nmilestone. This repo includes commits that have not been releases for\nthis cycle yet.\n\nIf the team is ready to process with this release, please leave a +1\nto indicate we should go ahead with the release.\n\nIf the team needs more time for things about to merge, or if there is\nsome other reason a release should not be done at this time, please\nleave a -1 with a comment indicating the status. Then update the patch\nwith the new commit hash to use once the team is ready to do the\nrelease.\n\n$ git log --oneline --no-merges 2.6.0..7ecc819\n7ecc819 setup.cfg: Replace dashes with underscores\n69031b5 Fix error in CLI \u0027manila list\u0027 with sorting key \u0027availability_zone\u0027.\n96a7942 Add Python3 xena unit tests\n71ebd96 Update master for stable/wallaby\nac127bd Fix test failures for \"test_list_shares_by_{filter}\"\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I98005d186a4137417d3f131adffc8965c91e365b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7e202148a8fd581c3b2b02fbc30bc5032a568a05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7e202148a8fd581c3b2b02fbc30bc5032a568a05"}]},"branch":"refs/heads/master"},"65b79de8c839a9cfc477a7f5ececec2ce8ee6714":{"kind":"REWORK","_number":2,"created":"2021-05-27 22:36:16.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/35/792935/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/35/792935/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/792935/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/792935/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/792935/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/35/792935/2"}}},"commit":{"parents":[{"commit":"1354f4acab5b77e927fdbbcae6be7e6ed18be786","subject":"Merge \"[watcher] Transition Train to EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1354f4acab5b77e927fdbbcae6be7e6ed18be786"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-05-25 10:28:10.000000000","tz":120},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-05-27 22:35:53.000000000","tz":-420},"subject":"Release python-manilaclient for Xena-1 milestone","message":"Release python-manilaclient for Xena-1 milestone\n\nThis is a library release for python-manilaclient for\nthe Xena-1 milestone.\n\n$ git log --oneline --no-merges 2.6.0..99ccf02\n99ccf02d0 [TrivialFix] Follow up per-share-gigabytes quota changes\n1f31818d2 Add per share gigabytes quota to the CLI\n5ff8b70cd [ci] fix neutron services in functional job\n7ecc819d1 (gerrit/master) setup.cfg: Replace dashes with underscores\n69031b558 Fix error in CLI \u0027manila list\u0027 with sorting key \u0027availability_zone\u0027.\n96a7942e1 Add Python3 xena unit tests\n71ebd961d Update master for stable/wallaby\nac127bd33 (bug/1821945) Fix test failures for \"test_list_shares_by_{filter}\"\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I98005d186a4137417d3f131adffc8965c91e365b\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/65b79de8c839a9cfc477a7f5ececec2ce8ee6714"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/65b79de8c839a9cfc477a7f5ececec2ce8ee6714"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
