)]}'
{"id":"openstack%2Freleases~793890","triplet_id":"openstack%2Freleases~master~Ib616837841a5d470b4d3fb1ef710fea4c8ead4c3","project":"openstack/releases","branch":"master","topic":"openstackdocstheme","hashtags":[],"change_id":"Ib616837841a5d470b4d3fb1ef710fea4c8ead4c3","subject":"Release openstackdocstheme 2.3.0","status":"MERGED","created":"2021-06-01 06:45:51.000000000","updated":"2021-06-02 15:26:56.000000000","submitted":"2021-06-02 15:26:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"793890-openstackdocstheme","meta_rev_id":"c37d76daef9bf83d4204a90bbf9344ed31fc94f1","_number":793890,"virtual_id_number":793890,"owner":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-02 15:26:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2021-06-02 15:11:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":2,"date":"2021-06-02 09:54:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2021-06-02 15:11:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-06-02 09:55:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-01 06:47:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-01 06:57:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-01 07:06:23.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"CC"},{"updated":"2021-06-01 08:25:51.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2021-06-01 09:25:51.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2021-06-02 09:54:21.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-06-02 15:11:32.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"15268bc4bbe98e86e3149dfbff53ba54f4ac5c1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-06-01 06:45:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08584c94529e441efe0eda7cb1eb3c00cf7ff935","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 06:57:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b33c722d2184aa7ac59f18435b82da4 : SUCCESS in 7m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ef707de5e3aa4cb9a56d85530c9f0197 : SUCCESS in 6m 02s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/0f618c10866545d3bbafc3b37a133113 : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":1},{"id":"70ab5440f7b5ad90026b28a523f4cb63f9fb6618","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-06-01 07:06:23.000000000","message":"Patch Set 1:\n\nI suggest to include the stack starting at https://review.opendev.org/c/openstack/openstackdocstheme/+/793818/ - if that one merges quickly. Those changes are needed for new Sphinx release and once they are in a new release is needed.","accounts_in_message":[],"_revision_number":1},{"id":"010d84a920103c483563d1a7c5d33a6b81d332cd","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-06-01 08:02:27.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e I suggest to include the stack starting at https://review.opendev.org/c/openstack/openstackdocstheme/+/793818/ - if that one merges quickly. Those changes are needed for new Sphinx release and once they are in a new release is needed.\n\nSounds good. The patch mentioned is simple enough and I hope it lands soon.","accounts_in_message":[],"_revision_number":1},{"id":"e4d9b9d81529cd68111637390b4668101bd8f227","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-06-01 08:05:56.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e I suggest to include the stack starting at https://review.opendev.org/c/openstack/openstackdocstheme/+/793818/ - if that one merges quickly. Those changes are needed for new Sphinx release and once they are in a new release is needed.\n\u003e \n\u003e Sounds good. The patch mentioned is simple enough and I hope it lands soon.\n\nI just thought the first patch of the series. I am not sure we land all of them quickly.\nI would like to wait for more input from docs folks.","accounts_in_message":[],"_revision_number":1},{"id":"1a78a12e7d897dd9b8c83a786135d1ddf268c640","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-06-01 08:25:51.000000000","message":"Patch Set 1:\n\nWe need the whole stack to merge - see http://lists.openstack.org/pipermail/openstack-discuss/2021-May/022750.html for the reason.","accounts_in_message":[],"_revision_number":1},{"id":"13e7e19bec8b51a33800a8c7aae140d4d2185101","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-01 09:25:51.000000000","message":"Patch Set 1:\n\n(1 comment)\n\n\u003e Patch Set 1:\n\u003e \n\u003e We need the whole stack to merge - see http://lists.openstack.org/pipermail/openstack-discuss/2021-May/022750.html for the reason.\n\nI fast approved some of these patches and I asked for another core review for the rest, let\u0027s wait a bit for these feedbacks.\n\nIf nobody review these remaining patches, then I suppose that we could \"single approve\" them to address the emergency.","accounts_in_message":[],"_revision_number":1},{"id":"26013caa50ad13294d6eef7d9c86219f79779d65","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-06-01 10:05:47.000000000","message":"Patch Set 1: Workflow-1\n\nAll discussion around sphinx 4.0 support makes sense to me. Let\u0027s wait for a while and update the commit hash once the patch series lands. Marking this as WIP.","accounts_in_message":[],"_revision_number":1},{"id":"f96713dc0f03ba05c8d33ed6dba0711bb095c3bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-06-02 03:35:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6d9f5bfe823275b6634bd126035555addaacce5a","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-06-02 03:35:53.000000000","message":"Patch Set 2:\n\nThe patch series for Sphinx 4 has landed. I updated the commit hash in the release file.","accounts_in_message":[],"_revision_number":2},{"id":"bad063683af2335631bff647bbb1b933c57e5ec8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 03:42:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21a90d03b9ea410facd53ebbb948545b : SUCCESS in 6m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a8b34ebb71d148898215d2f571fda4ad : SUCCESS in 5m 39s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/4ea8d5dc74a845f5ad0ccbad10f61413 : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":2},{"id":"b63ecba329c28fb658271008fe6625554b2c6990","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-02 09:54:21.000000000","message":"Patch Set 2: Code-Review+2\n\nStephen\u0027s changes are all merged and landed here, LGTM as release liaison for openstackdocsthem and LGTM as release team core member","accounts_in_message":[],"_revision_number":2},{"id":"78307bb53ba25e898e5b95c457e1514922458c3f","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 09:55:42.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"99183b367f3d449d44d4bcb006bcbc3f8a62102b","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-06-02 15:11:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLGTM:\n* version bump looks OK to me as explained in the commit message\n* approved by release liaison\n* no issues found in test job logs\nLet\u0027s release this.","accounts_in_message":[],"_revision_number":2},{"id":"37cdfac0dae12b153adcb0379958ddde84e1a615","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 15:18:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"da37703ac2a01a4f221afb50f47f514b35211cdf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 15:26:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e36961cca5ca47aeba976a2a97563476 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f13ac38d76f4903a632bbe182fdce1d : SUCCESS in 6m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/83a765fabaa2468c9a9b2eeaf266ada8 : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":2},{"id":"c37d76daef9bf83d4204a90bbf9344ed31fc94f1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 15:26:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5bfefe65729f9f047576b8929d151173bcf6b837","revisions":{"327e129546303d4b0837fc7f3157b97337bad20e":{"kind":"REWORK","_number":1,"created":"2021-06-01 06:45:51.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/90/793890/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/90/793890/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/793890/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/793890/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/793890/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/90/793890/1"}}},"commit":{"parents":[{"commit":"fa14e6c3ca4b4a6ce4111e9d7d4470d27eead17c","subject":"Merge \"Release python-tackerclient for Xena-1 milestone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fa14e6c3ca4b4a6ce4111e9d7d4470d27eead17c"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2021-06-01 06:43:14.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2021-06-01 06:43:14.000000000","tz":540},"subject":"Release openstackdocstheme 2.3.0","message":"Release openstackdocstheme 2.3.0\n\nReferences to Freenode in openstackdocstheme were replaced with OFTC.\nWe need to release it so that each project can consume it.\nReplacing freenode with OFTC sounds a feature rather than a bug fix,\nso the minor version bump is proposed as 2.3.0.\n\nChange-Id: Ib616837841a5d470b4d3fb1ef710fea4c8ead4c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/327e129546303d4b0837fc7f3157b97337bad20e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/327e129546303d4b0837fc7f3157b97337bad20e"}]},"branch":"refs/heads/master"},"5bfefe65729f9f047576b8929d151173bcf6b837":{"kind":"REWORK","_number":2,"created":"2021-06-02 03:35:10.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/90/793890/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/90/793890/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/793890/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/793890/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/793890/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/90/793890/2"}}},"commit":{"parents":[{"commit":"fa14e6c3ca4b4a6ce4111e9d7d4470d27eead17c","subject":"Merge \"Release python-tackerclient for Xena-1 milestone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fa14e6c3ca4b4a6ce4111e9d7d4470d27eead17c"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2021-06-01 06:43:14.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2021-06-02 03:33:12.000000000","tz":540},"subject":"Release openstackdocstheme 2.3.0","message":"Release openstackdocstheme 2.3.0\n\nReferences to Freenode in openstackdocstheme were replaced with OFTC.\nWe need to release it so that each project can consume it.\nReplacing freenode with OFTC sounds a feature rather than a bug fix,\nso the minor version bump is proposed as 2.3.0.\nThis release also includes fixes for Sphinx 4.\n\nChange-Id: Ib616837841a5d470b4d3fb1ef710fea4c8ead4c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5bfefe65729f9f047576b8929d151173bcf6b837"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5bfefe65729f9f047576b8929d151173bcf6b837"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
