)]}'
{"id":"openstack%2Freleases~794738","triplet_id":"openstack%2Freleases~master~I5d1c255914efb2ab0579a959a8d78a69ea3e64c5","project":"openstack/releases","branch":"master","topic":"puppet-openstack-xena-m1","hashtags":[],"change_id":"I5d1c255914efb2ab0579a959a8d78a69ea3e64c5","subject":"Puppet OpenStack Xena Milestone 1","status":"MERGED","created":"2021-06-04 08:30:04.000000000","updated":"2021-07-02 13:47:36.000000000","submitted":"2021-07-02 13:47:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"794738-puppet-openstack-xena-m1","meta_rev_id":"45dfd316b1435e3e4f961920df89b7bf07cc2f48","_number":794738,"virtual_id_number":794738,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-02 13:47:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2021-06-28 10:16:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2021-07-02 13:38:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2021-07-02 13:38:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-06-28 08:45:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-04 08:31:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-07 09:03:16.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2021-06-16 08:41:02.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-06-16 14:02:22.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-06-21 11:53:54.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"c6753054a7ec185c0ba831d507e1c775a90619f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-06-04 08:30:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"960d34e22a8fb21138283e4fee79f4442920f471","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 08:31:00.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"15cb253df3c189fd54bc7f5dafbe5ee531e19bf5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 08:40:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/623d8e791b8a4a02bd294bd1b748376f : SUCCESS in 6m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0417e9653ffe49e282afef1248593bb8 : FAILURE in 8m 28s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ce3d916344624cb2857668271ec0361b : SUCCESS in 8m 14s","accounts_in_message":[],"_revision_number":1},{"id":"fc2e6d2b17c92264969379df07b44014ac256608","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-06-05 02:36:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1615115af17a0da4df1384b3d6f66435e27701fc","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 02:37:00.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"3465b190b6ebf8b0a426698c92919d71049e60b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 02:45:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47071d7899124f11a0c257256d44b5c6 : SUCCESS in 6m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8f2a486421d54faca1b857a34a5ef99b : SUCCESS in 7m 52s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/8a7160dc190d44819c2c0348ac2b0163 : SUCCESS in 8m 14s","accounts_in_message":[],"_revision_number":2},{"id":"226ce2cac22689d34e05ac94648b51b0e90ca650","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-06-07 09:03:16.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1171d1ba996c3c120e0f39e1fa6284a1e3dcce48","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-06-11 07:58:04.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8c6eb623c2025885ac817f18047208d6110e589d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 08:12:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c1ec731839248adbfbd6aed33a2d493 : SUCCESS in 7m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/158ee01c09b4499b92ead4011de714c2 : FAILURE in 7m 19s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e6eac0b6f55a4e268d569f0a809e930d : SUCCESS in 8m 11s","accounts_in_message":[],"_revision_number":2},{"id":"2d788a3225a7808b25d7cfc02316b9d63fdaedfa","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-06-14 09:29:32.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"cc3ed9bb1d948585379e38628770c1f983a0784e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 09:43:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c67e758db6f4cc8911c1da5f1c9847e : SUCCESS in 6m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/42626d92e7424f80b89ec849de2fc969 : SUCCESS in 8m 05s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e05e973a8aa74429b42aeed22ef8a3c1 : SUCCESS in 7m 36s","accounts_in_message":[],"_revision_number":2},{"id":"355735f98936fee4e4a7b7c5ca8821414cd76f38","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-06-16 14:02:22.000000000","message":"Patch Set 2: Code-Review+1\n\nI see that though these deliverables are cycle-with-intermediary they were mostly doing major version bump between releases. And this patch follows the same rule. However, cycle-with-intermediary requires only to bump minor version. And \u0027reno semver\u0027 suggests the same, so only minor version bumps instead of major version bump in some cases here.\n\nI guess that is possible to bump major versions despite that the model is cycle-with-intermediary... Anyway, I\u0027m giving only +1 now and let you respond, as I\u0027m not 100% sure...","accounts_in_message":[],"_revision_number":2},{"id":"326f5edd08e7d187cba80a145f07ec22aba78a7f","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-06-21 01:34:38.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2: Code-Review+1\n\u003e \n\u003e I see that though these deliverables are cycle-with-intermediary they were mostly doing major version bump between releases. And this patch follows the same rule. However, cycle-with-intermediary requires only to bump minor version. And \u0027reno semver\u0027 suggests the same, so only minor version bumps instead of major version bump in some cases here.\n\u003e \n\u003e I guess that is possible to bump major versions despite that the model is cycle-with-intermediary... Anyway, I\u0027m giving only +1 now and let you respond, as I\u0027m not 100% sure...\n\nThe puppet openstack project has always been upgraded like this before","accounts_in_message":[],"_revision_number":2},{"id":"41b3fbedac29975a7136f98c96c9000291547651","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-21 11:46:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n\u003e Patch Set 2: Code-Review+1\n\u003e \n\u003e I see that though these deliverables are cycle-with-intermediary they were mostly doing major version bump between releases. And this patch follows the same rule. However, cycle-with-intermediary requires only to bump minor version. And \u0027reno semver\u0027 suggests the same, so only minor version bumps instead of major version bump in some cases here.\n\u003e \n\u003e I guess that is possible to bump major versions despite that the model is cycle-with-intermediary... Anyway, I\u0027m giving only +1 now and let you respond, as I\u0027m not 100% sure...\n\nUsually they start new cycles with a major bump, examples:\n\n- https://opendev.org/openstack/releases/src/branch/master/deliverables/train/puppet-aodh.yaml#L12\n- https://opendev.org/openstack/releases/src/branch/master/deliverables/ussuri/puppet-aodh.yaml#L12\n- https://opendev.org/openstack/releases/src/branch/master/deliverables/wallaby/puppet-aodh.yaml#L12","accounts_in_message":[],"_revision_number":2},{"id":"dfb79bdade8a2da282d31f2a5f2448a3abd46fc5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 11:46:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b8666f9438d0c4232b020861b3f99778aee15678","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-06-21 11:53:54.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a1af725404247cc9a36932fa78b622c617e3fa3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 11:59:58.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd860a1aa52f4deca9d7eef069ae42d7 : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7e08ea0bd174dc18053ba84c0da1952 : SUCCESS in 7m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/957d54019e4c471992e7f57628098377 : FAILURE in 8m 16s","accounts_in_message":[],"_revision_number":2},{"id":"3daff04d2b849411cfab3442e623df8fa678bc2c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-21 12:21:00.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3e2575e2e8f6e2628fc87f480d6c098bad5c98f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 12:35:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/026601f835024d478f95fd45c29f8161 : SUCCESS in 6m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/efae8e4c2b9744d0a0aeee1576b2d41d : FAILURE in 7m 46s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/3fd07538d19448f39c8ce49685f18980 : SUCCESS in 8m 30s","accounts_in_message":[],"_revision_number":2},{"id":"e0c6f92fdbc83f3693132317323dc1302f6d694b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-22 12:32:07.000000000","message":"Patch Set 2: Code-Review-1 -Workflow\n\nYour metadata don\u0027t match the given versions.\n\nExample with puppet-barbican:\n\nhttps://opendev.org/openstack/puppet-barbican/commit/0d364bc202cc1ff69522a32efd47681b66c0aa1b\n\nHere you ask for a major version (19.0.0) but your metadata request a minor version (19.1.0).\n\nThey should be aligned to allow us to continue.\nEither:\n1) you downgrade them on the project side (puppet-*) and you keep the version as they are\n2) you update the versions here to match your metadata\n\nHowever the second option will skip the major version and you\u0027ll move from 18.4.0 to 19.1.0 (e.g  puppet-barbican will skip 19.0.0). I really prefer the first option.\n\nI -1\u0027d to highlight the need of changes.\n\nI think that we didn\u0027t see early this validation issue due to a bug that have been fixed few days ago [1] in our validation module. We apologies for the side effects. \n\n[1] https://opendev.org/openstack/releases/commit/9504e96b5df4bd26d87c36681fecca5b27872a9d","accounts_in_message":[],"_revision_number":2},{"id":"af2e0a2e814458fac2262b30e16adacccf4fe424","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-06-28 08:44:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bb8caccbea503e96434082e4c0ef09eb266e6129","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 08:45:23.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"23121e2f2521bf4745cacfb1828d22f84ea4a701","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 08:59:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df474fcbe10040d89dd0e525240a6462 : SUCCESS in 9m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b865dc199cd449098723ea13b7a7c0fe : SUCCESS in 9m 34s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9b3dbf6f2f644520aac9da42d921852e : SUCCESS in 9m 09s","accounts_in_message":[],"_revision_number":3},{"id":"4388e76b23b64091aef27decf1873162e3579ca2","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-06-28 10:16:07.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a4777da6ea7375a020b46c8f63f9304cc97eba7f","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-07-02 13:38:39.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nhashes are corrected and validators passing. adding back +W.","accounts_in_message":[],"_revision_number":3},{"id":"d52c9de381357fd078d064c4f57bb622df109f18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 13:38:55.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2b25385ed35808442158ab17f23ac05bdaad9e6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 13:47:34.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89a2cbcdb38f4845b674b77a17e5f7aa : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c59d83fd795f48dbbdea0165896c86ba : SUCCESS in 6m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b60fe589d4f642779bdeded2febd7782 : SUCCESS in 7m 22s","accounts_in_message":[],"_revision_number":3},{"id":"45dfd316b1435e3e4f961920df89b7bf07cc2f48","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 13:47:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"de234095cbe2ad11c1c0bfb9c006d8d2a3685903","revisions":{"aad7ff2bb40960b2fde6095f3682240fe802fdea":{"kind":"REWORK","_number":1,"created":"2021-06-04 08:30:04.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/38/794738/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/38/794738/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/794738/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/794738/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/794738/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/38/794738/1"}}},"commit":{"parents":[{"commit":"f45222d69d323e1e52437f7b9a9e160ae02a7f2c","subject":"Merge \"[barbican] Transition Train to EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f45222d69d323e1e52437f7b9a9e160ae02a7f2c"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-06-04 08:29:02.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-06-04 08:29:02.000000000","tz":480},"subject":"Puppet OpenStack Xena Milestone 1","message":"Puppet OpenStack Xena Milestone 1\n\nChange-Id: I5d1c255914efb2ab0579a959a8d78a69ea3e64c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/aad7ff2bb40960b2fde6095f3682240fe802fdea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/aad7ff2bb40960b2fde6095f3682240fe802fdea"}]},"branch":"refs/heads/master"},"235e3663c66ffb7a0dd71e05aec37e6d48d836ce":{"kind":"REWORK","_number":2,"created":"2021-06-05 02:36:00.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/38/794738/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/38/794738/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/794738/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/794738/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/794738/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/38/794738/2"}}},"commit":{"parents":[{"commit":"f45222d69d323e1e52437f7b9a9e160ae02a7f2c","subject":"Merge \"[barbican] Transition Train to EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f45222d69d323e1e52437f7b9a9e160ae02a7f2c"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-06-04 08:29:02.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-06-05 02:35:39.000000000","tz":480},"subject":"Puppet OpenStack Xena Milestone 1","message":"Puppet OpenStack Xena Milestone 1\n\nChange-Id: I5d1c255914efb2ab0579a959a8d78a69ea3e64c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/235e3663c66ffb7a0dd71e05aec37e6d48d836ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/235e3663c66ffb7a0dd71e05aec37e6d48d836ce"}]},"branch":"refs/heads/master"},"de234095cbe2ad11c1c0bfb9c006d8d2a3685903":{"kind":"REWORK","_number":3,"created":"2021-06-28 08:44:07.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/38/794738/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/38/794738/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/794738/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/794738/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/38/794738/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/38/794738/3"}}},"commit":{"parents":[{"commit":"f45222d69d323e1e52437f7b9a9e160ae02a7f2c","subject":"Merge \"[barbican] Transition Train to EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f45222d69d323e1e52437f7b9a9e160ae02a7f2c"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-06-04 08:29:02.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-06-28 08:43:47.000000000","tz":480},"subject":"Puppet OpenStack Xena Milestone 1","message":"Puppet OpenStack Xena Milestone 1\n\nChange-Id: I5d1c255914efb2ab0579a959a8d78a69ea3e64c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/de234095cbe2ad11c1c0bfb9c006d8d2a3685903"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/de234095cbe2ad11c1c0bfb9c006d8d2a3685903"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
