)]}'
{"id":"openstack%2Freleases~795672","triplet_id":"openstack%2Freleases~master~I73f46b385d8ccba316c9588320a8306901433d82","project":"openstack/releases","branch":"master","topic":"ocata-eol","hashtags":[],"change_id":"I73f46b385d8ccba316c9588320a8306901433d82","subject":"[requirements] Transition Ocata to End of Life","status":"ABANDONED","created":"2021-06-09 19:18:03.000000000","updated":"2022-05-24 20:09:53.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bf894f4b8497f08215408f29f970ef91288cff44","_number":795672,"virtual_id_number":795672,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-06-09 19:26:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2021-08-05 15:34:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-09 19:18:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-09 19:20:16.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-06-09 19:26:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-21 13:13:02.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"f4e61c58a4da50d3865503722fbe18b2fbdf59b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-06-09 19:18:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0155bdcea1e1a27064f45871ac2045682bc8afd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 19:26:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2116ed0ad7b7470a9c4d7edab0e31188 : SUCCESS in 7m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/98bb45cf83194308bfffa5eb3ac02bd5 : SUCCESS in 5m 56s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ada6658258244eec985bbcf6cb1774d6 : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":1},{"id":"1ee404feaba452426fa8f47aaa52b32545594127","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-21 13:13:02.000000000","message":"Patch Set 1: Code-Review+1\n\nAs tags exists (1.0.0 and ocata-em), I think that ocata-eol make sense too.\n\nWaiting for Matthew\u0027s confirmation.","accounts_in_message":[],"_revision_number":1},{"id":"b21e9d47295499b7332ff21f686485c8e0d90ffa","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-06-21 13:46:29.000000000","message":"Patch Set 1:\n\nThere are some *-eol tag on requirements (see tags at: [1]).\n(The interesting thing is that newton-eol is missing)\n\n[1] https://opendev.org/openstack/requirements/src/branch/master","accounts_in_message":[],"_revision_number":1},{"id":"006d06803861d290e80c3f60263f74311cd42d3a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-28 10:49:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4c69e7b651a51c67e77566953a0cebbab5e23c46","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-28 10:49:31.000000000","message":"Patch Set 1:\n\nLet\u0027s go with that","accounts_in_message":[],"_revision_number":1},{"id":"5f246f0ed3e3b9d17de5a6d10b0821e66adc9634","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-05 15:16:36.000000000","message":"Patch Set 1:\n\n@Matthew: Please can you have a look, we are not sure how it would affect the not-yet-ocata-EOLd repositories (and their future transition to EOL).\n\nThanks by advance","accounts_in_message":[],"_revision_number":1},{"id":"91ca169844589b01e2f33cb1d57d5b85e1f6dd3d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-05 15:22:17.000000000","message":"Patch Set 1:\n\nI think we need to make sure there are no gate jobs using ocata running anywhere.  If we pull the branch the jobs will fail.  (I guess a beagle search is in order)","accounts_in_message":[],"_revision_number":1},{"id":"2ed4c5fed1fe9e7d0f2b5b0762969bed6a1327b9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-05 15:31:09.000000000","message":"Patch Set 1: -Code-Review\n\nthanks for your reply.","accounts_in_message":[],"_revision_number":1},{"id":"ef87dbc4a6a08be38cc8adff6c63f60ead0d6d8e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-05 15:34:50.000000000","message":"Patch Set 1:\n\nhttps://gist.github.com/a954c806611428b84b97f415d9c515a4","accounts_in_message":[],"_revision_number":1},{"id":"bf894f4b8497f08215408f29f970ef91288cff44","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-05-24 20:09:53.000000000","message":"Abandoned\n\nocata meanwhile transitioned to EOL and similar patch was merged instead of this :S https://review.opendev.org/c/openstack/releases/+/838568","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9ce0aa9d8045b71ddc5895d2a6f8b140f8408d12","revisions":{"9ce0aa9d8045b71ddc5895d2a6f8b140f8408d12":{"kind":"REWORK","_number":1,"created":"2021-06-09 19:18:03.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/72/795672/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/72/795672/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/795672/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/795672/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/795672/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/72/795672/1"}}},"commit":{"parents":[{"commit":"9ccdd595bed17a3365a1282588bab71447d4e39a","subject":"Merge \"New neutronclient release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9ccdd595bed17a3365a1282588bab71447d4e39a"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-06-09 19:17:42.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-06-09 19:17:42.000000000","tz":120},"subject":"[requirements] Transition Ocata to End of Life","message":"[requirements] Transition Ocata to End of Life\n\nThis transition the Ocata branch to End of Life. The last patch of the\nbranch will be tagged with ocata-eol tag. The ocata branch cannot be\nused anymore and will be deleted if this patch merges.\n\nThis is needed as stable/ocata is not actively maintained in the recent\nperiod and thus gate is broken due to job failures. By removing these\nbranches, infra resources will be freed up, too.\n\nPlease try to identify any zuul job, that is defined outside of the\nrepositories in this patch (for example in openstack-zuul-jobs, etc.)\nand won\u0027t be used anymore if stable/ocata is deleted, and propose a job\nremoval patch for them.\n\nPlease +1 if the team is ready for us to proceed with this transition,\nor -1 if there are still some activity on the branch and the team wants\nto continue to maintain it.\n\nChange-Id: I73f46b385d8ccba316c9588320a8306901433d82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9ce0aa9d8045b71ddc5895d2a6f8b140f8408d12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9ce0aa9d8045b71ddc5895d2a6f8b140f8408d12"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
