)]}'
{"id":"openstack%2Freleases~795681","triplet_id":"openstack%2Freleases~master~I19a14d8c4c571e098a1b03d00652de7774575123","project":"openstack/releases","branch":"master","topic":"ocata-eol","hashtags":[],"change_id":"I19a14d8c4c571e098a1b03d00652de7774575123","subject":"[tripleo] Transition Ocata to End of Life","status":"ABANDONED","created":"2021-06-09 19:32:45.000000000","updated":"2021-06-17 12:30:57.000000000","total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"dc8abb4c73230739c44a44989e829772538cffc8","_number":795681,"virtual_id_number":795681,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-06-09 19:41:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"disliked":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"date":"2021-06-17 06:48:10.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":-1,"date":"2021-06-16 19:05:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2021-06-09 20:17:15.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"date":"2021-06-17 11:18:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"date":"2021-06-17 11:19:55.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-09 19:33:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-09 19:41:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-09 20:17:15.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-06-09 20:20:10.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2021-06-14 11:57:24.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-06-16 20:14:46.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"45887091eecae6758a60cbaf1c4c4decb11632d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-06-09 19:32:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"247b5058bc1c9e3e6c75f1015ba231c46c2a4fc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 19:41:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f853436864eb4a7dbcf1eb2874c4e4e0 : SUCCESS in 7m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e42a06a9b7374959ab4584dad894f899 : SUCCESS in 5m 42s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/fff1ae54579f4b7d868ab50eef15b7ae : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":1},{"id":"2bbb690aba425124d043439b477bfe73a403e346","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-06-09 20:17:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d7e15707aa1ce3c67c8260f966cc580765b93e81","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-06-10 05:24:38.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9d4501d7c9d79a7b7fe4176273792b91525db5f7","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 05:25:35.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"11d093799c4b9807f9653a79600393d9b194e36f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-14 11:57:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8c07c04c0123d3c8035b5198594ccffd28f779b2","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-06-16 19:05:23.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nI don\u0027t know what to do with this patch, as:\n* there\u0027s no ocata-eol tag but\n* there\u0027s no stable/ocata branch either\nmaybe we can just simply abandon this patch and we can ignore that ocata-eol tag does not exists on this repository...","accounts_in_message":[],"_revision_number":1},{"id":"7921ba39f684d4a72667bbf00788357feb21c293","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-06-16 19:05:32.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"bde799bcc17a33064b32ed6710763fc7aac2e765","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-16 20:11:51.000000000","message":"Patch Set 1: -Code-Review\n\nYes the branch seems to not has been created, even in this file.\n\nI would suggest to simply abandon it.","accounts_in_message":[],"_revision_number":1},{"id":"2a6b861b742591e2518260e4cbd33d33b5e80512","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-16 20:14:46.000000000","message":"Patch Set 1:\n\nLet\u0027s see what thierry think about it since he is the most oldest release team member, maybe he will have some insights to share with us.","accounts_in_message":[],"_revision_number":1},{"id":"e48fdb10ab1788549650ef8db669c82344f57454","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-06-17 06:48:10.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)\n\napologies for the confusion I should have noticed this on my earlier visit","accounts_in_message":[],"_revision_number":1},{"id":"f13af559eeeee1b60c5d228074b639263a158884","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 06:49:36.000000000","message":"Patch Set 1: -PTL-Approved","accounts_in_message":[],"_revision_number":1},{"id":"9fafa91998e942a6f746cd437a8a8741b9fbce0d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-17 11:18:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dc8abb4c73230739c44a44989e829772538cffc8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-06-17 12:30:57.000000000","message":"Abandoned\n\nThanks for the replies. Abandoning according to discussion in comments.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bc2393a8b52fc46e7b28534626a365788e2ac382","revisions":{"bc2393a8b52fc46e7b28534626a365788e2ac382":{"kind":"REWORK","_number":1,"created":"2021-06-09 19:32:45.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/81/795681/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/81/795681/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/795681/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/795681/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/795681/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/81/795681/1"}}},"commit":{"parents":[{"commit":"9ccdd595bed17a3365a1282588bab71447d4e39a","subject":"Merge \"New neutronclient release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9ccdd595bed17a3365a1282588bab71447d4e39a"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-06-09 19:27:15.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-06-09 19:27:15.000000000","tz":120},"subject":"[tripleo] Transition Ocata to End of Life","message":"[tripleo] Transition Ocata to End of Life\n\nThis transition the Ocata branch to End of Life. The last patch of the\nbranch will be tagged with ocata-eol tag. The ocata branch cannot be\nused anymore and will be deleted if this patch merges.\n\nThis is needed as stable/ocata is not actively maintained in the recent\nperiod and thus gate is broken due to job failures. By removing these\nbranches, infra resources will be freed up, too.\n\nPlease try to identify any zuul job, that is defined outside of the\nrepositories in this patch (for example in openstack-zuul-jobs, etc.)\nand won\u0027t be used anymore if stable/ocata is deleted, and propose a job\nremoval patch for them.\n\nPlease +1 if the team is ready for us to proceed with this transition,\nor -1 if there are still some activity on the branch and the team wants\nto continue to maintain it.\n\nChange-Id: I19a14d8c4c571e098a1b03d00652de7774575123\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bc2393a8b52fc46e7b28534626a365788e2ac382"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bc2393a8b52fc46e7b28534626a365788e2ac382"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
