)]}'
{"id":"openstack%2Freleases~795836","triplet_id":"openstack%2Freleases~master~I19561a49544d1d855d41fe50e09974f6c30267b0","project":"openstack/releases","branch":"master","topic":"python-glanceclient","hashtags":[],"change_id":"I19561a49544d1d855d41fe50e09974f6c30267b0","subject":"python-glanceclient 3.4.0","status":"MERGED","created":"2021-06-10 15:34:05.000000000","updated":"2021-06-15 12:27:59.000000000","submitted":"2021-06-15 11:59:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"795836-python-glanceclient","meta_rev_id":"0d0cbb8f1f8b6beabadaf593d9f6729affb7521f","_number":795836,"virtual_id_number":795836,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-15 11:58:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-06-14 11:18:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2021-06-15 11:23:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2021-06-15 11:49:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-10 15:35:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-10 15:41:56.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-06-10 15:42:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-10 16:20:09.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2021-06-11 12:04:02.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-06-11 16:16:38.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-06-15 11:23:07.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"d7c61c65f4b3d82a37779e5cba03319055941c2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-06-10 15:34:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"652f833bdd44ea417f0d6040618620426530f381","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 15:42:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69f6ce66f98248e78cdec35c40a4ac59 : SUCCESS in 6m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2f1f2a42ef1f4915ad3ec10e18594052 : SUCCESS in 5m 15s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b99332c711564ce8b8203c358f953b62 : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":1},{"id":"220c9a73f6adb75b590837042f4bf36ad9cb8210","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-06-11 04:42:21.000000000","message":"Patch Set 1: Code-Review+1\n\nHash is correct!!","accounts_in_message":[],"_revision_number":1},{"id":"0df9f1d37188aa393f7d5471668f20c1c1064268","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 04:43:12.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"ae6d88077ea33c06d30c71b927f239054a19e36a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-06-11 12:04:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e769f5992f68d39923abac83a38b6ffad0f30079","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-06-11 16:16:38.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"11681ce21f98198a8bfdef73e926c3915301554c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-06-11 16:21:01.000000000","message":"Patch Set 1: -Code-Review\n\nOops, totally right Elod. I overlooked that part.\n\nI thought we had a job that checked for this exact scenario.","accounts_in_message":[],"_revision_number":1},{"id":"b6233b01329f0866980067abf6fe9c0f5a9b1fb6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-06-11 16:21:31.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: -Code-Review\n\u003e \n\u003e Oops, totally right Elod. I overlooked that part.\n\u003e \n\u003e I thought we had a job that checked for this exact scenario.\n\nAnd/or that the new release script just handled that.","accounts_in_message":[],"_revision_number":1},{"id":"537f8898e1964bfb3d16c5ca0921e70ad572d214","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-06-14 11:07:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"62355547faa558c8c6ae0e960d2044f83cac402b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-06-14 11:08:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2f37729f913beaece854f64c3d2d3b6edbed8b92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 11:16:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8ac66d6669a4177bc300236cd3c32ef : SUCCESS in 7m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6e5f2101d2e34b639e0dfa1eb9b04a9f : SUCCESS in 5m 53s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7a22d43b91714baa9c3ed26b38f0e96b : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":2},{"id":"2538c505cb628da3e42d5843ca44e1d4ac112058","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-06-14 11:18:18.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"cd52d5f51af3cb7612a2212a569ddaa058ed0e0e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-15 11:23:07.000000000","message":"Patch Set 2: Code-Review+2\n\n\u003e Patch Set 1: -Code-Review\n\u003e \n\u003e Oops, totally right Elod. I overlooked that part.\n\u003e \n\u003e I thought we had a job that checked for this exact scenario.\n\nYes you\u0027re right, normally we had a related check, I wonder why it doesn\u0027t warned us here, I\u0027ll have a look locally.","accounts_in_message":[],"_revision_number":2},{"id":"26e3805bc07776adea05c8633715e9df29f8a675","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-15 11:49:12.000000000","message":"Patch Set 2: Workflow+1\n\nBy using PS1 locally I can\u0027t trigger the warn about the need of a minor version either.\n\n+W\u0027ing these changes to move further with glanceclient, however, we need further investigations to determine why this check don\u0027t work as expected.","accounts_in_message":[],"_revision_number":2},{"id":"b7a07555b7db4fae053b1826cdc8c49a72f5e8c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 11:49:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e02a934164dc9af2f9fd299bddc87e7bfa5bd5a2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-15 11:54:21.000000000","message":"Patch Set 2:\n\nI just saw that the check is triggered by the `new_release` command, so, normally if ptl/liaison use this convenience tool they will be informed, however if they don\u0027t use it they won\u0027t be informed and in this case our validation job should check that too.\n\nExample with oslo.middleware which isn\u0027t yet released within Xena:\n```\n$ tools/new_release.sh xena oslo.middleware bugfix\n   INFO: creating temporary files in /tmp/releases-57jmeejy\nusage: new-release [-h] [-v] [-i] [--no-cleanup] [--force] [--debug] [--stable-branch] [--intermediate-branch]\n                   series deliverable {bugfix,feature,major,milestone,rc,procedural,eol,em,releasefix}\nnew-release: error: The first release for a series must be at least a feature release to allow for stable releases from the previous series.\n```","accounts_in_message":[],"_revision_number":2},{"id":"2a69a35e9a53df54765bc1d6c85a805e02b4d337","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 11:58:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7172e627b794a069b27675131d7a144 : SUCCESS in 8m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5932f8b563d24dd8ae6ee66ce93df59d : SUCCESS in 7m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fec0e7e6485f41d7bf3cfafd24851140 : SUCCESS in 6m 20s","accounts_in_message":[],"_revision_number":2},{"id":"27e570efebcd7fc0a3fdbe3553cd52f90d39ee0e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 11:59:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"0d0cbb8f1f8b6beabadaf593d9f6729affb7521f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-15 12:27:59.000000000","message":"Patch Set 2:\n\nValidation will be fixed by https://review.opendev.org/c/openstack/releases/+/796456","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2772b43b96d712f7defb40eef26081fd32bc5d8c","revisions":{"bb16c858c2182c224bc9ebc4e829dfcfca9486e7":{"kind":"REWORK","_number":1,"created":"2021-06-10 15:34:05.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/36/795836/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/36/795836/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/36/795836/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/36/795836/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/36/795836/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/36/795836/1"}}},"commit":{"parents":[{"commit":"156b6df7b599297a5c8abc5c49481b2c40b3b304","subject":"Merge \"Release oslo.limit for Xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/156b6df7b599297a5c8abc5c49481b2c40b3b304"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-06-10 15:33:04.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-06-10 15:33:04.000000000","tz":60},"subject":"python-glanceclient 3.3.1","message":"python-glanceclient 3.3.1\n\nUncap PrettyTable, plus some mechanical stable/wallaby-related changes.\n\nChange-Id: I19561a49544d1d855d41fe50e09974f6c30267b0\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bb16c858c2182c224bc9ebc4e829dfcfca9486e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bb16c858c2182c224bc9ebc4e829dfcfca9486e7"}]},"branch":"refs/heads/master"},"2772b43b96d712f7defb40eef26081fd32bc5d8c":{"kind":"REWORK","_number":2,"created":"2021-06-14 11:07:11.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/36/795836/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/36/795836/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/36/795836/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/36/795836/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/36/795836/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/36/795836/2"}}},"commit":{"parents":[{"commit":"156b6df7b599297a5c8abc5c49481b2c40b3b304","subject":"Merge \"Release oslo.limit for Xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/156b6df7b599297a5c8abc5c49481b2c40b3b304"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-06-10 15:33:04.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-06-14 11:07:01.000000000","tz":60},"subject":"python-glanceclient 3.4.0","message":"python-glanceclient 3.4.0\n\nUncap PrettyTable, plus some mechanical stable/wallaby-related changes.\nThis is a new minor release since we\u0027re in a new series.\n\nChange-Id: I19561a49544d1d855d41fe50e09974f6c30267b0\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2772b43b96d712f7defb40eef26081fd32bc5d8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2772b43b96d712f7defb40eef26081fd32bc5d8c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
