)]}'
{"id":"openstack%2Freleases~796456","triplet_id":"openstack%2Freleases~master~I5d9024528d08487a7582e0a0e73576b22708a6cf","project":"openstack/releases","branch":"master","topic":"fix-first-series-release-check","hashtags":[],"change_id":"I5d9024528d08487a7582e0a0e73576b22708a6cf","subject":"Fix validation to ensure that first release of series isn\u0027t a bugfix","status":"MERGED","created":"2021-06-15 12:27:14.000000000","updated":"2021-06-21 10:09:54.000000000","submitted":"2021-06-21 10:09:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"796456-fix-first-series-release-check","meta_rev_id":"b280aed202f24144dc251e3d8dbce491f1ed2322","_number":796456,"virtual_id_number":796456,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-21 10:09:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-06-18 14:45:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-06-21 09:52:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-06-21 09:52:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-15 12:43:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-16 13:34:02.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-06-16 14:33:08.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"CC"},{"updated":"2021-06-17 11:43:57.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"6e9e7db45fce0130110354c2bae1591298edce37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-15 12:27:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"861b7b7f577e68857ea539bb8e58e353533bf966","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-15 12:32:47.000000000","message":"Patch Set 1:\n\nI wonder if it\u0027s not the decorator `skip_existing_tags` that\u0027s broken.\nAt first glance it make sense to ignore existing tag with this test, however, this decorator excluded the new created version (e.g https://review.opendev.org/c/openstack/releases/+/795836).\n\nhttps://opendev.org/openstack/releases/commit/cdf80e14694afe7bf1697a1b82fd64de2af6cc48","accounts_in_message":[],"_revision_number":1},{"id":"7f22b839c5ed9ccb2f1569ef01fcf55888410b6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-15 12:34:46.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7f76c78e29f18a16fad19c49825e71c19bff5b88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 12:43:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d776e2e689ea4f389ab7f44ca13ffe5e : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b9462f7a45f4c4a87730d2b2f9d8eb2 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aeab0ce79f054bb48fbd83f7413d1103 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1291d81980cc41d093702bb05a785557 : SUCCESS in 6m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3a969ca8f1af4ea5a19a71eb99466ebd : SUCCESS in 4m 46s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/faaf7919da4c469bb4e19949622706d7 : SUCCESS in 5m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/32887666a47447b08bbb747efa36676d : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":2},{"id":"22313e75beb28cd0ea35d6d1b467956aa91cd623","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-15 13:41:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"aa5945db3a96a77a64e8ff75deb0ff5e71315f45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 13:53:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e627c581ff99466da3d748826d589f59 : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc1213df91a14d16a7e64f81368af5e3 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/897dd1a30d1340e3bd6238fa2f6259b8 : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0530896e066c42e48433fdf4fe3d6162 : SUCCESS in 7m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6238f38bb2b84b55b5b180252d80bfb4 : SUCCESS in 5m 33s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/a3f7e288e79142baa8a9f6fdb9344074 : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c568aa69ce9d442c8adcfefd7a49bd89 : SUCCESS in 5m 53s","accounts_in_message":[],"_revision_number":3},{"id":"c394786c84c66d8f4aae753efdfd0ae0da1ee470","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-15 14:13:04.000000000","message":"Patch Set 3:\n\nPoC available at https://review.opendev.org/c/openstack/releases/+/796485","accounts_in_message":[],"_revision_number":3},{"id":"f4f64c469b7c74644966ff7ce676603a7f823294","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-15 14:54:24.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"4a323d0f936faac7d81f32db8d85fdd87d8b5503","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 15:09:29.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e82f5d688f5142c2ab2785716a162f53 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9494072827b6482390ff249a9e1483e4 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a8a65cc3e6e4f52aa4b5cd901eebdd1 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dba8a55829794513a2b1b278ab780ed8 : SUCCESS in 8m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2f8fcc9d9cd140cbad643558ff3bc853 : SUCCESS in 5m 52s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c2e9503e74cf49f4ad09a521412d8d43 : SUCCESS in 4m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/71f61e03419e4702860f8b2b55894b24 : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":4},{"id":"b919ae1d722ef321d0bc0ec96d89979d96bc71e5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-06-16 13:34:02.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"c50d84eb9691e6bfed972f3ef9e8c1bc664ca284","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-06-16 14:33:08.000000000","message":"Patch Set 4:\n\nI think this should be worth a test case, so that we don\u0027t reintroduce it accidentally in the future","accounts_in_message":[],"_revision_number":4},{"id":"b10de3de168ce1ee001b244df6d3f206b93bd900","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-16 15:36:50.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"92608484ce04d2c3d3ed56ba4c69031c27f59474","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 15:50:06.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bf3a968f5da491795839198a7e8d2b0 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf860fdecbca4e8b950c45fc918ca96d : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d55cc08411904447a64d4eb8bb08190b : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91198bd9fee445e3840412127cb70a39 : SUCCESS in 7m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bece754bcc344038bd7ab3d92b706cc1 : SUCCESS in 5m 44s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e429f7c6878b4e5c991d2dcb0423b8b6 : SUCCESS in 6m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/42f1119e47cf4ce3a0e6a82d045600f9 : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":5},{"id":"eebfbea7070c314c9b634a81507756105cb7986c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-17 11:36:37.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"aaa5f0bda09bcc8d4e93fb610bf6a6733e41f784","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-17 11:40:57.000000000","message":"Uploaded patch set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"e8975351f09bf3c9cac9d0eae50047890e497b90","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-06-17 11:43:57.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"6ca24ca78276ef9c81a1d5d3c2660eca3166109c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-17 11:48:22.000000000","message":"Uploaded patch set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"bb7410ff89a34c6422cb02a688ccb404315fbe34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 11:58:38.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/962e25294c374c529ced01ae24b7d1da : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/111fda3fbd2749e9b3cae412a6f1d770 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0033d30dcd3c48ed83ad9c2733507d0f : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d65d445b6fa453c9e19470a361931e8 : SUCCESS in 6m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/646b33a37f3a4e01a498a56b0331d6b0 : SUCCESS in 5m 26s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b71ea4b1f83e4092a95736c552e22545 : SUCCESS in 4m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/07979c190b07484693fdf5cd38a08945 : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":8},{"id":"384b89f752e1163070a33e39331eb076d465fe9d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-06-18 14:45:58.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"5179a9ee1685db611078ef0978bdb16ad38f85d3","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-06-21 09:52:26.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"e4799afabad3033eecb4939d6c2d41e14b3b3325","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 09:52:42.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"8de5a10d0c687b64e85ced074f6692d1cdb58dd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 10:09:53.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37efb28927784285ab8c853c2f8c2a65 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5452b2e26e14663bd30666c7f0e479a : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c03769b6d5f14db4b88f7e4222bc6b31 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/845df92389124cd490c373e663abf972 : SUCCESS in 6m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2065d1ce6dbd4bdf88af6f2b3c8448cf : SUCCESS in 7m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6024005bc0e04e9daedf9d25496fc447 : SUCCESS in 8m 31s","accounts_in_message":[],"_revision_number":8},{"id":"b280aed202f24144dc251e3d8dbce491f1ed2322","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 10:09:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"9504e96b5df4bd26d87c36681fecca5b27872a9d","revisions":{"13876d836ed6c759523da721220c6b336b3b2458":{"kind":"REWORK","_number":1,"created":"2021-06-15 12:27:14.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/56/796456/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/56/796456/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/796456/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/796456/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/796456/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/56/796456/1"}}},"commit":{"parents":[{"commit":"156b6df7b599297a5c8abc5c49481b2c40b3b304","subject":"Merge \"Release oslo.limit for Xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/156b6df7b599297a5c8abc5c49481b2c40b3b304"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-15 12:22:49.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-15 12:26:23.000000000","tz":120},"subject":"Fix validation to ensure that first release of series isn\u0027t a bugfix","message":"Fix validation to ensure that first release of series isn\u0027t a bugfix\n\nWe recently observed that this check was skipped. Indeed it was\npossible to propose a bugfix version on the first release of a series\nwithout facing validation errors [1].\n\nThese changes reactivate this check.\n\n[1] https://review.opendev.org/c/openstack/releases/+/795836\n\nChange-Id: I5d9024528d08487a7582e0a0e73576b22708a6cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/13876d836ed6c759523da721220c6b336b3b2458"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/13876d836ed6c759523da721220c6b336b3b2458"}]},"branch":"refs/heads/master"},"cdcdfe7bcad31e1d522c4d9fc7cc90443f3ef37f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-06-15 12:34:46.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/56/796456/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/56/796456/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/796456/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/796456/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/796456/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/56/796456/2"}}},"commit":{"parents":[{"commit":"156b6df7b599297a5c8abc5c49481b2c40b3b304","subject":"Merge \"Release oslo.limit for Xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/156b6df7b599297a5c8abc5c49481b2c40b3b304"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-15 12:22:49.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-15 12:34:27.000000000","tz":120},"subject":"[WIP] Fix validation to ensure that first release of series isn\u0027t a bugfix","message":"[WIP] Fix validation to ensure that first release of series isn\u0027t a bugfix\n\nWe recently observed that this check was skipped. Indeed it was\npossible to propose a bugfix version on the first release of a series\nwithout facing validation errors [1].\n\nThese changes reactivate this check.\n\n[1] https://review.opendev.org/c/openstack/releases/+/795836\n\nChange-Id: I5d9024528d08487a7582e0a0e73576b22708a6cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cdcdfe7bcad31e1d522c4d9fc7cc90443f3ef37f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cdcdfe7bcad31e1d522c4d9fc7cc90443f3ef37f"}]},"branch":"refs/heads/master"},"2c4114729bc166e8992e7793935e9a61d1d842f6":{"kind":"REWORK","_number":3,"created":"2021-06-15 13:41:47.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/56/796456/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/56/796456/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/796456/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/796456/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/796456/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/56/796456/3"}}},"commit":{"parents":[{"commit":"156b6df7b599297a5c8abc5c49481b2c40b3b304","subject":"Merge \"Release oslo.limit for Xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/156b6df7b599297a5c8abc5c49481b2c40b3b304"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-15 12:22:49.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-15 13:39:04.000000000","tz":120},"subject":"Fix validation to ensure that first release of series isn\u0027t a bugfix","message":"Fix validation to ensure that first release of series isn\u0027t a bugfix\n\nWe recently observed that this check was skipped. Indeed it was\npossible to propose a bugfix version on the first release of a series\nwithout facing validation errors [1].\n\nThe described situation was allowed by checks that we introduced during\nWallaby [2]. Those was designed to ensure that we can build sdist from the\nproposed tag. Indeed, they temporarly created the new git tag to build sdist\nfrom it.  However we never delete this tag after our tests so this temporary\ntag is see as an existing tag and so that led us to the observed bug.\n\nWe check that the first version number of series isn\u0027t a bugfix after we tried\nto build the sdist, so the tag remain available, and so the check is skipped.\n\nWe should notice that surely impacted other checks in the same manner.\nHopefully it didn\u0027t have too much side effects.\n\nThese changes delete the temporary tag after usage to ensure to restore\nthe initial environment and to stop skipping next checks.\n\n[1] https://review.opendev.org/c/openstack/releases/+/795836\n[2] https://opendev.org/openstack/releases/commit/80652b523214ab7215928368246fbf726b78a645\n\nChange-Id: I5d9024528d08487a7582e0a0e73576b22708a6cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2c4114729bc166e8992e7793935e9a61d1d842f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2c4114729bc166e8992e7793935e9a61d1d842f6"}]},"branch":"refs/heads/master"},"3fa5ac14ffd88e2bfa05246691ab101116847981":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-06-15 14:54:24.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/56/796456/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/56/796456/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/796456/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/796456/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/796456/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/56/796456/4"}}},"commit":{"parents":[{"commit":"156b6df7b599297a5c8abc5c49481b2c40b3b304","subject":"Merge \"Release oslo.limit for Xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/156b6df7b599297a5c8abc5c49481b2c40b3b304"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-15 12:22:49.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-15 14:53:36.000000000","tz":120},"subject":"Fix validation to ensure that first release of series isn\u0027t a bugfix","message":"Fix validation to ensure that first release of series isn\u0027t a bugfix\n\nWe recently observed that this check was skipped. Indeed it was\npossible to propose a bugfix version on the first release of a series\nwithout facing validation errors [1].\n\nThe described situation was allowed by checks that we introduced during\nWallaby [2]. Those was designed to ensure that we can build sdist from the\nproposed tag. Indeed, they temporarly created the new git tag to build sdist\nfrom it.  However we never delete this tag after our tests so this temporary\ntag is see as an existing tag and so that led us to the observed bug.\n\nWe check that the first version number of series isn\u0027t a bugfix after we tried\nto build the sdist, so the tag remain available, and so the check is skipped.\n\nWe should notice that it surely impacted other checks in the same manner.\nHopefully it didn\u0027t have too much side effects.\n\nThese changes delete the temporary tag after usage to ensure to restore\nthe initial environment and to stop skipping next checks.\n\n[1] https://review.opendev.org/c/openstack/releases/+/795836\n[2] https://opendev.org/openstack/releases/commit/80652b523214ab7215928368246fbf726b78a645\n\nChange-Id: I5d9024528d08487a7582e0a0e73576b22708a6cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3fa5ac14ffd88e2bfa05246691ab101116847981"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3fa5ac14ffd88e2bfa05246691ab101116847981"}]},"branch":"refs/heads/master"},"93ac9ab74c4c6a979ca7743d25c1317f612c508c":{"kind":"REWORK","_number":5,"created":"2021-06-16 15:36:50.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/56/796456/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/56/796456/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/796456/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/796456/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/796456/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/56/796456/5"}}},"commit":{"parents":[{"commit":"156b6df7b599297a5c8abc5c49481b2c40b3b304","subject":"Merge \"Release oslo.limit for Xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/156b6df7b599297a5c8abc5c49481b2c40b3b304"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-15 12:22:49.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-16 15:36:31.000000000","tz":120},"subject":"Fix validation to ensure that first release of series isn\u0027t a bugfix","message":"Fix validation to ensure that first release of series isn\u0027t a bugfix\n\nWe recently observed that this check was skipped. Indeed it was\npossible to propose a bugfix version on the first release of a series\nwithout facing validation errors [1].\n\nThe described situation was allowed by checks that we introduced during\nWallaby [2]. Those was designed to ensure that we can build sdist from the\nproposed tag. Indeed, they temporarly created the new git tag to build sdist\nfrom it.  However we never delete this tag after our tests so this temporary\ntag is see as an existing tag and so that led us to the observed bug.\n\nWe check that the first version number of series isn\u0027t a bugfix after we tried\nto build the sdist, so the tag remain available, and so the check is skipped.\n\nWe should notice that it surely impacted other checks in the same manner.\nHopefully it didn\u0027t have too much side effects.\n\nThese changes delete the temporary tag after usage to ensure to restore\nthe initial environment and to stop skipping next checks.\n\n[1] https://review.opendev.org/c/openstack/releases/+/795836\n[2] https://opendev.org/openstack/releases/commit/80652b523214ab7215928368246fbf726b78a645\n\nChange-Id: I5d9024528d08487a7582e0a0e73576b22708a6cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/93ac9ab74c4c6a979ca7743d25c1317f612c508c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/93ac9ab74c4c6a979ca7743d25c1317f612c508c"}]},"branch":"refs/heads/master"},"455e34ae23f4cad5329ffb3b33d14c3344490e8f":{"kind":"REWORK","_number":6,"created":"2021-06-17 11:36:37.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/56/796456/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/56/796456/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/796456/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/796456/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/796456/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/56/796456/6"}}},"commit":{"parents":[{"commit":"156b6df7b599297a5c8abc5c49481b2c40b3b304","subject":"Merge \"Release oslo.limit for Xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/156b6df7b599297a5c8abc5c49481b2c40b3b304"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-15 12:22:49.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-17 11:36:26.000000000","tz":120},"subject":"Fix validation to ensure that first release of series isn\u0027t a bugfix","message":"Fix validation to ensure that first release of series isn\u0027t a bugfix\n\nWe recently observed that this check was skipped. Indeed it was\npossible to propose a bugfix version on the first release of a series\nwithout facing validation errors [1].\n\nThe described situation was allowed by checks that we introduced during\nWallaby [2]. Those was designed to ensure that we can build sdist from the\nproposed tag. Indeed, they temporarly created the new git tag to build sdist\nfrom it.  However we never delete this tag after our tests so this temporary\ntag is see as an existing tag and so that led us to the observed bug.\n\nWe check that the first version number of series isn\u0027t a bugfix after we tried\nto build the sdist, so the tag remain available, and so the check is skipped.\n\nWe should notice that it surely impacted other checks in the same manner.\nHopefully it didn\u0027t have too much side effects.\n\nThese changes delete the temporary tag after usage to ensure to restore\nthe initial environment and to stop skipping next checks.\n\n[1] https://review.opendev.org/c/openstack/releases/+/795836\n[2] https://opendev.org/openstack/releases/commit/80652b523214ab7215928368246fbf726b78a645\n\nChange-Id: I5d9024528d08487a7582e0a0e73576b22708a6cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/455e34ae23f4cad5329ffb3b33d14c3344490e8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/455e34ae23f4cad5329ffb3b33d14c3344490e8f"}]},"branch":"refs/heads/master"},"f2d061d685abd0313667346fc609009801560da2":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2021-06-17 11:40:57.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/56/796456/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/56/796456/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/796456/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/796456/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/796456/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/56/796456/7"}}},"commit":{"parents":[{"commit":"156b6df7b599297a5c8abc5c49481b2c40b3b304","subject":"Merge \"Release oslo.limit for Xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/156b6df7b599297a5c8abc5c49481b2c40b3b304"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-15 12:22:49.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-17 11:36:43.000000000","tz":120},"subject":"Fix validation to ensure that first release of series isn\u0027t a bugfix","message":"Fix validation to ensure that first release of series isn\u0027t a bugfix\n\nWe recently observed that this check was skipped. Indeed it was\npossible to propose a bugfix version on the first release of a series\nwithout facing validation errors [1].\n\nThe described situation was allowed by checks that we introduced during\nWallaby [2]. Those was designed to ensure that we can build sdist from the\nproposed tag. Indeed, they temporarly created the new git tag to build sdist\nfrom it.  However we never delete this tag after our tests so this temporary\ntag is see as an existing tag and so that led us to the observed bug.\n\nOur validation check that the first version number of series isn\u0027t a bugfix\nafter it tried to build the sdist from the new tag, so, without these\nchanges the new tag created previously remain available in the repo, and so\na couple of next checks are skipped, including the validation of the\nversion number, because these checks are decorated to ignore existing\ntags.\n\nWe should notice that it surely impacted other checks in the same manner.\nHopefully it didn\u0027t have too much side effects.\n\nThese changes delete the temporary tag after usage to ensure to restore\nthe initial environment and to stop skipping next checks.\n\n[1] https://review.opendev.org/c/openstack/releases/+/795836\n[2] https://opendev.org/openstack/releases/commit/80652b523214ab7215928368246fbf726b78a645\n\nChange-Id: I5d9024528d08487a7582e0a0e73576b22708a6cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f2d061d685abd0313667346fc609009801560da2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f2d061d685abd0313667346fc609009801560da2"}]},"branch":"refs/heads/master"},"9504e96b5df4bd26d87c36681fecca5b27872a9d":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2021-06-17 11:48:22.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/56/796456/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/56/796456/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/796456/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/796456/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/796456/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/56/796456/8"}}},"commit":{"parents":[{"commit":"156b6df7b599297a5c8abc5c49481b2c40b3b304","subject":"Merge \"Release oslo.limit for Xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/156b6df7b599297a5c8abc5c49481b2c40b3b304"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-15 12:22:49.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-17 11:41:12.000000000","tz":120},"subject":"Fix validation to ensure that first release of series isn\u0027t a bugfix","message":"Fix validation to ensure that first release of series isn\u0027t a bugfix\n\nWe recently observed that this check was skipped. Indeed it was\npossible to propose a bugfix version on the first release of a series\nwithout facing validation errors [1].\n\nThe described situation was allowed by checks that we introduced\nduring Wallaby [2]. Those was designed to ensure that we can\nbuild sdist from the proposed tag. Indeed, they temporarly created\nthe new git tag to build sdist from it. However we never delete this\ntag after our tests so this temporary tag is see as an existing tag\nand so that led us to the observed bug.\n\nOur validation check that the first version number of series isn\u0027t a\nbugfix after it tried to build the sdist from the new tag, so, without\nthis patch the tag newly created by the sdist check remains available\nin the repo, and so a couple of next checks are skipped, including the\nvalidation of the version number, because these checks are decorated\nto ignore existing tags.\n\nIndeed the function decorated with `skip_existing_tags` detect this\ntemporary tag as an existing tag, so, the decorated functions are\nskipped.\n\nWe should notice that it surely impacted other checks in the same manner.\nIndeed, all validation functions decorated with this function\n`skip_existing_tags` are impacted by this bug.\nHopefully it didn\u0027t have too much side effects.\n\nThese changes delete the temporary tag after usage to ensure to restore\nthe initial environment and to stop skipping next checks.\n\n[1] https://review.opendev.org/c/openstack/releases/+/795836\n[2] https://opendev.org/openstack/releases/commit/80652b523214ab7215928368246fbf726b78a645\n\nChange-Id: I5d9024528d08487a7582e0a0e73576b22708a6cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9504e96b5df4bd26d87c36681fecca5b27872a9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9504e96b5df4bd26d87c36681fecca5b27872a9d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
