)]}'
{"id":"openstack%2Freleases~796683","triplet_id":"openstack%2Freleases~master~I6a144d78a3dd275b11d7dfbe2859e992794ffd74","project":"openstack/releases","branch":"master","topic":"fix-deliverable-type","hashtags":[],"change_id":"I6a144d78a3dd275b11d7dfbe2859e992794ffd74","subject":"Adding more way to determine the deliverable type","status":"ABANDONED","created":"2021-06-16 14:23:27.000000000","updated":"2021-06-16 14:52:39.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0661bd9d1c4c93cc45d8cfb2c750ef666f8c24f3","_number":796683,"virtual_id_number":796683,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-06-16 14:35:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":-1,"date":"2021-06-16 14:26:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-16 14:26:47.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2021-06-16 14:35:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b850457a352e54986ba52318632d7e5c5f34da51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-16 14:23:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e4dc5e2ca6ca1acbc1a1137d1dd064a7d168b47","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-06-16 14:26:47.000000000","message":"Patch Set 1: Code-Review-1\n\nI don\u0027t think that will be necessary, there is a specific flag for retired repos so that they skip the release job test, we should probably just leverage that feature.","accounts_in_message":[],"_revision_number":1},{"id":"8eac37a06ae5b2a2e0202f3a9887533557dd0144","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 14:35:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74117975071d412787cdf6c2620f0aab : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91f2807967ce46b780da1f9d67886b04 : FAILURE in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d57341654ba0447da2dace95f1d4ebc8 : FAILURE in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b21502494d14ea4a1a0abec828ead3a : SUCCESS in 7m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e73b06f421694d3dae6d55c48c03174a : SUCCESS in 6m 24s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/cdd11ee9f2754a9db079050f4b547368 : SUCCESS in 5m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c8e04abf5c72442dac88e58acc7c613a : FAILURE in 6m 08s","accounts_in_message":[],"_revision_number":1},{"id":"0661bd9d1c4c93cc45d8cfb2c750ef666f8c24f3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-16 14:52:39.000000000","message":"Abandoned\n\nother mechanism allow us to handle that case https://review.opendev.org/c/openstack/project-team-guide/+/796687 . Abandoning these changes","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e08e22644d0568f7aa51c4e81e383425997db2ed","revisions":{"e08e22644d0568f7aa51c4e81e383425997db2ed":{"kind":"REWORK","_number":1,"created":"2021-06-16 14:23:27.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/83/796683/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/83/796683/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/796683/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/796683/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/796683/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/83/796683/1"}}},"commit":{"parents":[{"commit":"8c5f37aabe149cb2f246ddbbc718de0ef82f2977","subject":"Merge \"[searchlight] Transition Ocata to End of Life\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8c5f37aabe149cb2f246ddbbc718de0ef82f2977"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-16 14:21:04.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-16 14:21:04.000000000","tz":120},"subject":"Adding more way to determine the deliverable type","message":"Adding more way to determine the deliverable type\n\nWe faced issue with puppet-panko that has been recently retired [1] and our\nmachinery search for puppet module in the repo content to determine the\ndeliverable type.\n\nPuppet repositories names are standardized, use the name to allow us to\ndetermine the deliverable type when a repo is retired.\n\n[1] https://review.opendev.org/c/openstack/releases/+/792820\n\nChange-Id: I6a144d78a3dd275b11d7dfbe2859e992794ffd74\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e08e22644d0568f7aa51c4e81e383425997db2ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e08e22644d0568f7aa51c4e81e383425997db2ed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
