)]}'
{"id":"openstack%2Freleases~796692","triplet_id":"openstack%2Freleases~master~Ibffd79798c2ddbc61880bb4ebb60a40c97497244","project":"openstack/releases","branch":"master","topic":"release-yaql","hashtags":[],"change_id":"Ibffd79798c2ddbc61880bb4ebb60a40c97497244","subject":"Release yaql 1.1.4","status":"ABANDONED","created":"2021-06-16 15:06:50.000000000","updated":"2021-07-09 14:18:05.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4c66b26ee92dfa8a58be4d4229ff594b783217ea","_number":796692,"virtual_id_number":796692,"owner":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-06-17 12:03:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":-1,"date":"2021-07-07 14:33:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-16 15:13:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-16 15:20:21.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2021-06-16 15:23:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-17 11:54:29.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2021-07-07 14:33:04.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"51ad129de015fa38b2ace938d8d0acd5d0c7801c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2021-06-16 15:06:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0af02485168e570b300839d303a9c877f5e7a936","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 15:23:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dea98c99bd3742baaef332d31673b249 : SUCCESS in 7m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b9da61eb9d244e41afc2422943c38ba5 : SUCCESS in 5m 56s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/68584aa10d28417886aaaaacdf240bf8 : POST_FAILURE in 6m 19s","accounts_in_message":[],"_revision_number":1},{"id":"012f6594c73c81a7c235eb54f85d05afecaf4b96","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-17 11:54:29.000000000","message":"Patch Set 1:\n\nrecheck\n\nLooks like the recent ssh host key conflict failures have been in rackspace.\nLet\u0027s simply recheck.","accounts_in_message":[],"_revision_number":1},{"id":"efe63e69410d916851399170b7c49bb19892fc7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 12:03:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc9a67f940b94d1c9064d59bfd697ffb : SUCCESS in 7m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5574d2177d4647209a3b002286a99138 : SUCCESS in 5m 22s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/fa52e7bcda25485b9eb39930143152b6 : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":1},{"id":"d4076bb0b3ad970f5ddd844430363cd55a43afa4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-07-02 14:25:48.000000000","message":"Patch Set 1:\n\nHello PTL,\n\nPlease can you have a look to this patch?","accounts_in_message":[],"_revision_number":1},{"id":"87dd7603086679837aafa1368dc9b3e49c329f5a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-07-05 14:55:29.000000000","message":"Patch Set 1:\n\nDuplicated with https://review.opendev.org/c/openstack/releases/+/797996\nThe other patch propose a version 1.2.0","accounts_in_message":[],"_revision_number":1},{"id":"d69f4cf9034f918bc1f2cc8fe06896563d1f0fc4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-07-05 14:58:04.000000000","message":"Patch Set 1:\n\nI prefer to continue with the other version (1.2.0) as requirements have been updated, it will better reflect this kind of change.","accounts_in_message":[],"_revision_number":1},{"id":"e6d6431eaa376ce6e8571827d8c5238db85ade40","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-07-07 14:33:04.000000000","message":"Patch Set 1: Code-Review-1\n\n-1 to highlight my previous comments","accounts_in_message":[],"_revision_number":1},{"id":"4c66b26ee92dfa8a58be4d4229ff594b783217ea","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-07-09 14:18:05.000000000","message":"Abandoned\n\nWe decided to continue with https://review.opendev.org/c/openstack/releases/+/797996","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"db384ae148ed1097853091287b7d4bffc8823aa2","revisions":{"db384ae148ed1097853091287b7d4bffc8823aa2":{"kind":"REWORK","_number":1,"created":"2021-06-16 15:06:50.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/92/796692/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/92/796692/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/796692/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/796692/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/796692/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/92/796692/1"}}},"commit":{"parents":[{"commit":"9830659e169da9fb622974327d3f9ce2d5b0a0b6","subject":"Merge \"release oslo\u0027s deliverable for victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9830659e169da9fb622974327d3f9ce2d5b0a0b6"}]}],"author":{"name":"ricolin","email":"ricolin@ricolky.com","date":"2021-06-16 15:03:21.000000000","tz":480},"committer":{"name":"ricolin","email":"ricolin@ricolky.com","date":"2021-06-16 15:03:21.000000000","tz":480},"subject":"Release yaql 1.1.4","message":"Release yaql 1.1.4\n\n1.1.3 contains some collection usage which will stop working after\n  python 3.9. So we need to release commit c3bda9ee.\n\nChange-Id: Ibffd79798c2ddbc61880bb4ebb60a40c97497244\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/db384ae148ed1097853091287b7d4bffc8823aa2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/db384ae148ed1097853091287b7d4bffc8823aa2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
