)]}'
{"id":"openstack%2Freleases~796841","triplet_id":"openstack%2Freleases~master~I6f81ab30fab147a85d759b4db45b6265720e6f27","project":"openstack/releases","branch":"master","topic":"ocata-eol","hashtags":[],"change_id":"I6f81ab30fab147a85d759b4db45b6265720e6f27","subject":"[dragonflow] Set eol to Dragonflow Ocata","status":"MERGED","created":"2021-06-17 13:36:39.000000000","updated":"2021-08-19 13:55:48.000000000","submitted":"2021-08-19 13:55:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"796841-ocata-eol","meta_rev_id":"4ec2076f65f59aab43aee9ef9ad2dfec5d94b88b","_number":796841,"virtual_id_number":796841,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-19 13:55:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2021-08-19 13:39:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2021-08-05 15:14:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2021-08-19 13:39:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-17 13:37:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-17 13:48:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-16 10:32:47.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-08-19 13:39:55.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"b6c806cce8a623f0a153da59c52ba39e67430280","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-06-17 13:36:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"80d9a54ed612fa8eba41dc1a856a0770a913cbce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 13:48:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd2e56a219e74334a30675a75b3b62b3 : FAILURE in 6m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/278d8899e90e4aec948cb967c8404020 : SUCCESS in 5m 20s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/846a1b05728c4612917c7aff432deee8 : SUCCESS in 5m 23s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: file doc/source/newton/index.rst has only 24 lines\u0027\n  Comments left for invalid file doc/source/newton/index.rst","accounts_in_message":[],"_revision_number":1},{"id":"04e12a48726821e73a6d7dd85bdfc5c653c2382e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-07-05 17:14:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f96c975a6c95ae5eaf77d68fccad2c016775ab91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 17:32:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dfcdf5c59734d56af5b7a9ac6127e5b : SUCCESS in 9m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9713537f8b6c4c509f9171962fe2473c : SUCCESS in 5m 30s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/039712ed202b42f1b7baabfc2f711a60 : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":2},{"id":"74763078e3cd61f5ebc2170f1a3a3119159f40c1","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-07-16 10:32:47.000000000","message":"Patch Set 2: Code-Review+2\n\nI think that we can move further with this one","accounts_in_message":[],"_revision_number":2},{"id":"f9fa7be99b5b53ce149d4979208faa1aed37f1cf","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-07-16 10:57:10.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"1e7b1a7ed997e0ebc09b11076f6237e6af49a7c1","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-07-16 11:09:26.000000000","message":"Patch Set 2: Workflow-1\n\nlet me remove the newton-eol first as it seems there\u0027s already a tag like that [1]. That issue needs to be addressed in another patch.\n\n[1] https://opendev.org/openstack/dragonflow/src/tag/newton-eol","accounts_in_message":[],"_revision_number":2},{"id":"ffc4deb4b01ff0943550cfaf1aa43b8e4c89e123","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-07-16 11:12:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7d5024fdcd5baa2e77f6f7c924184b1571f79933","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-07-16 11:16:49.000000000","message":"Patch Set 3:\n\nack no problem","accounts_in_message":[],"_revision_number":3},{"id":"44bbcf2760a5430675a5c92d17743bfb54a88b0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 11:21:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/124f5bef29da4284893aadf81b21c861 : SUCCESS in 6m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/11d03a18861743b4a0b43a66e2f3309d : SUCCESS in 7m 04s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/5bb1932d41664f67b1b918e99a871f24 : SUCCESS in 7m 08s","accounts_in_message":[],"_revision_number":3},{"id":"b2027fce463cb54ae22bb73f2459c719253fd41e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-04 12:35:18.000000000","message":"Patch Set 3:\n\nI think that we can merge this patch.","accounts_in_message":[],"_revision_number":3},{"id":"98a479af622fa069a02c287d5f54d5b4d201cee0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-05 15:14:55.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"dd8e09909f5d91db83fc43f3f186f96567735ea9","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-08-19 13:39:55.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5a11aba99d27b6dae9c50597e2f6445822b34e96","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 13:40:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"67a645be979f6eacf22b293b3a0dcc34fb3c430e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 13:55:47.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb3f01e3edb8492990cb2abf9a1ae69d : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e31b1a4d14a4bcb85be7f998395bbf0 : SUCCESS in 7m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2422b76e0ba64e5293c40c5e2d981a19 : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":3},{"id":"4ec2076f65f59aab43aee9ef9ad2dfec5d94b88b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 13:55:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0ba15565393e1378b9a397a740f66d6ab5e2d352","revisions":{"a81ec65623800e5cb5daf46b3a89d669ad6ec145":{"kind":"REWORK","_number":1,"created":"2021-06-17 13:36:39.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/41/796841/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/41/796841/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/796841/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/796841/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/796841/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/41/796841/1"}}},"commit":{"parents":[{"commit":"2c1077c51a52fb7a21fbce443496462f8d5e2448","subject":"Merge \"ovsdbapp 1.11.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2c1077c51a52fb7a21fbce443496462f8d5e2448"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-06-17 13:33:06.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-06-17 13:33:06.000000000","tz":120},"subject":"WIP: Set eol to Dragonflow","message":"WIP: Set eol to Dragonflow\n\nDragonflow is not listed under deliverables ocata, however it has\nstable/ocata branch open (and even stable/newton).\n\nThis patch adds dragonflow under ocata and newton deliverables just to\nmake it possible to tag them with *-eol and delete the branches.\n\nChange-Id: I6f81ab30fab147a85d759b4db45b6265720e6f27\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a81ec65623800e5cb5daf46b3a89d669ad6ec145"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a81ec65623800e5cb5daf46b3a89d669ad6ec145"}]},"branch":"refs/heads/master"},"0f3f7ff50b084f3886fc6df827e89f5da021ffa2":{"kind":"REWORK","_number":2,"created":"2021-07-05 17:14:34.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/41/796841/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/41/796841/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/796841/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/796841/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/796841/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/41/796841/2"}}},"commit":{"parents":[{"commit":"2c1077c51a52fb7a21fbce443496462f8d5e2448","subject":"Merge \"ovsdbapp 1.11.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2c1077c51a52fb7a21fbce443496462f8d5e2448"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-06-17 13:33:06.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-07-05 17:13:03.000000000","tz":120},"subject":"[dragonflow] Set eol to Dragonflow Ocata and Newton","message":"[dragonflow] Set eol to Dragonflow Ocata and Newton\n\nDragonflow is not listed under deliverables ocata, however it has\nstable/ocata branch open (and even stable/newton).\n\nThis patch adds dragonflow under ocata and newton deliverables just to\nmake it possible to tag them with *-eol and delete the branches.\n\nChange-Id: I6f81ab30fab147a85d759b4db45b6265720e6f27\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0f3f7ff50b084f3886fc6df827e89f5da021ffa2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0f3f7ff50b084f3886fc6df827e89f5da021ffa2"}]},"branch":"refs/heads/master"},"0ba15565393e1378b9a397a740f66d6ab5e2d352":{"kind":"REWORK","_number":3,"created":"2021-07-16 11:12:49.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/41/796841/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/41/796841/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/796841/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/796841/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/796841/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/41/796841/3"}}},"commit":{"parents":[{"commit":"2c1077c51a52fb7a21fbce443496462f8d5e2448","subject":"Merge \"ovsdbapp 1.11.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2c1077c51a52fb7a21fbce443496462f8d5e2448"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-06-17 13:33:06.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-07-16 11:10:52.000000000","tz":120},"subject":"[dragonflow] Set eol to Dragonflow Ocata","message":"[dragonflow] Set eol to Dragonflow Ocata\n\nDragonflow is not listed under deliverables of ocata, however it has\nstable/ocata branch open.\n\nThis patch adds dragonflow under ocata deliverables just to make it\npossible to tag it with ocata-eol and delete the branch.\n\nChange-Id: I6f81ab30fab147a85d759b4db45b6265720e6f27\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0ba15565393e1378b9a397a740f66d6ab5e2d352"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0ba15565393e1378b9a397a740f66d6ab5e2d352"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
