)]}'
{"id":"openstack%2Freleases~797670","triplet_id":"openstack%2Freleases~master~I6b5179c15c961f968670529c0a41038bcbeb40c8","project":"openstack/releases","branch":"master","topic":"force-tags","hashtags":[],"change_id":"I6b5179c15c961f968670529c0a41038bcbeb40c8","subject":"Allow us to disable our validation","status":"ABANDONED","created":"2021-06-23 15:05:44.000000000","updated":"2021-06-24 09:40:19.000000000","total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"08e0baac4b2188967c907a24a5f94b0b66685231","_number":797670,"virtual_id_number":797670,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-06-24 06:35:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-23 15:31:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-24 08:55:20.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"CC"}],"messages":[{"id":"d41653e502192a57aba50ccd98a90a62484a9cb4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-23 15:05:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f0ba80afacb3e6de5e86aed3c3855f2a814faf8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-23 15:09:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"714011c1e234e02742511314c08e9f5cf5376374","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-23 15:10:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d3ddd86aed08ddfabd9982802f01b7a8c7195203","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 15:31:04.000000000","message":"Patch Set 2: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d664b3905b64fc9af46f1ef2d105327 : FAILURE in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11e731605af940b8bb3fef6d1bfdafc0 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6e46e9154bc480ebbe5d118a8392900 : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36b9ed2e888c49f894689e9b10db0371 : SUCCESS in 7m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/50dd4884852f4a63b46060f2bc18454f : SUCCESS in 5m 48s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d7534578f3604f7d96bfbb566f67fa3d : SUCCESS in 5m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1782861e761b487084b78b521ba2a977 : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":2},{"id":"4a50fa8f1f076280bcc9d262f44b91d6da748c0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-23 16:28:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ed23de8c3ff3191b030752ba12d153fff7cb3f1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-23 16:39:08.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"bc9b2960ba39b47b81edb186afddcc3750f81d32","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-23 16:40:48.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8fb13e3c9ad338f317da30e5f0c4a93d513d95d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 16:50:31.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94940c8afe1a43398f995c4a3447cc1b : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/287d254871014b7fa3a27f928a3d0e42 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96dc9b17321a46b38484244742e61d7f : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a24c1d864e34294a25f6129c5590818 : SUCCESS in 9m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dbb97d550ee8490f93d52dc9984177af : SUCCESS in 6m 40s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9cd1702045ad424fbc120cd2d0866c9b : SUCCESS in 5m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/480d0ed370444d9d85ebbc3d3be21665 : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":4},{"id":"88b9c05062480ea33f7d110d9cf16c6b89fa1785","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-24 06:25:53.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c5cd01ed8050d8fe0ffffe76d4d23981512dd2bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 06:35:59.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fac0d48a1b80498f891d834d023d656a : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5405e0e14dc4310a5020638dba9e77b : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1977d79d9aa34b3698f50ab5725c07fd : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/794cf00b196e4028b99392e9700a4037 : SUCCESS in 8m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c391ad0f904646b88f55adb75ff6fb10 : SUCCESS in 6m 13s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/1dec52b1a02142e98d3be95a890b1d3e : SUCCESS in 5m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5e6d25bf0859422381fb8fece5f49fb2 : SUCCESS in 6m 25s","accounts_in_message":[],"_revision_number":5},{"id":"de546c8b6a0a1b14ca081dce1728d257a38974b0","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-06-24 08:55:20.000000000","message":"Patch Set 5:\n\nI fear this is a recipe for future problems... For example, a given change bypasses validation, but once merged, creates a situation that will result in all future validation tests to fail. Or... the change bypasses validation but creates post-job issues. Like.... we don\u0027t even know how to represent those changes on releases.openstack.org.\n\nThe main drawback of the current approach (push some tags manually) is that we end up with tags in the repository that are not documented in releases.openstack.org. There has always been a number of extra tags in the repositories that are not resulting in true \"releases\", so I don\u0027t think it is that much of a problem.\n\nIt definitely creates two classes of \"releases\": the official OpenStack ones that follow the openstack model of a clear number channels named after series. And extra ones like \"bugfix releases\" that some projects want to do in addition. The fact that only the official OpenStack releases are documented in the openstack/releases repository and the OpenStack releases website (releases.openstack.org) is a feature for me, not a bug.","accounts_in_message":[],"_revision_number":5},{"id":"6de4212553d91a6c9012f4667cf2bf76d6b72b41","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-24 09:40:12.000000000","message":"Patch Set 5:\n\n\u003e Patch Set 5:\n\u003e \n\u003e I fear this is a recipe for future problems... For example, a given change bypasses validation, but once merged, creates a situation that will result in all future validation tests to fail. Or... the change bypasses validation but creates post-job issues. Like.... we don\u0027t even know how to represent those changes on releases.openstack.org.\n\u003e \n\u003e The main drawback of the current approach (push some tags manually) is that we end up with tags in the repository that are not documented in releases.openstack.org. There has always been a number of extra tags in the repositories that are not resulting in true \"releases\", so I don\u0027t think it is that much of a problem.\n\u003e \n\u003e It definitely creates two classes of \"releases\": the official OpenStack ones that follow the openstack model of a clear number channels named after series. And extra ones like \"bugfix releases\" that some projects want to do in addition. The fact that only the official OpenStack releases are documented in the openstack/releases repository and the OpenStack releases website (releases.openstack.org) is a feature for me, not a bug.\n\nI agree with you, let\u0027s abandon this idea.","accounts_in_message":[],"_revision_number":5},{"id":"08e0baac4b2188967c907a24a5f94b0b66685231","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-24 09:40:19.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"bfb41a9bf15ec76c9ac7783842b91fe21281465e","revisions":{"873f5b30d3c0de84467e7d4c075f21679743312b":{"kind":"REWORK","_number":1,"created":"2021-06-23 15:05:44.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/70/797670/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/70/797670/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/797670/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/797670/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/797670/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/70/797670/1"}}},"commit":{"parents":[{"commit":"1e50b850dfc1daad08ea8be04bec25465df6baa3","subject":"Merge \"[cloudkitty] Transition Ocata to End of Life\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1e50b850dfc1daad08ea8be04bec25465df6baa3"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-23 12:59:25.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-23 15:02:37.000000000","tz":120},"subject":"Allow us to disable our validation","message":"Allow us to disable our validation\n\nOnly the release team PTL is allowed to disable the validation.\n\nThis is related to our previous discussion with the ironic team [1] and\nrelated to the need of the non regular tags.\n\n[1] https://meetings.opendev.org/irclogs/%23openstack-release/%23openstack-release.2021-06-22.log.html#t2021-06-22T15:37:05\n\nChange-Id: I6b5179c15c961f968670529c0a41038bcbeb40c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/873f5b30d3c0de84467e7d4c075f21679743312b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/873f5b30d3c0de84467e7d4c075f21679743312b"}]},"branch":"refs/heads/master"},"30d4667ac278980a60f417642905340b3d019539":{"kind":"REWORK","_number":2,"created":"2021-06-23 15:10:58.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/70/797670/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/70/797670/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/797670/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/797670/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/797670/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/70/797670/2"}}},"commit":{"parents":[{"commit":"1e50b850dfc1daad08ea8be04bec25465df6baa3","subject":"Merge \"[cloudkitty] Transition Ocata to End of Life\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1e50b850dfc1daad08ea8be04bec25465df6baa3"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-23 12:59:25.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-23 15:10:38.000000000","tz":120},"subject":"Allow us to disable our validation","message":"Allow us to disable our validation\n\nOnly the release team PTL is allowed to disable the validation.\n\nThis is related to our previous discussion with the ironic team [1] and\nrelated to the need of the non regular tags.\n\n[1] https://meetings.opendev.org/irclogs/%23openstack-release/%23openstack-release.2021-06-22.log.html#t2021-06-22T15:37:05\n\nChange-Id: I6b5179c15c961f968670529c0a41038bcbeb40c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/30d4667ac278980a60f417642905340b3d019539"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/30d4667ac278980a60f417642905340b3d019539"}]},"branch":"refs/heads/master"},"2f1bd5508b188865b121b14ee6f60f33a310d731":{"kind":"REWORK","_number":3,"created":"2021-06-23 16:28:53.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/70/797670/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/70/797670/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/797670/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/797670/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/797670/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/70/797670/3"}}},"commit":{"parents":[{"commit":"1e50b850dfc1daad08ea8be04bec25465df6baa3","subject":"Merge \"[cloudkitty] Transition Ocata to End of Life\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1e50b850dfc1daad08ea8be04bec25465df6baa3"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-23 12:59:25.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-23 16:28:37.000000000","tz":120},"subject":"Allow us to disable our validation","message":"Allow us to disable our validation\n\nOnly the release team PTL is allowed to disable the validation.\n\nThis is related to our previous discussion with the ironic team [1] and\nrelated to the need of the non regular tags.\n\n[1] https://meetings.opendev.org/irclogs/%23openstack-release/%23openstack-release.2021-06-22.log.html#t2021-06-22T15:37:05\n\nChange-Id: I6b5179c15c961f968670529c0a41038bcbeb40c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2f1bd5508b188865b121b14ee6f60f33a310d731"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2f1bd5508b188865b121b14ee6f60f33a310d731"}]},"branch":"refs/heads/master"},"ae6553fa2b47f6cc8d968603d846a716bcef2d27":{"kind":"REWORK","_number":4,"created":"2021-06-23 16:39:08.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/70/797670/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/70/797670/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/797670/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/797670/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/797670/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/70/797670/4"}}},"commit":{"parents":[{"commit":"1e50b850dfc1daad08ea8be04bec25465df6baa3","subject":"Merge \"[cloudkitty] Transition Ocata to End of Life\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1e50b850dfc1daad08ea8be04bec25465df6baa3"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-23 12:59:25.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-23 16:38:54.000000000","tz":120},"subject":"Allow us to disable our validation","message":"Allow us to disable our validation\n\nOnly the release team PTL is allowed to disable the validation.\n\nThis is related to our previous discussion with the ironic team [1] and\nrelated to the need of the non regular tags.\n\n[1] https://meetings.opendev.org/irclogs/%23openstack-release/%23openstack-release.2021-06-22.log.html#t2021-06-22T15:37:05\n\nChange-Id: I6b5179c15c961f968670529c0a41038bcbeb40c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ae6553fa2b47f6cc8d968603d846a716bcef2d27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ae6553fa2b47f6cc8d968603d846a716bcef2d27"}]},"branch":"refs/heads/master"},"bfb41a9bf15ec76c9ac7783842b91fe21281465e":{"kind":"REWORK","_number":5,"created":"2021-06-24 06:25:53.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/70/797670/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/70/797670/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/797670/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/797670/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/797670/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/70/797670/5"}}},"commit":{"parents":[{"commit":"1e50b850dfc1daad08ea8be04bec25465df6baa3","subject":"Merge \"[cloudkitty] Transition Ocata to End of Life\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1e50b850dfc1daad08ea8be04bec25465df6baa3"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-23 12:59:25.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-24 06:25:34.000000000","tz":120},"subject":"Allow us to disable our validation","message":"Allow us to disable our validation\n\nOnly the release team PTL is allowed to disable the validation.\n\nThis is related to our previous discussion with the ironic team [1] and\nrelated to the need of the non regular tags.\n\n[1] https://meetings.opendev.org/irclogs/%23openstack-release/%23openstack-release.2021-06-22.log.html#t2021-06-22T15:37:05\n\nChange-Id: I6b5179c15c961f968670529c0a41038bcbeb40c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bfb41a9bf15ec76c9ac7783842b91fe21281465e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bfb41a9bf15ec76c9ac7783842b91fe21281465e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
