)]}'
{"id":"openstack%2Freleases~797672","triplet_id":"openstack%2Freleases~master~I96bd4e65fa2aaf6192f5b69440581751e722b205","project":"openstack/releases","branch":"master","topic":"force-tags","hashtags":[],"change_id":"I96bd4e65fa2aaf6192f5b69440581751e722b205","subject":"DNM [force_tagging] PoC 2 - force won\u0027t be allowed","status":"ABANDONED","created":"2021-06-23 15:05:44.000000000","updated":"2021-06-24 09:40:32.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"987c07df2d6119e959ca16dc9daf5b0eaa1ee19c","_number":797672,"virtual_id_number":797672,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-06-24 06:36:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-06-24 09:40:32.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-23 15:07:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"80db855412f905d6ceea049f906439b9b6c7c380","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-23 15:05:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d8ca929f10f6b896b603048faae6cffbfe982807","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 15:07:02.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"fc7b88e1ddf1de959da67076144914765f9dbee3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-23 15:10:58.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"f32ccca58a5ecc0e2a4bb7eb28a246a820e79e3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 15:20:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e860bb5fcdd4172a6a17ddde010e9af : SUCCESS in 8m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/097d3b673e244180b6a9eccaec0b2afb : FAILURE in 6m 26s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/636192e62bac4a0ba4e21aa8f74b3770 : FAILURE in 5m 40s","accounts_in_message":[],"_revision_number":2},{"id":"5a9b046c2d675bed24db899dd482f125a2344845","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-23 16:28:53.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"14ca6e3d2ceab12b17943fa4dc1e9901cda668de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-23 16:33:39.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5ac5845beaeaaaa067a608af8339e8a49c9046fd","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 16:35:03.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"ac76963c6a32e52dede34598051ae35451443c81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-23 16:39:08.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"05f1c629f8b7d57a062c3e842696487ef2278eef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 16:53:00.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6536ce1dabc415ca853611ecf07162b : FAILURE in 6m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7ca12ce3d41345cc835e2a1278200829 : FAILURE in 6m 14s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ae3a640782a84334b750ca27186c76e9 : SUCCESS in 6m 56s","accounts_in_message":[],"_revision_number":5},{"id":"3ed76450d4d08c551b8fe8278a1e5846063c0e4c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-23 17:08:53.000000000","message":"Patch Set 5:\n\n(1 comment)\n\nrecheck\n\nLogs are not available and tests pass locally","accounts_in_message":[],"_revision_number":5},{"id":"f407b4cb23f36060d49656364d67b2733bac0964","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-24 05:58:53.000000000","message":"Patch Set 5:\n\nrecheck\n\nJobs logs are still unavailable","accounts_in_message":[],"_revision_number":5},{"id":"dfa44f8935a35ceb79646141eb2f6cb106a455c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 06:05:32.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ca4a5ba376a4d98bf0076ed804353f2 : FAILURE in 5m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b959797230104cde9827df9960e6a7c1 : FAILURE in 5m 47s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/a58b268db39944d397761aeb8c2ad396 : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":5},{"id":"3658bb455c916b2d915ebc388e106b9f90383c4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-24 06:25:53.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"7411744627d318e1a13691bd0d1cfd9aa23c4db9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 06:36:08.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3600e27f96a946de89a620e904c3e636 : FAILURE in 6m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/888f3aa1868b4543bbf5f45cd3e662dd : FAILURE in 7m 26s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/306a15da204a410582f6d0ab195e9522 : SUCCESS in 6m 16s","accounts_in_message":[],"_revision_number":6},{"id":"0ee45f7d3edf01552f81a29656fdc8e6c67b6877","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-24 09:40:32.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"19d44f0c602144aa22d4e173ab1a189f109ec1da","revisions":{"60dd16b994116ba2176c21d22de0be098588bf86":{"kind":"REWORK","_number":1,"created":"2021-06-23 15:05:44.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/72/797672/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/72/797672/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/797672/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/797672/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/797672/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/72/797672/1"}}},"commit":{"parents":[{"commit":"a5fd6d5844ba3c09875c4f2dad34e9f9324acc7b","subject":"DNM [force_tagging] PoC 1 - force will be allowed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a5fd6d5844ba3c09875c4f2dad34e9f9324acc7b"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-23 15:04:57.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-23 15:05:24.000000000","tz":120},"subject":"DNM [force_tagging] PoC 2 - force won\u0027t be allowed","message":"DNM [force_tagging] PoC 2 - force won\u0027t be allowed\n\nChange-Id: I96bd4e65fa2aaf6192f5b69440581751e722b205\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/60dd16b994116ba2176c21d22de0be098588bf86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/60dd16b994116ba2176c21d22de0be098588bf86"}]},"branch":"refs/heads/master"},"c067419aa031d1d600a1f0aecb13fd85892b4984":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-06-23 15:10:58.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/72/797672/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/72/797672/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/797672/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/797672/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/797672/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/72/797672/2"}}},"commit":{"parents":[{"commit":"cbf7c3b2072dcefe53087ec3b7fe19e6e9927b7d","subject":"DNM [force_tagging] PoC 1 - force will be allowed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cbf7c3b2072dcefe53087ec3b7fe19e6e9927b7d"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-23 15:04:57.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-23 15:10:38.000000000","tz":120},"subject":"DNM [force_tagging] PoC 2 - force won\u0027t be allowed","message":"DNM [force_tagging] PoC 2 - force won\u0027t be allowed\n\nChange-Id: I96bd4e65fa2aaf6192f5b69440581751e722b205\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c067419aa031d1d600a1f0aecb13fd85892b4984"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c067419aa031d1d600a1f0aecb13fd85892b4984"}]},"branch":"refs/heads/master"},"da95407a12cd28950b8d17a3e816ff8e14701914":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-06-23 16:28:53.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/72/797672/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/72/797672/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/797672/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/797672/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/797672/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/72/797672/3"}}},"commit":{"parents":[{"commit":"b659a929b2ca4f3a52ad9cd178c3388a10e717dd","subject":"DNM [force_tagging] PoC 1 - force will be allowed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b659a929b2ca4f3a52ad9cd178c3388a10e717dd"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-23 15:04:57.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-23 16:28:37.000000000","tz":120},"subject":"DNM [force_tagging] PoC 2 - force won\u0027t be allowed","message":"DNM [force_tagging] PoC 2 - force won\u0027t be allowed\n\nChange-Id: I96bd4e65fa2aaf6192f5b69440581751e722b205\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/da95407a12cd28950b8d17a3e816ff8e14701914"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/da95407a12cd28950b8d17a3e816ff8e14701914"}]},"branch":"refs/heads/master"},"e39707f0fd4d6cfe923b400efc79ae052b3474e7":{"kind":"REWORK","_number":4,"created":"2021-06-23 16:33:39.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/72/797672/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/72/797672/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/797672/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/797672/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/797672/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/72/797672/4"}}},"commit":{"parents":[{"commit":"9769cc246ec60f919e063695fad32c7e8a03f914","subject":"DNM [force_tagging] PoC 1 - force will be allowed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9769cc246ec60f919e063695fad32c7e8a03f914"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-23 15:04:57.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-23 16:33:23.000000000","tz":120},"subject":"DNM [force_tagging] PoC 2 - force won\u0027t be allowed","message":"DNM [force_tagging] PoC 2 - force won\u0027t be allowed\n\nChange-Id: I96bd4e65fa2aaf6192f5b69440581751e722b205\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e39707f0fd4d6cfe923b400efc79ae052b3474e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e39707f0fd4d6cfe923b400efc79ae052b3474e7"}]},"branch":"refs/heads/master"},"1c072af2f3ade1a367168fe61936bf31f5a6932b":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-06-23 16:39:08.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/72/797672/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/72/797672/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/797672/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/797672/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/797672/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/72/797672/5"}}},"commit":{"parents":[{"commit":"435cb9f6db28b504a7332eae1b2196ac941b7a1b","subject":"DNM [force_tagging] PoC 1 - force will be allowed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/435cb9f6db28b504a7332eae1b2196ac941b7a1b"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-23 15:04:57.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-23 16:38:54.000000000","tz":120},"subject":"DNM [force_tagging] PoC 2 - force won\u0027t be allowed","message":"DNM [force_tagging] PoC 2 - force won\u0027t be allowed\n\nChange-Id: I96bd4e65fa2aaf6192f5b69440581751e722b205\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1c072af2f3ade1a367168fe61936bf31f5a6932b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1c072af2f3ade1a367168fe61936bf31f5a6932b"}]},"branch":"refs/heads/master"},"19d44f0c602144aa22d4e173ab1a189f109ec1da":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-06-24 06:25:53.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/72/797672/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/72/797672/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/797672/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/797672/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/797672/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/72/797672/6"}}},"commit":{"parents":[{"commit":"7a1357cf0af568983c176b2422c31c4d09fc2ba6","subject":"DNM [force_tagging] PoC 1 - force will be allowed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7a1357cf0af568983c176b2422c31c4d09fc2ba6"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-23 15:04:57.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-24 06:25:34.000000000","tz":120},"subject":"DNM [force_tagging] PoC 2 - force won\u0027t be allowed","message":"DNM [force_tagging] PoC 2 - force won\u0027t be allowed\n\nChange-Id: I96bd4e65fa2aaf6192f5b69440581751e722b205\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/19d44f0c602144aa22d4e173ab1a189f109ec1da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/19d44f0c602144aa22d4e173ab1a189f109ec1da"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
