)]}'
{"id":"openstack%2Freleases~798084","triplet_id":"openstack%2Freleases~master~I41f2387dc674a64247e0f275c08a17aae4c05e2f","project":"openstack/releases","branch":"master","topic":"oslo-db-june","hashtags":[],"change_id":"I41f2387dc674a64247e0f275c08a17aae4c05e2f","subject":"Release oslo.db 10.0.0 for Xena","status":"MERGED","created":"2021-06-25 13:13:10.000000000","updated":"2021-07-01 12:30:24.000000000","submitted":"2021-07-01 12:30:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"798084-oslo-db-june","meta_rev_id":"bc51dfdf12cfd5aa5f3ca554bfe54f6419ab7a6f","_number":798084,"virtual_id_number":798084,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-01 12:30:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2021-07-01 12:21:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2021-06-30 14:16:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2021-06-30 09:22:18.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2021-07-01 12:21:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"date":"2021-06-30 09:24:02.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-06-25 13:14:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}],"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-25 13:14:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-25 17:08:49.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"},{"updated":"2021-06-30 14:16:56.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-07-01 12:21:43.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"75da8a0794a13c7e6496ccff7bf8ac57fa5d0f6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-25 13:13:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"320597567b1e91e65cae66ad8135727b4374fd9e","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-25 13:14:13.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"326b7cc067e9a1d349a739276ecffce3670bbd3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-25 13:24:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57c0aadfd2b04abb98f96a7f348033f9 : SUCCESS in 8m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bd2128d538424c4e8609acb054aed09b : SUCCESS in 7m 03s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f64d31ef98c84d329ca5f69d6a18c0bb : SUCCESS in 7m 27s","accounts_in_message":[],"_revision_number":1},{"id":"31d547a7be51f889284ae740802f03184c0b9e80","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-06-25 17:08:49.000000000","message":"Patch Set 1:\n\nI have a few more deprecated things that I\u0027d like to remove in oslo.db. I wonder if it would make sense to revert the deprecated option bindings patch, release a 9.x patch, and then reapply it with the additional changes and cut a 10.0.0 release? If so, I can propose the revert and reapply. If not, consider me +1 on this.","accounts_in_message":[],"_revision_number":1},{"id":"cdfccd17c2e6865fff21965af27d3b966d6f35be","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-28 09:49:38.000000000","message":"Patch Set 1: Workflow-1\n\n\u003e Patch Set 1:\n\u003e \n\u003e I have a few more deprecated things that I\u0027d like to remove in oslo.db. I wonder if it would make sense to revert the deprecated option bindings patch, release a 9.x patch, and then reapply it with the additional changes and cut a 10.0.0 release? If so, I can propose the revert and reapply. If not, consider me +1 on this.\n\nThe question is more are you in rush?\n\nConcerning myself I do not particularly need a new version quickly. You asked me if we could  release oslo.db ASAP after the merge of [1], so this is the reason of this patch. If you are not too much in rush to allow us to revert things as you suggested, then I agree that it would be worth to bring all the non backward compat changes together through a single major version. \n\nI -W this patch for now to allow you to reply. I let you handle the needed reverts and reapply them. Let me know if you need reviews and then I\u0027ll process them with an high priority.\n\nAnyway, tags are cheaps so we could release a new major version again.\n\n[1] https://opendev.org/openstack/oslo.db/commit/f3bdd4d4220a884e4752ce6b61c2605bfd72f2fb","accounts_in_message":[],"_revision_number":1},{"id":"2f1f65d6d190a5eb6ad1c44f9616bc41ff2354f4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-30 09:22:18.000000000","message":"Patch Set 1: Code-Review+1 -Workflow\n\nSo, Stephen and I decided to continue with this current version.\n\nWe will release another major version later to land the other deprecated/removed stuffs","accounts_in_message":[],"_revision_number":1},{"id":"db372a0021b3411584855893330fc69cec0bf67d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-30 09:24:02.000000000","message":"Patch Set 1:\n\nOnly +1 to reflect my oslo caps and let the other release core members validate these changes","accounts_in_message":[],"_revision_number":1},{"id":"49dd7b62bc85918d4a0cccf29fa7058e99279f67","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-06-30 14:16:56.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM:\n* major version bump is explained and looks OK (shouldn\u0027t reno show the same major version bump?)\n* uploaded/approved by release liaison\n* no issues found in test job logs","accounts_in_message":[],"_revision_number":1},{"id":"369aaa8b82844fb352c9b5158af5e3f9e5c00ac9","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-07-01 12:21:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"90e38c08c35d5ad9b13c3a49bd373bf1e4e9bd49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 12:21:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3684b830fa3952b55d9b6725095de5433ba00793","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 12:30:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c41906a13d1418ba929d957542049b7 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b5b686c870f45ff86cad5a8c5f61d17 : SUCCESS in 7m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c32b0929a80543e8ab5ad766e24b86a8 : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":1},{"id":"bc51dfdf12cfd5aa5f3ca554bfe54f6419ab7a6f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 12:30:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b737a323b47e0f0d3da3a3a38ac19bfc5739020c","revisions":{"b737a323b47e0f0d3da3a3a38ac19bfc5739020c":{"kind":"REWORK","_number":1,"created":"2021-06-25 13:13:10.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/84/798084/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/84/798084/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/84/798084/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/84/798084/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/84/798084/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/84/798084/1"}}},"commit":{"parents":[{"commit":"bf465181cbc37f3be43a22f605bb729c73f90763","subject":"Merge \"Create pike-eol tag for horizon project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bf465181cbc37f3be43a22f605bb729c73f90763"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-25 13:10:11.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-25 13:11:47.000000000","tz":120},"subject":"Release oslo.db 10.0.0 for Xena","message":"Release oslo.db 10.0.0 for Xena\n\nDeprecated option bindings have been removed [1] so this is not a\nbackward compatible version for this reason we need to release a\nmajor version.\n\n[1] https://opendev.org/openstack/oslo.db/commit/0e76c0216069a0d3f7dea1fe01d50330f03fd344\n\nChange-Id: I41f2387dc674a64247e0f275c08a17aae4c05e2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b737a323b47e0f0d3da3a3a38ac19bfc5739020c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b737a323b47e0f0d3da3a3a38ac19bfc5739020c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
