)]}'
{"id":"openstack%2Freleases~798520","triplet_id":"openstack%2Freleases~master~I97f38b96bce39d66072128141470679d2c03cc1e","project":"openstack/releases","branch":"master","topic":"osc-lib-2.4.1","hashtags":[],"change_id":"I97f38b96bce39d66072128141470679d2c03cc1e","subject":"osc-lib release 2.4.1","status":"MERGED","created":"2021-06-29 08:34:17.000000000","updated":"2021-07-05 09:18:02.000000000","submitted":"2021-07-05 09:18:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"798520-osc-lib-2.4.1","meta_rev_id":"c4ab7a33a2e08ca32c8f89f55f8cd1620dfe37c9","_number":798520,"virtual_id_number":798520,"owner":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-05 09:18:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2021-07-05 09:07:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2021-07-02 14:50:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2021-07-02 14:57:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2021-07-02 08:20:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":1,"date":"2021-06-30 14:08:15.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-02 14:22:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2021-07-05 09:07:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-07-02 14:51:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-29 08:49:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-29 09:05:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-29 11:20:20.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2021-06-30 10:25:24.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"},{"updated":"2021-06-30 14:08:15.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-06-30 14:16:33.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2021-06-30 15:10:50.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2021-07-01 10:33:25.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2021-07-02 08:20:46.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2021-07-02 14:57:11.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-07-02 15:25:33.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2021-07-02 15:25:33.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"5bc4cd32f8eded48912f816a6fe20670ef56dbe8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2021-06-29 08:34:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b5d73c0fa88a0de605f6b50ee2d1ca9023371d7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 09:05:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/925d1202996946299f5a6d17e09e3e43 : SUCCESS in 8m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/73fd05fef71f49d7a7caa65be55e27c7 : SUCCESS in 8m 28s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/382b870300ca4cd29a94f1e1bbe470cd : SUCCESS in 7m 14s","accounts_in_message":[],"_revision_number":1},{"id":"69307aa32ec35ec304424dbab691210653150350","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-06-30 10:25:24.000000000","message":"Patch Set 1:\n\nMay be can also include https://review.opendev.org/c/openstack/osc-lib/+/798808 in this release if feels ok.","accounts_in_message":[],"_revision_number":1},{"id":"3c5a2ee1f7d09ad14d366e727bb6204939fdd60e","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-06-30 12:30:51.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e May be can also include https://review.opendev.org/c/openstack/osc-lib/+/798808 in this release if feels ok.\n\nMerged, so will include here.","accounts_in_message":[],"_revision_number":1},{"id":"cbb5a94268ec6aa2b44f16ea3db41b2051f21775","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-06-30 12:37:44.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"759f16a8fe08ec7061c2727435ab2c367ec707d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 12:49:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8210c6eb2b548f19dc30a7343b9de84 : SUCCESS in 8m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e7efec99ab044e5b910823c3fba9f285 : SUCCESS in 7m 46s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/551d05fce3da433b9ad762928531c7b0 : SUCCESS in 7m 35s","accounts_in_message":[],"_revision_number":2},{"id":"395a190b2d2006bf5daa7a29e1cd69d0d43e9a90","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-30 14:08:15.000000000","message":"Patch Set 2: Code-Review+1\n\nLGTM. Waiting for PTL/liaison their validations","accounts_in_message":[],"_revision_number":2},{"id":"16aaf7cb4f202eb8e61ba5376afc436496bfa065","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-06-30 15:10:50.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"7970cc65f7a007835bb7fa822b047dfcfc8f202f","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-07-01 10:33:25.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2: Code-Review+1\n\u003e \n\u003e LGTM. Waiting for PTL/liaison their validations\n\nbut as i see there is no PTL","accounts_in_message":[],"_revision_number":2},{"id":"aef5335b9f0007131bb440c0dcc4350503cf8df9","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-07-01 10:36:07.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e \u003e Patch Set 2: Code-Review+1\n\u003e \u003e \n\u003e \u003e LGTM. Waiting for PTL/liaison their validations\n\u003e \n\u003e but as i see there is no PTL\n\nsorry wrong commet, i misread earlier https://github.com/openstack/governance/blob/c056a73f663e93080f5370c3602d0bef52dba543/reference/projects.yaml#L2055-L2059","accounts_in_message":[],"_revision_number":2},{"id":"f5f0b029ea6244d81c93a49ec868766330c5c642","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-07-02 05:50:56.000000000","message":"Patch Set 2:\n\n@Artem, @dtantsur can u please have a look and vote, Thanks.","accounts_in_message":[],"_revision_number":2},{"id":"ce5cd4064186c04651f765590c9373dd1fb08cd9","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-07-02 08:20:46.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5d0c35d87488e43cd4df7ff3b6200812911d430b","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-07-02 14:22:41.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5a9d7dc4034c2b7becda81d6ea74e7e6b8ae00fe","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-07-02 14:50:41.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"6c9f3ad3fe36bc248b35577d89aec515aba1940c","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-07-02 14:50:55.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5c555e223740fa49d934f855ef5f50963c670ef8","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 14:51:51.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"b626102589bb89edbe1622b843500f3a98c338bd","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-07-02 14:57:11.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM:\n* version bump looks OK\n* PTL-Approved flag is there\n* no issues found in test job logs","accounts_in_message":[],"_revision_number":2},{"id":"4098c4028899a90f77688107ca0b6d8c0ce08d80","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-07-05 09:07:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c9a51435fe72095590e23b80b45ea65996810825","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 09:07:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c72c44248443089f97037a29b46333acafa2a1bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 09:18:01.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c237e9374936457db19109b7ac1d9041 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb56913f5e6d4fa48a43e56d9aadddb9 : SUCCESS in 7m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4eb60a8cf3ca4bedb6a4a83e191a5501 : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":2},{"id":"c4ab7a33a2e08ca32c8f89f55f8cd1620dfe37c9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 09:18:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6e2eaf3dcda7f6038f74e9ff1cdc0da6d8708349","revisions":{"87e45d012a7d83c6b0b310ffcf1d3c7d86cdc0ee":{"kind":"REWORK","_number":1,"created":"2021-06-29 08:34:17.000000000","uploader":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"ref":"refs/changes/20/798520/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/20/798520/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/20/798520/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/20/798520/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/20/798520/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/20/798520/1"}}},"commit":{"parents":[{"commit":"fff63ef24122ac7e5921179508bff9bce78d6f49","subject":"Merge \"Create Wallaby branch for OpenStack-Ansible Roles\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fff63ef24122ac7e5921179508bff9bce78d6f49"}]}],"author":{"name":"James Page","email":"james.page@ubuntu.com","date":"2021-06-29 08:32:39.000000000","tz":60},"committer":{"name":"James Page","email":"james.page@ubuntu.com","date":"2021-06-29 08:34:13.000000000","tz":60},"subject":"osc-lib release 2.4.1","message":"osc-lib release 2.4.1\n\nPatch release of osc-lib which resolves issues during shell\ncleanup, specifically when generating command completions for\nshell integration.\n\nChange-Id: I97f38b96bce39d66072128141470679d2c03cc1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/87e45d012a7d83c6b0b310ffcf1d3c7d86cdc0ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/87e45d012a7d83c6b0b310ffcf1d3c7d86cdc0ee"}]},"branch":"refs/heads/master"},"6e2eaf3dcda7f6038f74e9ff1cdc0da6d8708349":{"kind":"REWORK","_number":2,"created":"2021-06-30 12:37:44.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/20/798520/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/20/798520/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/20/798520/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/20/798520/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/20/798520/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/20/798520/2"}}},"commit":{"parents":[{"commit":"fff63ef24122ac7e5921179508bff9bce78d6f49","subject":"Merge \"Create Wallaby branch for OpenStack-Ansible Roles\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fff63ef24122ac7e5921179508bff9bce78d6f49"}]}],"author":{"name":"James Page","email":"james.page@ubuntu.com","date":"2021-06-29 08:32:39.000000000","tz":60},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2021-06-30 12:37:38.000000000","tz":0},"subject":"osc-lib release 2.4.1","message":"osc-lib release 2.4.1\n\nPatch release of osc-lib which resolves issues during shell\ncleanup, specifically when generating command completions for\nshell integration. Also includes commit for cleaning up of session\nobject if available.\n\nChange-Id: I97f38b96bce39d66072128141470679d2c03cc1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6e2eaf3dcda7f6038f74e9ff1cdc0da6d8708349"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6e2eaf3dcda7f6038f74e9ff1cdc0da6d8708349"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
