)]}'
{"id":"openstack%2Freleases~798924","triplet_id":"openstack%2Freleases~master~I5893202c31dfadbe5009a87f4e545787fd065368","project":"openstack/releases","branch":"master","topic":"tripleo-queens-eol","hashtags":[],"change_id":"I5893202c31dfadbe5009a87f4e545787fd065368","subject":"Tag stable/queens tripleo repos as EOL","status":"MERGED","created":"2021-06-30 15:57:46.000000000","updated":"2021-07-15 10:42:40.000000000","submitted":"2021-07-12 14:17:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"798924-tripleo-queens-eol","meta_rev_id":"0d07335b3dbf808c97892338c80682247b5d22a9","_number":798924,"virtual_id_number":798924,"owner":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":22722,"name":"Jason Joyce","email":"fuzzball81@gmail.com","username":"jjoyce"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-12 14:17:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":22722,"name":"Jason Joyce","email":"fuzzball81@gmail.com","username":"jjoyce"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2021-07-05 19:18:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2021-07-14 14:12:04.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":1,"date":"2021-07-06 13:24:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-07-12 14:09:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2021-07-15 10:42:40.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":22722,"name":"Jason Joyce","email":"fuzzball81@gmail.com","username":"jjoyce"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-07-12 14:09:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":22722,"name":"Jason Joyce","email":"fuzzball81@gmail.com","username":"jjoyce"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22722,"name":"Jason Joyce","email":"fuzzball81@gmail.com","username":"jjoyce"},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-30 15:58:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-30 16:02:44.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2021-06-30 16:07:19.000000000","updated_by":{"_account_id":22722,"name":"Jason Joyce","email":"fuzzball81@gmail.com","username":"jjoyce"},"reviewer":{"_account_id":22722,"name":"Jason Joyce","email":"fuzzball81@gmail.com","username":"jjoyce"},"state":"REVIEWER"},{"updated":"2021-06-30 16:10:09.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-07-01 10:24:44.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2021-07-01 13:22:02.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"state":"REVIEWER"},{"updated":"2021-07-01 13:48:14.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2021-07-02 09:37:33.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-07-12 14:09:48.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-07-15 10:42:40.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"REVIEWER"}],"messages":[{"id":"3f5e608d80bef59028cc3c453be6c171674eebd8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-06-30 15:57:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"da0cec8fd442c27d0eef31b9c285dd2c71ca22ff","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 15:58:54.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"40813387b1841e9a1af93a92a7c7f62c73258a6a","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-06-30 16:02:44.000000000","message":"Patch Set 1: Code-Review+1\n\nwoot","accounts_in_message":[],"_revision_number":1},{"id":"78fd14e4e1d875d2fcfa9bbdb4cfb2f0ff1d20b1","author":{"_account_id":22722,"name":"Jason Joyce","email":"fuzzball81@gmail.com","username":"jjoyce"},"date":"2021-06-30 16:07:19.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9a485e3459dd36768b467f888e8b9d316b473509","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 16:09:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fd4d49070154937abad74843ffcf8ad : SUCCESS in 10m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/97d133cef8c04b7eb455a1f6240066db : SUCCESS in 7m 03s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7a688fd2fe7d41ab86ad9026a5f83049 : SUCCESS in 9m 39s","accounts_in_message":[],"_revision_number":1},{"id":"06d98c2e100f8333c2076d1a4cf93a3c59dd7ffe","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-06-30 16:10:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"23bbfd618d5ce77cc2aa97c95e24fa8ce871e915","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-07-01 10:24:44.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5810240045489df0c417b24b835e5e99b1182a9a","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-07-01 13:22:02.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"eb852bcb60c16540d5dcb1a5025dacb7c4fcc296","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2021-07-01 13:48:14.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"44df7428b5a2d5e310686f0e0afd5d9b32c64286","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-07-02 09:37:33.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)\n\nIn general, the patch looks OK (hashes are OK, too), however I see some TripleO repositories that are missing from the list (see my comment inline).\n\n@Marios I would ask one more thing: could you please check whether there are job definitions (in openstack-zuul-jobs and other repositories) that was used only in TripleO\u0027s queens repos and we won\u0027t need them anymore? If you find such job definitions/templates, please propose a patch that removes those definitions. Thanks!\n\n(Giving -1 to get attention to my questions)","accounts_in_message":[],"_revision_number":1},{"id":"20bffeab829f9b67a47f35b11b33d2b36bd8b318","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-07-05 15:17:38.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e (3 comments)\n\u003e \n\u003e In general, the patch looks OK (hashes are OK, too), however I see some TripleO repositories that are missing from the list (see my comment inline).\n\nack thanks for the very careful review as always and apologies for delay in reply I was out at the end of last week.\n\n\u003e \n\u003e @Marios I would ask one more thing: could you please check whether there are job definitions (in openstack-zuul-jobs and other repositories) that was used only in TripleO\u0027s queens repos and we won\u0027t need them anymore? If you find such job definitions/templates, please propose a patch that removes those definitions. Thanks!\n\u003e \n\nyes of course we are/will do this... in general we don\u0027t have a large number of jobs left we have removed some already. However removal of all jobs is obviously one of the things we are very interested in (less maintenance for our tripleo-ci team) ... one example recently is there https://review.opendev.org/q/topic:stein-tripleo for stein from Takashi\n\n\n\n\n\u003e (Giving -1 to get attention to my questions)","accounts_in_message":[],"_revision_number":1},{"id":"822cf2dd371d7618bc28c39634076cd236376c12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-07-05 15:30:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"96118d9186fa26b5f8b63fd685e642b3797a72ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-07-05 15:44:29.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"532dbb8ca073bb57857b6ab5e63d85d9015aff77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 15:58:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b45b7cfce4f4f14996b038259cc375d : SUCCESS in 6m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6283f4a1b32d44e78841c61914e26281 : SUCCESS in 5m 09s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e6dec0277f7a480d990f52c4a7ccfa5a : SUCCESS in 8m 56s","accounts_in_message":[],"_revision_number":3},{"id":"5d5ad7a9417eeb4af26583d0d335d5f23e6be62a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-07-05 19:18:36.000000000","message":"Patch Set 3: Code-Review+2\n\nThanks for the update Marios! LGTM.","accounts_in_message":[],"_revision_number":3},{"id":"47c9157314d8fc5dcd002373deef856a4403492a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-07-05 19:22:27.000000000","message":"Patch Set 3:\n\n+ the mail was sent out 2+ weeks ago so this patch is ready for approval in my opinion.","accounts_in_message":[],"_revision_number":3},{"id":"40ad5a9de2a0a4a84a3a07a7a381e2306424993f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-07-06 13:24:57.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"96593c295bca73c1dd23b00c2d0e9486768ffdd8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-07-12 14:09:48.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d03e6316f68422558f0200579acda12a6fe58989","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 14:10:07.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5fd877cee80fa919e5ebd1ce424fdcb9f3e31c72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 14:17:41.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc9a23b581d546a0b9c484f2e582b74d : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/886a0b86219e4e20974dbdecacef124f : SUCCESS in 6m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/552d291e96df4ba18d9b67d1a78bb388 : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":3},{"id":"632f392d17994548e3e37e43f3d4ccddb82ca734","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 14:17:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"207c426472555891279757ef0493abb9bbf89701","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-07-14 14:12:04.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"0d07335b3dbf808c97892338c80682247b5d22a9","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2021-07-15 10:42:40.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"37d51ffe3b28b8680572a9cbc91971bd93d82b2d","revisions":{"0cbc908eb4b3a59874acaa690ddbfe3443c08dfb":{"kind":"REWORK","_number":1,"created":"2021-06-30 15:57:46.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/24/798924/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/24/798924/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/24/798924/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/24/798924/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/24/798924/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/24/798924/1"}}},"commit":{"parents":[{"commit":"013a0d73b59f103c3742a467dbea7f5a5a612a00","subject":"Merge \"puppet: Create ocata/pike-eol tag for puppet-openstack-integration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/013a0d73b59f103c3742a467dbea7f5a5a612a00"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-06-30 15:42:20.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-06-30 15:54:53.000000000","tz":180},"subject":"Tag stable/queens tripleo repos as EOL","message":"Tag stable/queens tripleo repos as EOL\n\nAs discussed at [1] this tags the last commits to the\nstable/queens branch as queens-eol for tripleo repos [2] except\nthose that are already retired such as tripleo-ui.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2021-June/023115.html\n[2] https://releases.openstack.org/teams/tripleo.html#queens\n\nChange-Id: I5893202c31dfadbe5009a87f4e545787fd065368\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0cbc908eb4b3a59874acaa690ddbfe3443c08dfb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0cbc908eb4b3a59874acaa690ddbfe3443c08dfb"}]},"branch":"refs/heads/master"},"0924a82bf60d391ad28287d0bf5b0e74649956ef":{"kind":"REWORK","_number":2,"created":"2021-07-05 15:30:27.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/24/798924/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/24/798924/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/24/798924/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/24/798924/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/24/798924/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/24/798924/2"}}},"commit":{"parents":[{"commit":"7a1d33209b5314906fbc34e1c8161c18d6170e18","subject":"Merge \"[solum] Transition Ocata to End of Life\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7a1d33209b5314906fbc34e1c8161c18d6170e18"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-07-05 15:28:48.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-07-05 15:28:48.000000000","tz":180},"subject":"Tag stable/queens tripleo repos as EOL","message":"Tag stable/queens tripleo repos as EOL\n\nAs discussed at [1] this tags the last commits to the\nstable/queens branch as queens-eol for tripleo repos [2] except\nthose that are already retired such as tripleo-ui.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2021-June/023115.html\n[2] https://releases.openstack.org/teams/tripleo.html#queens\n\nChange-Id: I5893202c31dfadbe5009a87f4e545787fd065368\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0924a82bf60d391ad28287d0bf5b0e74649956ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0924a82bf60d391ad28287d0bf5b0e74649956ef"}]},"branch":"refs/heads/master"},"37d51ffe3b28b8680572a9cbc91971bd93d82b2d":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-07-05 15:44:29.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/24/798924/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/24/798924/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/24/798924/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/24/798924/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/24/798924/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/24/798924/3"}}},"commit":{"parents":[{"commit":"7a1d33209b5314906fbc34e1c8161c18d6170e18","subject":"Merge \"[solum] Transition Ocata to End of Life\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7a1d33209b5314906fbc34e1c8161c18d6170e18"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-07-05 15:28:48.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-07-05 15:44:22.000000000","tz":0},"subject":"Tag stable/queens tripleo repos as EOL","message":"Tag stable/queens tripleo repos as EOL\n\nAs discussed at [1] this tags the last commits to the\nstable/queens branch as queens-eol for tripleo repos [2].\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2021-June/023115.html\n[2] https://releases.openstack.org/teams/tripleo.html#queens\n\nChange-Id: I5893202c31dfadbe5009a87f4e545787fd065368\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/37d51ffe3b28b8680572a9cbc91971bd93d82b2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/37d51ffe3b28b8680572a9cbc91971bd93d82b2d"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
