)]}'
{"id":"openstack%2Freleases~799473","triplet_id":"openstack%2Freleases~master~I8550787cdd313bdcedeb3440039b369c68aeba1b","project":"openstack/releases","branch":"master","topic":"odl-stale-eol","hashtags":[],"change_id":"I8550787cdd313bdcedeb3440039b369c68aeba1b","subject":"Create ocata-eol tag for Neutron stadium projects","status":"MERGED","created":"2021-07-05 13:26:05.000000000","updated":"2021-07-07 14:54:09.000000000","submitted":"2021-07-07 14:54:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"799473-odl-stale-eol","meta_rev_id":"054fde7b54fd70d041cc5c5ef8d2b5a6070d82d8","_number":799473,"virtual_id_number":799473,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-07 14:54:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-07-06 15:04:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-07-06 15:44:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2021-07-06 12:04:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2021-07-06 10:44:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":2,"date":"2021-07-07 14:32:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":1,"date":"2021-07-07 14:32:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-07-06 10:45:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-05 13:27:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-05 13:30:44.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-07-05 13:30:44.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2021-07-05 13:30:44.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-07-05 13:30:44.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-07-05 13:43:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-05 15:40:01.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-07-07 14:32:25.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"2eb989b5c2f235fb45840d9132450020048dd9ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-07-05 13:26:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2029f13417bf8bfc44342a4ba5ce38f6baebccb5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-07-05 13:29:12.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"3f3462c036d7d3c588a853fba843b2b91247987e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 13:43:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1828d910a82b406d9bfe7e8f2e6d746c : SUCCESS in 7m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/45eb0d5b8a504ef9a650872777df55a1 : SUCCESS in 5m 32s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6612c8fff1fc474a89b1f62440055283 : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":2},{"id":"28e2334db9c78dbc15759284e210dc9a1b5387be","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-07-05 14:09:17.000000000","message":"Patch Set 2: Code-Review-1\n\nthis patch looks OK to me, however I see some more neutron related repos that has stable/ocata open and could be EOL\u0027d:\n\nnetworking-bagpipe has stable/ocata!\nnetworking-bgpvpn has stable/ocata!\nnetworking-generic-switch has stable/ocata!\nnetworking-hyperv has stable/ocata!\nnetworking-l2gw has stable/ocata!\nnetworking-midonet has stable/ocata!\nnetworking-odl has stable/ocata!\nnetworking-onos has stable/ocata!\nnetworking-powervm has stable/ocata!\nnetworking-sfc has stable/ocata!\nneutron-lbaas-dashboard has stable/ocata!\nneutron-vpnaas has stable/ocata!\npython-neutronclient has stable/ocata!\n\nCould you add these, too?\n\n(and I guess the following are not directly owned/was owned by neutron team so probably it\u0027s not relevant for this patch:\n\ndeb-networking-arista has stable/ocata!\ndeb-networking-hyperv has stable/ocata!\ndeb-networking-l2gw has stable/ocata!\ndeb-networking-midonet has stable/ocata!\ndeb-networking-mlnx has stable/ocata!\ndeb-networking-odl has stable/ocata!\ndeb-networking-ovn has stable/ocata!\ndeb-neutron has stable/ocata!\ndeb-neutron-fwaas has stable/ocata!\ndeb-neutron-lbaas has stable/ocata!\ndeb-neutron-taas has stable/ocata!\ndeb-neutron-vpnaas has stable/ocata!\ndeb-python-neutron-lib has stable/ocata!\ndeb-python-neutronclient has stable/ocata!\n)","accounts_in_message":[],"_revision_number":2},{"id":"f9b0eeca417154dd1ae3e1258ae9009abb2939b9","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-07-05 15:35:41.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2: Code-Review-1\n\u003e \n\u003e this patch looks OK to me, however I see some more neutron related repos that has stable/ocata open and could be EOL\u0027d:\n\u003e \n\u003e networking-bagpipe has stable/ocata!\n\u003e networking-bgpvpn has stable/ocata!\n\u003e networking-generic-switch has stable/ocata!\n\u003e networking-hyperv has stable/ocata!\n\u003e networking-l2gw has stable/ocata!\n\u003e networking-midonet has stable/ocata!\n\u003e networking-odl has stable/ocata!\n\u003e networking-onos has stable/ocata!\n\u003e networking-powervm has stable/ocata!\n\u003e networking-sfc has stable/ocata!\n\u003e neutron-lbaas-dashboard has stable/ocata!\n\u003e neutron-vpnaas has stable/ocata!\n\u003e python-neutronclient has stable/ocata!\n\u003e \n\u003e Could you add these, too?\n\u003e \n\u003e (and I guess the following are not directly owned/was owned by neutron team so probably it\u0027s not relevant for this patch:\n\u003e \n\u003e deb-networking-arista has stable/ocata!\n\u003e deb-networking-hyperv has stable/ocata!\n\u003e deb-networking-l2gw has stable/ocata!\n\u003e deb-networking-midonet has stable/ocata!\n\u003e deb-networking-mlnx has stable/ocata!\n\u003e deb-networking-odl has stable/ocata!\n\u003e deb-networking-ovn has stable/ocata!\n\u003e deb-neutron has stable/ocata!\n\u003e deb-neutron-fwaas has stable/ocata!\n\u003e deb-neutron-lbaas has stable/ocata!\n\u003e deb-neutron-taas has stable/ocata!\n\u003e deb-neutron-vpnaas has stable/ocata!\n\u003e deb-python-neutron-lib has stable/ocata!\n\u003e deb-python-neutronclient has stable/ocata!\n\u003e )\n\nThanks, I think we have to cover in this patch the projects under Neutron governance:\nhttps://governance.openstack.org/tc/reference/projects/neutron.html\n\nI am not sure how to find an ocata version of this page, but looking for it :-)","accounts_in_message":[],"_revision_number":2},{"id":"160798181b77c31122564f6799303a36ce498900","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-07-05 15:50:17.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2: Code-Review-1\n\u003e \n\u003e this patch looks OK to me, however I see some more neutron related repos that has stable/ocata open and could be EOL\u0027d:\n\u003e \n\u003e Could you add these, too?\n\nSome networking related repositories are/were under the neutron team,\nbut some are/were never under the neutron governance.\n\n\u003e networking-bagpipe has stable/ocata!\n\u003e networking-bgpvpn has stable/ocata!\n\u003e networking-midonet has stable/ocata!\n\u003e networking-odl has stable/ocata!\n\u003e networking-sfc has stable/ocata!\n\u003e neutron-vpnaas has stable/ocata!\n\u003e python-neutronclient has stable/ocata!\n\nThe above are under the neutron governance.\nWe (as the neutron team) can mark them as EOL in this patch.\n\n\u003e networking-l2gw has stable/ocata!\n\u003e networking-onos has stable/ocata!\n\nThe above were under the neutron governance and they are not retired now.\nIf the release team is fine with the neutron team ack, we can mark them as EOL in this patch.\n\nThe below are not under the neutron governance.\n\n\u003e networking-generic-switch has stable/ocata!\n\nunder ironic team\n\n\u003e networking-hyperv has stable/ocata!\n\nunder winstackers team\n\n\u003e networking-powervm has stable/ocata!\n\nunder PowerVMStackers team\n\n\u003e neutron-lbaas-dashboard has stable/ocata!\n\nunder octavia team","accounts_in_message":[],"_revision_number":2},{"id":"53ac711761223edd5e6e7ec515104257774286c4","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-07-05 16:04:28.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e \u003e Patch Set 2: Code-Review-1\n\u003e \u003e \n\u003e \u003e this patch looks OK to me, however I see some more neutron related repos that has stable/ocata open and could be EOL\u0027d:\n\u003e \u003e \n\u003e \u003e networking-bagpipe has stable/ocata!\n\u003e \u003e networking-bgpvpn has stable/ocata!\n\u003e \u003e networking-generic-switch has stable/ocata!\n\u003e \u003e networking-hyperv has stable/ocata!\n\u003e \u003e networking-l2gw has stable/ocata!\n\u003e \u003e networking-midonet has stable/ocata!\n\u003e \u003e networking-odl has stable/ocata!\n\u003e \u003e networking-onos has stable/ocata!\n\u003e \u003e networking-powervm has stable/ocata!\n\u003e \u003e networking-sfc has stable/ocata!\n\u003e \u003e neutron-lbaas-dashboard has stable/ocata!\n\u003e \u003e neutron-vpnaas has stable/ocata!\n\u003e \u003e python-neutronclient has stable/ocata!\n\u003e \u003e \n\u003e \u003e Could you add these, too?\n\u003e \u003e \n\u003e \u003e (and I guess the following are not directly owned/was owned by neutron team so probably it\u0027s not relevant for this patch:\n\u003e \u003e \n\u003e \u003e deb-networking-arista has stable/ocata!\n\u003e \u003e deb-networking-hyperv has stable/ocata!\n\u003e \u003e deb-networking-l2gw has stable/ocata!\n\u003e \u003e deb-networking-midonet has stable/ocata!\n\u003e \u003e deb-networking-mlnx has stable/ocata!\n\u003e \u003e deb-networking-odl has stable/ocata!\n\u003e \u003e deb-networking-ovn has stable/ocata!\n\u003e \u003e deb-neutron has stable/ocata!\n\u003e \u003e deb-neutron-fwaas has stable/ocata!\n\u003e \u003e deb-neutron-lbaas has stable/ocata!\n\u003e \u003e deb-neutron-taas has stable/ocata!\n\u003e \u003e deb-neutron-vpnaas has stable/ocata!\n\u003e \u003e deb-python-neutron-lib has stable/ocata!\n\u003e \u003e deb-python-neutronclient has stable/ocata!\n\u003e \u003e )\n\u003e \n\u003e Thanks, I think we have to cover in this patch the projects under Neutron governance:\n\u003e https://governance.openstack.org/tc/reference/projects/neutron.html\n\u003e \n\u003e I am not sure how to find an ocata version of this page, but looking for it :-)\n\nFYI: information on the retired repositories is found at legacy.yaml in the governance repo.\nhttps://opendev.org/openstack/governance/src/branch/master/reference/legacy.yaml#L552","accounts_in_message":[],"_revision_number":2},{"id":"85202f36911cb26dfc44ba85812afbbd9a7706b8","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-07-05 16:11:04.000000000","message":"Patch Set 2:\n\n\u003e \u003e networking-l2gw has stable/ocata!\n\u003e \u003e networking-onos has stable/ocata!\n\u003e \n\u003e The above were under the neutron governance and they are not retired now.\n\u003e If the release team is fine with the neutron team ack, we can mark them as EOL in this patch.\n\u003e \nMinor correction: ... and *they are retired now*.\n(Somehow I first wrote \"they are not the neutron governance now and rephrased it partially :( )","accounts_in_message":[],"_revision_number":2},{"id":"8ce669c5762973e3b7cb28cc17cda04110af0dc2","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-07-05 16:45:09.000000000","message":"Patch Set 2:\n\nThanks @Akihiro for the explanations! :)","accounts_in_message":[],"_revision_number":2},{"id":"4ab29e0a321f682d723ba693d73575bb4d1878bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-07-06 08:31:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e2580bb533e847691610092b3147c5acc1c28661","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-07-06 08:32:35.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 2:\n\u003e \n\u003e Thanks @Akihiro for the explanations! :)\n\nI think I added all Neutron governed projects now that is missing","accounts_in_message":[],"_revision_number":3},{"id":"26ea4ed4b6cf643338ddb4e7b6a9401bdb3c70d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 08:42:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5aa6a91897942c7850b5a7f7450c4c2 : SUCCESS in 6m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1df73f521b69407088691c32718d8cf0 : SUCCESS in 5m 41s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/029d31c85eab4f31893bfc6eb5c6b93b : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":3},{"id":"147ffa650807035079ddea4bb6ebfca0aab29c10","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-07-06 10:44:16.000000000","message":"Patch Set 3: Code-Review+1\n\nThanks for the update.\nAs far as I checked, it covers all repositories with stable/ocata branch under the neutron governance.","accounts_in_message":[],"_revision_number":3},{"id":"1b1622645ca154c2fcfc715412af6fd366475d69","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 10:45:20.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"e71bbc05308ef35f39a3111dbe7de9371395acda","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-07-06 12:04:00.000000000","message":"Patch Set 3: Code-Review+1\n\nList looks good to me on Ocata yes","accounts_in_message":[],"_revision_number":3},{"id":"e5a35a0d5d473ae9e00ad723e82c1f7d7d42c052","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-06 15:04:33.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"e9c02c4e8ac70ac6770d0f7c5c55b57ca513c0b0","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-07-06 15:43:57.000000000","message":"Patch Set 3:\n\n(2 comments)\n\nThanks for the update! LGTM.\n\n(I think \u0027release-model\u0027 and \u0027stable-branch-type\u0027 are not necessary in the yamls but does not cause any harm so I\u0027m OK with leaving them there)","accounts_in_message":[],"_revision_number":3},{"id":"c0accfd5b55bfb46f77e48ae253d45d10b1d0c7c","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-07-06 15:44:12.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"8cee09735f77708cf26670c2440bcf8791774358","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-07-07 14:32:25.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"cea2fb4a079e056a00bc1171c1bbf58c48819420","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 14:32:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d8edc7078abe055ced4971f795e496f54d043a69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 14:54:08.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9acdf034b4245e0aeef64e83ceaf239 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48593e44b8c84b12b74008dcb17f04ee : SUCCESS in 6m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1de5a090b20845e8a1e4248a07ea288f : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":3},{"id":"054fde7b54fd70d041cc5c5ef8d2b5a6070d82d8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 14:54:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7eff128679485ef4bb30d9f903decfdddcfdb760","revisions":{"99cf75ec12dc17cc852df41f4bc5128282040293":{"kind":"REWORK","_number":1,"created":"2021-07-05 13:26:05.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/73/799473/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/73/799473/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/799473/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/799473/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/799473/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/73/799473/1"}}},"commit":{"parents":[{"commit":"3186ae575612dc63d267ecde99c980d474c91f68","subject":"Merge \"Release OpenStack-Ansible RC2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3186ae575612dc63d267ecde99c980d474c91f68"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-07-05 13:21:55.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-07-05 13:21:55.000000000","tz":120},"subject":"Create ocata-eol tag for Neutron stadium projects","message":"Create ocata-eol tag for Neutron stadium projects\n\nThe stable/ocata branch of Neutron stadium projects were EOLed (see [1])\nThis patch adds ocata-eol tag for these projects, so Ocata branch can be\nretired.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2021-May/022269.html\n\nChange-Id: I8550787cdd313bdcedeb3440039b369c68aeba1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/99cf75ec12dc17cc852df41f4bc5128282040293"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/99cf75ec12dc17cc852df41f4bc5128282040293"}]},"branch":"refs/heads/master"},"9952a8f081b5bad44a8ee33e6fb5d196c5c5a49f":{"kind":"REWORK","_number":2,"created":"2021-07-05 13:29:12.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/73/799473/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/73/799473/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/799473/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/799473/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/799473/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/73/799473/2"}}},"commit":{"parents":[{"commit":"3186ae575612dc63d267ecde99c980d474c91f68","subject":"Merge \"Release OpenStack-Ansible RC2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3186ae575612dc63d267ecde99c980d474c91f68"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-07-05 13:21:55.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2021-07-05 13:29:07.000000000","tz":0},"subject":"Create ocata-eol tag for Neutron stadium projects","message":"Create ocata-eol tag for Neutron stadium projects\n\nThe stable/ocata branch of Neutron stadium projects were EOLed (see [1])\nThis patch adds ocata-eol tag for these projects, so Ocata branch can be\nretired.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2021-May/022269.html\n\nChange-Id: I8550787cdd313bdcedeb3440039b369c68aeba1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9952a8f081b5bad44a8ee33e6fb5d196c5c5a49f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9952a8f081b5bad44a8ee33e6fb5d196c5c5a49f"}]},"branch":"refs/heads/master"},"7eff128679485ef4bb30d9f903decfdddcfdb760":{"kind":"REWORK","_number":3,"created":"2021-07-06 08:31:11.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/73/799473/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/73/799473/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/799473/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/799473/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/799473/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/73/799473/3"}}},"commit":{"parents":[{"commit":"3186ae575612dc63d267ecde99c980d474c91f68","subject":"Merge \"Release OpenStack-Ansible RC2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3186ae575612dc63d267ecde99c980d474c91f68"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-07-05 13:21:55.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-07-06 08:30:48.000000000","tz":120},"subject":"Create ocata-eol tag for Neutron stadium projects","message":"Create ocata-eol tag for Neutron stadium projects\n\nThe stable/ocata branch of Neutron stadium projects were EOLed (see [1])\nThis patch adds ocata-eol tag for these projects, so Ocata branch can be\nretired.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2021-May/022269.html\n\nChange-Id: I8550787cdd313bdcedeb3440039b369c68aeba1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7eff128679485ef4bb30d9f903decfdddcfdb760"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7eff128679485ef4bb30d9f903decfdddcfdb760"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
