)]}'
{"id":"openstack%2Freleases~799543","triplet_id":"openstack%2Freleases~master~I503828b13ae9203c48e4ea948af969aaf2671b1e","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I503828b13ae9203c48e4ea948af969aaf2671b1e","subject":"Create queens-eol tag for horizon project","status":"MERGED","created":"2021-07-05 20:02:17.000000000","updated":"2021-07-16 10:38:46.000000000","submitted":"2021-07-16 10:38:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"799543","meta_rev_id":"207bc7d401f4bec4353a56a872d6a153c5d2818b","_number":799543,"virtual_id_number":799543,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-16 10:38:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2021-07-16 10:13:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2021-07-07 15:27:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":2,"date":"2021-07-16 10:28:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":1,"date":"2021-07-16 10:28:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-07-05 20:03:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}],"REVIEWER":[{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-05 20:03:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-07 15:27:21.000000000","updated_by":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"REVIEWER"},{"updated":"2021-07-08 11:56:25.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-07-09 13:07:23.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-07-15 16:05:40.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"}],"messages":[{"id":"d8e18e706520375a0d6a6dfd99689b9baffaef7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-07-05 20:02:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a48388e951954141767c682afaa72667c55c0b3c","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 20:03:12.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"57048f574ccf42307f262213f3500196fc3ed5ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 20:09:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d4d21343cf749d6992a5261581f145a : SUCCESS in 6m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e3d617fea86a4d06a2525f67eb58021c : SUCCESS in 4m 59s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/8b3221963cd6497281feb1264d5927df : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":1},{"id":"7a60fe4e96ecf2390be33704eed3095ccabd85cd","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2021-07-07 15:27:21.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ea514d207e8580ba1758e7e16a13afb36af898e2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-07-08 11:56:25.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM\n\n- hash refer to the latest changes in stable/queens\n- no patch remained opened https://review.opendev.org/q/branch:stable/queens+project:openstack/horizon+status:open","accounts_in_message":[],"_revision_number":1},{"id":"77c2677e68eeba5155d8a7555bc38008f620d87e","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-07-09 13:07:23.000000000","message":"Patch Set 1: Code-Review-1\n\nA couple of things here:\n\n* The last patch on horizon\u0027s stable/queens was merged in March, that is not that old. Since there are not so much activity on this branch (in Extended Maintenance) it does not need so much attention to keep it open (at least as long as gate is working) so in my understanding it could be kept open as long as gate is fine, without any pain. Does the team really want to close this branch?\n* Have you checked whether there are zuul job definitions outside openstack/horizon repository that can be deleted as soon as stable/queens is deleted? (2nd step in EOL process [1])\n* Is there any other repository that is handled by Horizon team and its stable/queens can be EOL\u0027d together with Horizon\u0027s stable/queens?\n\nI\u0027ll give -1 for now to get some attention to my comments. (Of course if everything is good to go, then I\u0027ll change it to +2)\n\n[1] https://docs.openstack.org/project-team-guide/stable-branches.html#end-of-life","accounts_in_message":[],"_revision_number":1},{"id":"59e88065245e1980543a1dd317628fc1610d143e","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-07-12 07:21:06.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e A couple of things here:\n\u003e \n\u003e * The last patch on horizon\u0027s stable/queens was merged in March, that is not that old. Since there are not so much activity on this branch (in Extended Maintenance) it does not need so much attention to keep it open (at least as long as gate is working) so in my understanding it could be kept open as long as gate is fine, without any pain. Does the team really want to close this branch?\n\u003e\u003e yeah horizon team discussed this topic and agreed to move stable/queens to EOL.\n\n\u003e * Have you checked whether there are zuul job definitions outside openstack/horizon repository that can be deleted as soon as stable/queens is deleted? (2nd step in EOL process [1])\n\u003e\u003e I have gone through the horizon stable/queens``.zuul.yaml`` [2] and didn\u0027t find any zuul job definitions that can be removed once stable/queens is deleted.\n[2] https://github.com/openstack/horizon/blob/stable/queens/.zuul.yaml\n\n\u003e * Is there any other repository that is handled by Horizon team and its stable/queens can be EOL\u0027d together with Horizon\u0027s stable/queens?\n\u003e \n\u003e\u003e As you may already know horizon have many horizon plugins and I looked at each horizon plugins ``.zuul.yaml`` and found one plugin i.e. openstack/manila-ui [3] depends-on horizon django-job templates. So we can\u0027t move further until manila-ui stable/queens(currently em) is moved to EOL. \n\n[3] https://github.com/openstack/manila-ui/blob/stable/queens/.zuul.yaml#L9-L10\n\n\u003e I\u0027ll give -1 for now to get some attention to my comments. (Of course if everything is good to go, then I\u0027ll change it to +2)\n\u003e \n\u003e [1] https://docs.openstack.org/project-team-guide/stable-branches.html#end-of-life","accounts_in_message":[],"_revision_number":1},{"id":"a5ba264555b5ad0660ae9af3d680fed0ea3eded8","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-07-12 10:49:39.000000000","message":"Patch Set 1:\n\n\u003e \u003e * Is there any other repository that is handled by Horizon team and its stable/queens can be EOL\u0027d together with Horizon\u0027s stable/queens?\n\u003e \u003e \n\u003e \u003e\u003e As you may already know horizon have many horizon plugins and I looked at each horizon plugins ``.zuul.yaml`` and found one plugin i.e. openstack/manila-ui [3] depends-on horizon django-job templates. So we can\u0027t move further until manila-ui stable/queens(currently em) is moved to EOL. \n\nOr simply that jobs are removed from manila-ui (or the job definition is moved there from horizon).\nThanks Vishal for checking these!\n\n\u003e \n\u003e [3] https://github.com/openstack/manila-ui/blob/stable/queens/.zuul.yaml#L9-L10","accounts_in_message":[],"_revision_number":1},{"id":"727c7b6f489a5caaa6c0580a898a0b8e34738691","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-07-12 14:10:56.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"30c419357b955e85227413103f2e3f07085ea21f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-07-15 16:05:40.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e \u003e * Is there any other repository that is handled by Horizon team and its stable/queens can be EOL\u0027d together with Horizon\u0027s stable/queens?\n\u003e \u003e \u003e \n\u003e \u003e \u003e\u003e As you may already know horizon have many horizon plugins and I looked at each horizon plugins ``.zuul.yaml`` and found one plugin i.e. openstack/manila-ui [3] depends-on horizon django-job templates. So we can\u0027t move further until manila-ui stable/queens(currently em) is moved to EOL. \n\u003e \n\u003e Or simply that jobs are removed from manila-ui (or the job definition is moved there from horizon).\n\u003e Thanks Vishal for checking these!\n\u003e \n\u003e \u003e \n\u003e \u003e [3] https://github.com/openstack/manila-ui/blob/stable/queens/.zuul.yaml#L9-L10\n\n\nThanks for checking Elod and Vishal. We discussed this at the manila weekly IRC meeting and decided to pursue EOL\u0027ing of the stable/queens branch on that repository - so no job changes are necessary\n\n[1] https://review.opendev.org/c/openstack/releases/+/800963","accounts_in_message":[],"_revision_number":1},{"id":"f1774a177c8bbb2bad616b7909f88455b47e329f","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-07-16 10:13:31.000000000","message":"Patch Set 1: Code-Review+2\n\nWith the manila-ui\u0027s stable/queens EOL patch I think we are good to go here. Thanks for looking after the zuul job definitions!","accounts_in_message":[],"_revision_number":1},{"id":"766e0e0d3b36ca606f4ab3f650207ab77a5d266e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-07-16 10:28:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ff9568753e25b4450eae5400976a926b11b2c7c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 10:28:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"440f37a0a349008c96c7868b6b9f7d574dc5e1bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 10:38:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33d3bc956ffc464487352e43ee642bc7 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6900a1a58ae34344b18ccef17b7e004f : SUCCESS in 6m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/be189198229b4347b11370a8c36ffb7c : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":1},{"id":"207bc7d401f4bec4353a56a872d6a153c5d2818b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 10:38:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aa83f51b3bc3b90ba05556215c1d18c72a7f8dae","revisions":{"aa83f51b3bc3b90ba05556215c1d18c72a7f8dae":{"kind":"REWORK","_number":1,"created":"2021-07-05 20:02:17.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/43/799543/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/43/799543/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/799543/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/799543/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/799543/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/43/799543/1"}}},"commit":{"parents":[{"commit":"7a1d33209b5314906fbc34e1c8161c18d6170e18","subject":"Merge \"[solum] Transition Ocata to End of Life\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7a1d33209b5314906fbc34e1c8161c18d6170e18"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-07-05 19:59:29.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-07-05 19:59:29.000000000","tz":330},"subject":"Create queens-eol tag for horizon project","message":"Create queens-eol tag for horizon project\n\nHorizon Team is no longer accepting patches to stable/queens\nbranch, this patch proposes to create queens-eol tags for\nhorizon project.\n\nAnnouncement email: http://lists.openstack.org/pipermail/openstack-discuss/2021-June/023381.html\n\nChange-Id: I503828b13ae9203c48e4ea948af969aaf2671b1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/aa83f51b3bc3b90ba05556215c1d18c72a7f8dae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/aa83f51b3bc3b90ba05556215c1d18c72a7f8dae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
