)]}'
{"id":"openstack%2Freleases~801143","triplet_id":"openstack%2Freleases~master~I3a32fb875a2bd3ab7aec21921f198b80a86f68cf","project":"openstack/releases","branch":"master","topic":"queens-eol","hashtags":[],"change_id":"I3a32fb875a2bd3ab7aec21921f198b80a86f68cf","subject":"[manila-ui] Transition stable/rocky to eol","status":"MERGED","created":"2021-07-16 21:57:48.000000000","updated":"2021-07-29 12:18:15.000000000","submitted":"2021-07-29 12:18:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"801143-queens-eol","meta_rev_id":"cf51354ebf688be2565c6b5403bf52e93f669669","_number":801143,"virtual_id_number":801143,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-29 12:18:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"recommended":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"all":[{"value":1,"date":"2021-07-22 15:13:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"value":2,"date":"2021-07-29 12:05:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2021-07-24 07:54:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2021-07-17 10:39:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"value":1,"date":"2021-07-29 12:05:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2021-07-16 22:09:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-16 21:58:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-16 22:20:13.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2021-07-19 17:38:37.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-07-22 15:13:31.000000000","updated_by":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"reviewer":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"state":"REVIEWER"},{"updated":"2021-07-27 20:16:33.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2021-07-27 20:16:59.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"a60433d0ed37622cfd1db651ee05566af3ec66d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-07-16 21:57:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d96d5952fee15956b6f1c9f1d0bbc6113a6a93ba","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 21:58:42.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"d7e6bcffbc7caa00bdb9c33405b87594f40aeab2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-07-16 22:09:02.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ed35f82c70c4c9e18e671efaeacb2277079fdfb0","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 22:09:59.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"bcba0b1d318ca02914e2daa3a9a45d3bc313f344","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 22:18:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04293deb5fe943deacb0d77256c6a920 : SUCCESS in 7m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2dfebbc4cf4e4748b0fbd461ca9d30ee : SUCCESS in 5m 45s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/972df819481546a48d2eb6ce3941fb89 : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":2},{"id":"384ba31e5b0f2e96986c6cc2f59a251d24aeab9f","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-07-17 10:39:49.000000000","message":"Patch Set 2: Code-Review+1\n\nThanks, Goutham!","accounts_in_message":[],"_revision_number":2},{"id":"9554f7ec88cf80ef7534eec300795787c058ed86","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-07-19 17:38:37.000000000","message":"Patch Set 2: Code-Review+1\n\nLooks OK (hash is correct).\nAnd as usual my question:\n- is there any zuul job definition outside manila-ui repository, that can be removed as won\u0027t be used anymore?\n\nAlso, let\u0027s wait 2-3 days to see if there is anyone who replies to mailing list and wants to maintain manila-ui\u0027s rocky.","accounts_in_message":[],"_revision_number":2},{"id":"f22dbce40340ab9da6de3201f1d432c6259a0a4b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-07-20 21:23:58.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2: Code-Review+1\n\u003e \n\u003e Looks OK (hash is correct).\n\u003e And as usual my question:\n\u003e - is there any zuul job definition outside manila-ui repository, that can be removed as won\u0027t be used anymore?\n\nhttps://opendev.org/openstack/manila-ui/src/branch/stable/rocky/.zuul.yaml\n\nWith my search-foo, i couldn\u0027t find any test job that was exclusive to this repository.\n\n\u003e \n\u003e Also, let\u0027s wait 2-3 days to see if there is anyone who replies to mailing list and wants to maintain manila-ui\u0027s rocky.","accounts_in_message":[],"_revision_number":2},{"id":"1813c1d796cdb60ff9f646209084eebdf716845c","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2021-07-22 15:13:31.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"66df645a828d650486790952a87bb407cbbfabcb","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-07-24 07:54:49.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks for this, too, Goutham! We have waited with this enough time, too, let\u0027s upgrade to +2.","accounts_in_message":[],"_revision_number":2},{"id":"07aee9c9663ab01ff18f67263dddc21abca2d0ca","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-07-27 20:15:42.000000000","message":"Patch Set 2:\n\nbump","accounts_in_message":[],"_revision_number":2},{"id":"626ab720b15a1b9fb486f16e518666811163216b","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-07-29 12:05:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2c17de4d7cb9b0ac9a7c169cb38fd4b83caf13a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 12:05:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"de66bdf16c87ac2f7ecd17c0430313824e42fd89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 12:18:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5a4f62afe064cc89fa8990f5ffed6c4 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e09001c4dca74104b98167c7b213b078 : SUCCESS in 6m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1ac1969e7f78476aa810831ad49ee2cb : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":2},{"id":"cf51354ebf688be2565c6b5403bf52e93f669669","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 12:18:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"237f9ad65d7f8d4c4a602fb9c3475e6572142fb0","revisions":{"b8c762ae60df4db6fe5a33e1e0fe2a9ea03dd972":{"kind":"REWORK","_number":1,"created":"2021-07-16 21:57:48.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/43/801143/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/43/801143/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/801143/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/801143/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/801143/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/43/801143/1"}}},"commit":{"parents":[{"commit":"3da0153e40e380045db115c8ae5e5b24fcf6cdf9","subject":"Merge \"Propose to retire monasca-analytics from deliverables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3da0153e40e380045db115c8ae5e5b24fcf6cdf9"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-07-16 21:56:18.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-07-16 21:56:18.000000000","tz":-420},"subject":"[manila-ui] Transition stable/rocky to eol","message":"[manila-ui] Transition stable/rocky to eol\n\nManila\u0027s maintainers would like to create the eol release tag\nwith the latest code from the stable/rocky branch on the manila-ui\nrepository and delete this release branch from git once\nthis tags has been created. See the discussion on the\nOpenStack Mailing list [1] for the announcement.\n\n[1]\n\nChange-Id: I3a32fb875a2bd3ab7aec21921f198b80a86f68cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b8c762ae60df4db6fe5a33e1e0fe2a9ea03dd972"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b8c762ae60df4db6fe5a33e1e0fe2a9ea03dd972"}]},"branch":"refs/heads/master"},"237f9ad65d7f8d4c4a602fb9c3475e6572142fb0":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-07-16 22:09:02.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/43/801143/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/43/801143/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/801143/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/801143/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/801143/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/43/801143/2"}}},"commit":{"parents":[{"commit":"3da0153e40e380045db115c8ae5e5b24fcf6cdf9","subject":"Merge \"Propose to retire monasca-analytics from deliverables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3da0153e40e380045db115c8ae5e5b24fcf6cdf9"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-07-16 21:56:18.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-07-16 22:08:52.000000000","tz":0},"subject":"[manila-ui] Transition stable/rocky to eol","message":"[manila-ui] Transition stable/rocky to eol\n\nManila\u0027s maintainers would like to create the eol release tag\nwith the latest code from the stable/rocky branch on the manila-ui\nrepository and delete this release branch from git once\nthis tags has been created. See the discussion on the\nOpenStack Mailing list [1] for the announcement.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2021-July/023702.html\n\nChange-Id: I3a32fb875a2bd3ab7aec21921f198b80a86f68cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/237f9ad65d7f8d4c4a602fb9c3475e6572142fb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/237f9ad65d7f8d4c4a602fb9c3475e6572142fb0"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
